)]}'
{"id":"openstack%2Frequirements~756139","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I6a847dd3702c726843e477a4f9e936d17af1bb7d","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I6a847dd3702c726843e477a4f9e936d17af1bb7d","subject":"update constraint for os-collect-config to new release 11.0.1","status":"MERGED","created":"2020-10-05 17:26:03.000000000","updated":"2020-10-07 21:49:18.000000000","submitted":"2020-10-07 21:49:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756139-1602107358185-5cec80e2","meta_rev_id":"d8f848801c63aaa76541a280207801900ca86735","_number":756139,"virtual_id_number":756139,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-07 21:49:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-10-06 01:10:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-10-05 17:46:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-10-06 01:10:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 17:46:23.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-06 01:10:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-07 21:49:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac29cbab3915eb7220857caf93a9db5bc6d05acf","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-05 17:26:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47789df2e9ac3319b3f9bef924f0ac5cf5779f6f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-05 17:46:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"70054a537373ce20f020c81e9119c4e123f230a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 23:23:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3835d11df30e4f189f24983c6e9a9214 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0384fddc80c34425af445def2e83522f : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aec60a10f4c04901966a8b08072aee3b : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9cd1498dc37845069f4ca674dabdec7d : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/163625f7b47648caa569e659890e4b67 : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7ee12664151a4d989401b1e0a575cf32 : SUCCESS in 7m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/092d1fb173f24cceafc6ee11566bdd0c : SUCCESS in 8m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/92e60b6e7e234bd0a77721d1e998df68 : SUCCESS in 5m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9ba075e207ae4a2fa9bdf537b52a2822 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2ce8c08d532743b3b3c9ce8b9e98a05d : SUCCESS in 7m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/73fb65349cf14b3dbafc08752a8ea9ba : SUCCESS in 14m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cf9091c5a05e4e23b71f96d262b3900a : SUCCESS in 27m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c349cc3393b94983ae09e8060d43271d : SUCCESS in 16m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/59cff6516aaf49dc804b6de89a9fc9b6 : SUCCESS in 13m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d1966616b1964b9a9a8030d9c56bf8a6 : SUCCESS in 16m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ff44111612384f638c8c65915996fc70 : SUCCESS in 5m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8a69b8ec48cc4a459fb6350b1ef2ba35 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/07625ff87895440687826d21f8bb4679 : SUCCESS in 6m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0b65dc89221e420b95ada0c9cb6a2200 : SUCCESS in 9m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/254d2cbc3fde49ac9bf225ee0a0712c1 : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ed04f2ac7644db28c83a97b6ef6ed6d : SUCCESS in 46m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7728d598116041a48010aaa599d9a109 : SUCCESS in 1h 11m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9a67482bf6241c1bbd2449a6a1a2034 : SUCCESS in 1h 11m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/b61e4f024d784990bc8f4aa646b7dc0c : SUCCESS in 1h 07m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9fc777d74eb28d7297cbbbb4583be5124cd0b978","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-06 01:10:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"72716eece3165710ba54bca10db97cf3e84c7540","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 01:38:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0867cdd29744acf004629103e0a427f6e4c4efd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 07:26:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e0e205ca0fb4d52a4521d5b07c3e061 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/158a3d4e1c21479298320e2633472cc0 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/808656a16514405d8a92bd7b164d58d3 : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a61017cd1ed64990a28f59b230128071 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78b86119c91c49339eec20b5cc32cddd : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/23df4c27c9b845a2970e3c4ceeeff68a : SUCCESS in 7m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d510dead0c1d46b9afbade930b37743a : SUCCESS in 8m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/60077d6f7faf45bebe905cad18d0fd33 : SUCCESS in 5m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5978fd2e77dd42ec9a80eb5c8437fe8b : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f0fde8c758b4711b3f22c19bdb8fe44 : SUCCESS in 6m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4314ffe5986946519c68e08f9b0837d5 : SUCCESS in 12m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/78d17c3a205b42979f73b2927a38467e : SUCCESS in 22m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d57a5b6c33f54b2d805fb2e40b030dda : SUCCESS in 12m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1d758824d3364b1e9340abcca07dd8d0 : FAILURE in 13m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b20021dba7174e978e58e07fccaa9c3f : SUCCESS in 24m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/17a274a549244765ae56f3fb8e6830cf : SUCCESS in 5m 07s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f0c51f5a079e44a492f92066523a18fb : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52de6f7c81de49cca2a3651641524411 : SUCCESS in 6m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5ed5c123951a4a8ab8cabfe6aa2d9fc7 : SUCCESS in 10m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/de05f176d7954f1487d8edf6e0caa8d7 : SUCCESS in 14m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1c4af503c6945f48e678fbec4f1a54b : SUCCESS in 56m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a2e59cb2cd74660b32fb6208987304a : SUCCESS in 1h 26m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da4d4a66b15540009abbf57d3dadb29a : SUCCESS in 1h 23m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/e0a3cb9511f14aa690e54082add83ef2 : SUCCESS in 1h 15m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1e970edf5f5e3994facaac5de01ce879a0444254","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-07 14:13:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7d09922781b08b2bfeaf2c8c1980ce0725ad727e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:28:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac5ca1d10679404188c0bcfc8f7dd5dd : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4192408d62d44d94ad7b055473ea0e58 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a174e58ac5ec4bee9280d8aacbd57aab : SUCCESS in 5m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/72e7d4a4b0b54cde8bac465796d54546 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/175f4271c16d42e981ace72dbfd629e8 : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d671020ec04845b2b388ad3115c55c5c : SUCCESS in 6m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/595d3a54ddf64c1f9d7702c85d858162 : SUCCESS in 8m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7961feb37c6f4e9884274d1fd8212860 : SUCCESS in 4m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5cd05c112e27405faa69f9407c623e25 : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bcc029f57e864e5a8b61e0f214062297 : SUCCESS in 7m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8acdcdd489264661a9458e0360d3a1c9 : SUCCESS in 18m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4d8e43150b954b96acdb03809de32c0f : SUCCESS in 31m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5663f4e20c8048489613c5ae05716e59 : SUCCESS in 15m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7f82eb39bb204bcc895157ee91962a0f : SUCCESS in 16m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e9901c1318e247b8b668abcba1eac63a : SUCCESS in 15m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e629aa0df52945259cb5783ad90c2708 : SUCCESS in 5m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cedb716673674edb84a31c205e89c704 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15c746607dd14464945508f225618c59 : SUCCESS in 6m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/39da2a4d251e47af82cdbefce02ef7fe : SUCCESS in 9m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/46498c5be62f4d28a4f2374942304f19 : SUCCESS in 9m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2ebaf5c1c824a2cb0086a58c7e8110e : SUCCESS in 55m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2069dec78064118bfaf5f4d11f5a7ca : SUCCESS in 1h 23m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/533cce242e6d4cd084c673d973f23276 : SUCCESS in 1h 14m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/537960f3b40a45b5bfa2814d38d48516 : SUCCESS in 1h 09m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f3f67148f7f969dcc530cf2ef22f2ba8239f95af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:28:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea84d56baf52bcc3410182149000896eccd699c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 21:49:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/164e4505fb79460e8f06b556559c6a41 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58be9e1c09914372a99f56ec199eb028 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c03ed3d8d5b4ee59f872c73a029455d : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f31f1209391a4f069ccdfab96c8cd61a : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c826ea66241a432683e2bf7eb9e74ddb : SUCCESS in 3m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0b374648f6cb46e091acf21f3a2bec07 : SUCCESS in 6m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8c9fc5068961440bbc20cc9f0f17ea25 : SUCCESS in 8m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a3824ffcf9b340a5b2859762b1066320 : SUCCESS in 4m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cd759da8f73941bfbdc2f27913a33fe0 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f040c052d44649beae34c9d3875c830e : SUCCESS in 7m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/829e075074eb4aae8a812ad5cdf25bd4 : SUCCESS in 17m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c1c80381c6ae4af98a7d9987202a6f40 : SUCCESS in 24m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/14eea9554e51401d8e7500001a73d130 : SUCCESS in 11m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9790bb258ad0436ea14ca5a17c0af4b7 : SUCCESS in 14m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e6367b2870f455d8badcad72291012f : SUCCESS in 17m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3a8ad8271d66429fa6a654de79605a41 : SUCCESS in 4m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a6f0d7b52d3749d288524f007f3c571e : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c62dc3989e2c49188f9df065f9192ed2 : SUCCESS in 6m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/48968506d0da40a58b26031721f2a0fb : SUCCESS in 11m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/140b35cb74254776ae9dd903f3a26bed : SUCCESS in 9m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e57cfa61c52435fb2a85327ee6c0ac7 : SUCCESS in 48m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bdad78015a64bf9b42ab0776b28f045 : SUCCESS in 1h 30m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/363f4e18100c43b99120f80228977c0d : SUCCESS in 1h 15m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/cb1fe755d4894a9d8cbc11bd545afb6f : SUCCESS in 1h 16m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e132bedce0be1fbd7cf8d16d5e5c871c3a52dd73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 21:49:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0443beec904ae4a9cf213b0524fe0b765378114","revisions":{"a0443beec904ae4a9cf213b0524fe0b765378114":{"kind":"REWORK","_number":1,"created":"2020-10-05 17:26:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/756139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/756139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/756139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/756139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/756139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/756139/1"}}},"commit":{"parents":[{"commit":"3bb7de97826ce2e972cf51da8afb5b2e1871db87","subject":"update constraint for python-troveclient to new release 3.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3bb7de97826ce2e972cf51da8afb5b2e1871db87"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-05 17:26:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-05 17:26:01.000000000","tz":0},"subject":"update constraint for os-collect-config to new release 11.0.1","message":"update constraint for os-collect-config to new release 11.0.1\n\nChange-Id: I6a847dd3702c726843e477a4f9e936d17af1bb7d\nmeta:version: 11.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: MSA \u003cmarios@redhat.com\u003e\nmeta:release:Commit: MSA \u003cmarios@redhat.com\u003e\nmeta:release:Change-Id: I803eafba77e45fecc00d15ab2c7e72294363ca27\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta:release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0443beec904ae4a9cf213b0524fe0b765378114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0443beec904ae4a9cf213b0524fe0b765378114"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
