)]}'
{"id":"openstack%2Frequirements~756510","triplet_id":"openstack%2Frequirements~stable%2Fstein~I01f53acca86ecd47f85bcf5d73ee985d29e91ba2","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I01f53acca86ecd47f85bcf5d73ee985d29e91ba2","subject":"update constraint for python-heatclient to new release 1.17.1","status":"MERGED","created":"2020-10-07 13:26:23.000000000","updated":"2020-10-07 19:19:02.000000000","submitted":"2020-10-07 19:19:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756510-1602098342008-d979b326","meta_rev_id":"b56f622c52247948ef1161b1f5d49d4dea4dc634","_number":756510,"virtual_id_number":756510,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-07 19:19:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-07 14:03:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-10-07 14:57:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-10-07 14:01:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-10-07 14:03:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-07 14:01:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-07 14:03:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-07 14:57:20.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-07 19:19:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"341b25ead3fa8608f718f035e931fcf1601941c6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-07 13:26:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ba62c06df2fc075f5a6e5c2a69bb19dc9eb0695","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-07 14:01:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6f7195282510074e527d4079d56a881727f79249","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-07 14:03:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0695d74599e1b602bcef123f401bbb02f22d05e6","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-07 14:57:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e9ce4c28a1cae4395cc3bf79e9b3b1db50f48408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 17:28:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb987fdc4ab8431482a884d1f92ed9c6 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3ba79ba8a414a60a36651b8727f4355 : SUCCESS in 6m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/56c08ed2ef0247fdb995f7474d58a64f : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a528a84002e4669b4a3d10d0c09b7ff : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56493f54ec4e42e088db4b870fff40fd : SUCCESS in 5m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aaaf86c2d63542cca9641eaf1c58dfad : SUCCESS in 5m 37s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/37149397cc344c49949a68e8103b5063 : SUCCESS in 9m 58s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/91ccba9e510d483d9ac7518af2471b4a : SUCCESS in 10m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bfeddc29ec9d416e8b3067629d1a4315 : SUCCESS in 8m 10s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/cee6d86693e440b4bd38bfa730072614 : SUCCESS in 8m 37s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/fab6c29dbf20423596b18a914954886a : SUCCESS in 9m 35s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/89bcd97c80374e21a6a3b0edcb09fa9c : SUCCESS in 5m 31s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/0b6bd1ee732c4e4c9266acc44cf5e8c9 : SUCCESS in 6m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/324e19057e8c4f589a31d1167f584256 : SUCCESS in 9m 46s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/dff6050ffec344c2aeb6430a553e1e9f : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4d046a951f614b1184c409cc4ac603cf : SUCCESS in 5m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/538786cd2b904e3ab92559283cd26bf0 : SUCCESS in 22m 04s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/c40a66312e1d4b8d826a5e0b00be1240 : SUCCESS in 33m 55s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/44ba101bcfc2408bb1aec583603266d1 : SUCCESS in 24m 12s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/e7a3e28d34be46f180e64446064c1fc4 : SUCCESS in 25m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b1170a9239094096831a1f2fa1851553 : SUCCESS in 14m 38s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/49cde63da52c41cf869c4e55864924ec : SUCCESS in 15m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/260b35d8e63947e7a1ea134a4f6bd5a3 : SUCCESS in 17m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/aefdd852122f45358c70c119181d6999 : SUCCESS in 8m 31s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/f2cc7f778a314024802ce043b3e687b5 : SUCCESS in 7m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec79e42cf1a04144a792bf814eb4a2e2 : SUCCESS in 1h 01m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/97d432698d474073bcf9fa94a932e518 : SUCCESS in 1h 26m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b8a3e7519b114ad69167d01b914cd5ff : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6881d88f1e7d2d07a941c037104d74d524e21c63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 17:28:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e65839ef00ece44d3b3f2469b92149fa4680edb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 19:19:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93d7187d8b75464fb5e9da4a9911c392 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c97b9cf6dfb9498cace8fb3c2120c044 : SUCCESS in 4m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/46645507c6274fabb44ec3d10882f1a5 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f77f0c76fa7f4b1b8c7c76f69e5eacbd : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b619bd14e6445d0b2e7522d8afcc2d1 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df5d1d41d99a4bb7995eed9d0462f4b1 : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/07f74982f2074da5b5f89f1452316f4f : SUCCESS in 8m 31s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/9a1ecb2b4b5b431c983a71a443190ac1 : SUCCESS in 9m 33s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/76231c71df5b41d29b40a0ddea27b651 : SUCCESS in 9m 41s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9d39b5096e814cf9b11f3eb1b0b2bd8f : SUCCESS in 9m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2b84e80775fc475e933f86de59913f6a : SUCCESS in 5m 57s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/9f05df3bfeee43419564d79d8ce8de1e : SUCCESS in 6m 40s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0fa790ed5d6e47dc8904ed72f5bf0d55 : SUCCESS in 10m 48s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/3253a340521545fab7378059f488eb4a : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5999045400684dd5b5ab0adeaea1931b : SUCCESS in 6m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fea20d1440db46e2bc13c6f7ffbbee8e : SUCCESS in 21m 08s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/48db48d3af974114befd8ebfb0972108 : SUCCESS in 33m 13s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/62aa45fcbfff4fae9c9b24a5b28d548c : SUCCESS in 20m 38s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/2a256fd85a6c433f926405950046d03b : SUCCESS in 18m 08s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ba52835e62c44698bfa32c3cd56b3c53 : SUCCESS in 13m 34s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/b92824cec3534702a70125174fc31b54 : SUCCESS in 13m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f08ffbbf7c184464bd481716ebba0b06 : SUCCESS in 19m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4a1f252491c649dbb9b88657224ecff5 : SUCCESS in 16m 51s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/8574fc722cd745d8bb7181b63bd6891d : SUCCESS in 9m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/467888caa2fa4b5fa2a67749a1f63403 : SUCCESS in 1h 03m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bcea658bab4d4c32b106a9e7a6cd8d60 : SUCCESS in 1h 18m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8a443ff41a7646f7a7c750541b47c127 : SUCCESS in 49m 08s","accounts_in_message":[],"_revision_number":1},{"id":"748e2587bf59f2ef7e4d50bcfa4765bfddaabee9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 19:19:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6649d86c8e53f510009720e820882fd8a7a0b49b","revisions":{"6649d86c8e53f510009720e820882fd8a7a0b49b":{"kind":"REWORK","_number":1,"created":"2020-10-07 13:26:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/10/756510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/756510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/756510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/756510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/756510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/756510/1"}}},"commit":{"parents":[{"commit":"a1d12a52ae060b471fa5bbfff48f972d0e1b1be2","subject":"update constraint for os-brick to new release 2.8.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1d12a52ae060b471fa5bbfff48f972d0e1b1be2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-07 13:26:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-07 13:26:20.000000000","tz":0},"subject":"update constraint for python-heatclient to new release 1.17.1","message":"update constraint for python-heatclient to new release 1.17.1\n\nChange-Id: I01f53acca86ecd47f85bcf5d73ee985d29e91ba2\nmeta:version: 1.17.1\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: ricolin \u003crico.lin.guanyu@gmail.com\u003e\nmeta:release:Commit: ricolin \u003crico.lin.guanyu@gmail.com\u003e\nmeta:release:Change-Id: I2c170cf3e75c18a803931c2da24f8775bfec57ca\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6649d86c8e53f510009720e820882fd8a7a0b49b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6649d86c8e53f510009720e820882fd8a7a0b49b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
