)]}'
{"id":"openstack%2Frequirements~756512","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ib08a15ded1af8775f6e4177b2041f8563e71335c","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Ib08a15ded1af8775f6e4177b2041f8563e71335c","subject":"update constraint for python-heatclient to new release 1.18.1","status":"MERGED","created":"2020-10-07 13:31:23.000000000","updated":"2020-10-07 21:49:16.000000000","submitted":"2020-10-07 21:49:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756512-1602107356766-8f1aa1ae","meta_rev_id":"ac461a3af5141047479a7d093a0327e5f10b8400","_number":756512,"virtual_id_number":756512,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-07 21:49:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-07 14:03:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-10-07 14:56:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-10-07 14:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-10-07 14:03:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-07 14:02:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-07 14:03:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-07 14:56:58.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-07 21:49:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f63a9e74bf23b716d68a4ad9342cd7defa84784","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-07 13:31:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41ec4a9629e16b8601f034412fec1eb4c231249d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-07 14:02:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff7a85f90f6a90d111a17e9b12ccc2e1b571e0e5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-07 14:03:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3120bca2f731e69e9ad6ea6ca0043eff4262d5d4","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-07 14:56:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3b073b65ae8ef4cd3d7a65ebbaab82f27adf7af6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 17:41:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdd41422be50413dbbd0c5ab16395bfe : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c155bcec3ac4b8f9d99c9f9c462614b : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86c31fbde48848cd9f435a1ffc4c6a61 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c5c7888c674479da6f8bb40898abef6 : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e1cc133498c84e37906a1100423997a1 : SUCCESS in 5m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be5d2235475c42c8a553655a67a882df : SUCCESS in 4m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3efdbb215ff442e5919fd3a66ae72d6e : SUCCESS in 17m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2eb0e6cd7ecc418eb4b3596e504985cd : SUCCESS in 7m 31s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c0cb9f743ba547a4ab8f5bab9f263a3f : SUCCESS in 9m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/79d6378aea0a4e14a7ec5642933036fd : SUCCESS in 10m 45s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3707213a4fbe434db8c25ff5456acb32 : SUCCESS in 5m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bb423a2ce35941009fff53a088409e8b : SUCCESS in 5m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/286e0073e73e4a49a6b095a7514d1e05 : SUCCESS in 8m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b9bc1c710b034fc5b18e85661cb9c680 : SUCCESS in 8m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e41aa897e224723bc1104056c471673 : SUCCESS in 9m 11s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/724c57299d4c42d1b7c56a25f755398d : SUCCESS in 41m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2026aa1549d747219917b64a1bdc098a : SUCCESS in 13m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/87d5744b4bdc40c98cbcbe24f6f4a521 : SUCCESS in 16m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4782cc48ff0e4cbe87bdf9ed127b035f : SUCCESS in 16m 23s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dba9d9e6f87d47d3a3e57e1a97cf819b : SUCCESS in 14m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/69d60e1e8bc14bc1af5b4eaf16dcd667 : SUCCESS in 20m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/49700cab6bfd4d74802c969538280504 : SUCCESS in 22m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ec3685b7cdeb4cacab004b07abdf6ce7 : SUCCESS in 21m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4719ca6de75240caab0fa0af78385a54 : SUCCESS in 6m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e7fc925a922543afb41447b7ad75583a : SUCCESS in 15m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6ff0af90ad0a45c2a153b4cc3ad1f13f : SUCCESS in 14m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/22368e79cc2341eba8badc27fae82c0f : SUCCESS in 1h 08m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3f55d87c06784d8da25131f30dab940c : SUCCESS in 1h 22m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4da327156f71451ea8740e107ac0c1f3 : SUCCESS in 51m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3054b2039a0407e49441759a4a96db8553d91f8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 17:42:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba5610fef9dc47a983e0fe8f06d12a1b17a16bd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 21:49:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f39530e3ed740582b56753bd195a1059923178da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 21:49:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e63dd2f52c54c898bcfe7c45faf3ced : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/906032610e684204a09f225b89f6d06e : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fc4e689dd214dac9d864e3e6707f721 : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce23a320f56547958d5c8789e27bb4bd : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f70ee37575de481fb4e062045f3b841f : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e701184939d1429a92a34aba0f50fce2 : SUCCESS in 4m 48s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e8cb1d985870499697992e2a76429cf5 : SUCCESS in 8m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/51a3e8e2a14d4c08a0118957f8f0c25d : SUCCESS in 8m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0cec1fcd9ca7485e83e1514886fbecdf : SUCCESS in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/28be5d35a4694b2d9cee616b8dd33af4 : SUCCESS in 9m 20s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4f13343105d6463295f94111e54c60b3 : SUCCESS in 5m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1d0161481fab47859208043adfe53f61 : SUCCESS in 5m 10s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f88a12a5d9c94e81845abe3b4e3769c2 : SUCCESS in 10m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b7787953aaca43ac82c1c50e476d1f20 : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6bcb241e44e44eb7a60958ed9fad3be8 : SUCCESS in 6m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/343464c363404f348d59cba66703cf86 : SUCCESS in 18m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dbba6da1bcb54e1aa96391f0bff2053c : SUCCESS in 16m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2b500bb9f89d46d4ae89a16f16b59cdb : SUCCESS in 21m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4c8539463a9642be82af12bd4ed74358 : SUCCESS in 18m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9142c9ca96e540c293251f44f82357db : SUCCESS in 24m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/83b6c93aaa88486db53aebfaf0702fef : SUCCESS in 16m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/10b60fa2e5c64fe9bd5e48ced78d1626 : SUCCESS in 11m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7abe6b5f3e58433d908fcae0a8fe06ba : SUCCESS in 17m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7fe012508dd04dd58a0a0e4cc6a85e4f : SUCCESS in 5m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/16ebd25afca94ba0b609e023a2417fb1 : SUCCESS in 9m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3a43abff9c8f4e178723586ced75d2f4 : SUCCESS in 8m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f0049413a164678940d3720ab5b051a : SUCCESS in 1h 10m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/13beb173754946c8bad7dc2c15d36b81 : SUCCESS in 1h 36m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/355b84727fe54e0db311465fc4a5b3a8 : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2836c70f359d600a850e1771e59f7a34e2c40c85","revisions":{"2836c70f359d600a850e1771e59f7a34e2c40c85":{"kind":"REWORK","_number":1,"created":"2020-10-07 13:31:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/12/756512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/12/756512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/756512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/756512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/756512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/12/756512/1"}}},"commit":{"parents":[{"commit":"ceef2d7da6aa88f427a2bf40b095d26ac1d32152","subject":"update constraint for oslo.privsep to new release 1.33.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ceef2d7da6aa88f427a2bf40b095d26ac1d32152"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-07 13:31:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-07 13:31:18.000000000","tz":0},"subject":"update constraint for python-heatclient to new release 1.18.1","message":"update constraint for python-heatclient to new release 1.18.1\n\nChange-Id: Ib08a15ded1af8775f6e4177b2041f8563e71335c\nmeta:version: 1.18.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: ricolin \u003crico.lin.guanyu@gmail.com\u003e\nmeta:release:Commit: ricolin \u003crico.lin.guanyu@gmail.com\u003e\nmeta:release:Change-Id: I2c170cf3e75c18a803931c2da24f8775bfec57ca\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2836c70f359d600a850e1771e59f7a34e2c40c85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2836c70f359d600a850e1771e59f7a34e2c40c85"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
