)]}'
{"id":"openstack%2Frequirements~756529","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ida57c8559621f42c8fcfcc8bf8a247959e4ecaa6","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ida57c8559621f42c8fcfcc8bf8a247959e4ecaa6","subject":"update constraint for neutron to new release 14.4.0","status":"MERGED","created":"2020-10-07 14:50:28.000000000","updated":"2020-10-10 22:31:48.000000000","submitted":"2020-10-10 22:31:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756529-1602369108520-3e075d27","meta_rev_id":"b21b333835334e97bd905443a34144e0f5cfde64","_number":756529,"virtual_id_number":756529,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-10 22:31:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-07 14:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-10 19:22:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-10-09 14:35:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-10-07 15:19:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":1,"date":"2020-10-07 15:19:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-07 14:58:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-07 15:19:51.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-09 14:35:35.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-10 19:22:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-10 22:31:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b8d4db0ff015be58125331331dfb2187c04f94af","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-07 14:50:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c221e3d5c531ba74c607fc4001350bda75fdb091","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-07 14:58:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"301d663e02550df46059b66dc4b1ae0597827a20","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-07 15:19:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6cf4a41d9f2d818a99bb240e27ff3042c741effc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 19:12:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e2f30e62bf7429289cc776fff8e1d5d : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b91f51fcbab4b628004250d585f9f78 : SUCCESS in 3m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3e2e78af45244b1f85fbaa693566817f : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8962921b88764db891a6aac8023fd11d : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e35a3e6c875343b0b0217567a3d65536 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/817f86f05d8341fca5badd8c8d37b5e2 : SUCCESS in 3m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5a7d207a61514dc784f65874cda8efab : SUCCESS in 7m 38s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/cf080864dd0f4515b8d266a52fdb9676 : SUCCESS in 20m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/597145bb2f2a4549b24ef477e8c3217c : SUCCESS in 7m 41s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/74678df9853b46eab37acd08d9b35df2 : SUCCESS in 10m 09s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/41ef0d52c9e54997a98753d9dd81519d : SUCCESS in 8m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0c7f89ecdae64c1cbbb382723d693313 : SUCCESS in 5m 19s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/43bd55f9e1284dadb63349d05d3a27f0 : SUCCESS in 4m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/32a955e60d934d63b9825c1d2b49e448 : SUCCESS in 11m 19s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/961dc8a5a8574f00b70d986fd2761f66 : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/04b0ce99dfd64796978d2656c1077570 : SUCCESS in 7m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6db1eb847ef441f39872cc66c4d3c56d : SUCCESS in 19m 37s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/eaa89f10b71e4eb58d92955459a46ff8 : SUCCESS in 36m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d4092748abdb47148aa7dbf7e24fdfbc : SUCCESS in 27m 44s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/d6418b2d26d840f79b6f76e265587670 : SUCCESS in 30m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d31d1823b53843c3ab06f445f2b5140d : SUCCESS in 13m 39s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/b697bd25c4a743268b948616d520adb8 : SUCCESS in 13m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2d9b030dd7d435eb9ca99f30f14ea46 : SUCCESS in 17m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/32c8e308e9924be58bc3c5c84431eda5 : SUCCESS in 7m 28s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e9a9453904d8412d90dc8dc720c7c25c : SUCCESS in 7m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3a5a6b8dfe2345ddacdbbd3ae730913a : SUCCESS in 1h 07m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/29f866895b1444019ddbd6c356002d63 : SUCCESS in 1h 16m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c284d8e61ff04701ac944f21a8453d27 : SUCCESS in 57m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e22c9d09d3bc43233e08a0b81e872e31945a9a3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 19:13:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57f763832a5833b94a81cf91cf6237a34d85b1fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 21:49:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96051815cbbf4c668c961feb71ab2bb5 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1458f7e946894e6b847f268d962e9a1c : SUCCESS in 5m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3663266c45a84c5eb09b5d60352b2aef : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c71188a399c9424eb645f691f4538f64 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/875bc39f202c417393fa13eeb3bdacdf : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/13e6e5064a324c498a37d74d8858825c : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1a0e9fffcfd340a4a55844ef1e17d532 : SUCCESS in 6m 31s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/de960a943da740cea27d71e4a26058cd : SUCCESS in 8m 04s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fd18788389c34b0da49512bbad592e39 : SUCCESS in 9m 22s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/0779c06eca744953be796acf5da1bfc2 : SUCCESS in 9m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e201eb40a6f7482cb10e8420af79bc0d : SUCCESS in 4m 56s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/a06b0a42f935481cbba7be58d8f1493a : SUCCESS in 4m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a251649ae29143b19bc738dc4410ef3a : SUCCESS in 10m 24s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/eb1b2858241445dfb338be541f762e26 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7faf45dd2019451f82961da9a3dca9b6 : SUCCESS in 6m 34s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7608f043ab444c56b621af7dd3a9f74c : SUCCESS in 21m 52s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/4dfa4c0d98c84acf9307354b4b063c07 : SUCCESS in 30m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2b4ea15dd79c4478bf62610341a02489 : SUCCESS in 25m 26s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/ab86c7867c1b454c9e11552d492c6fb3 : SUCCESS in 26m 01s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c51428be6ebc4007b79bdfb7ea4dfa2c : SUCCESS in 15m 32s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/2dd2226894094c55b4d39478e9838216 : SUCCESS in 13m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b617e66ae5e5433a9d0c59cb325bdaeb : SUCCESS in 19m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/35ba4361307b4cd19b65e05e6e4542aa : SUCCESS in 8m 37s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/70958e8c063440d6a10408b11b339920 : SUCCESS in 9m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a4e7573ca0c455191aa7e7576a10054 : SUCCESS in 1h 11m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7f0897b7798c43dc890b08cd2fedcf0c : FAILURE in 1h 15m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/139aa243931f4a4f9fab6b65eaa6d2a4 : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":1},{"id":"7fd429503de13eb39bd1fca9c24ff226b9415313","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-07 21:52:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0576de907c4b3dd1194c57e39bc4f5f1f4e581ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 00:09:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b3803e0a63e445eac8a7954adb8397d : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b545caae1c2647b0aaa61c3e24351bb8 : SUCCESS in 3m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6c50dab0109d4559927e2d316d5822c3 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e36ed6bd3edd4c92be3a307f4510c896 : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5e682e720bb4e009557fbeb4ddc2866 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/042a51a4feb74713ad1b9d1c9e069f3c : SUCCESS in 3m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/25b5f5c7503d4ef6b908a97345fe14cf : SUCCESS in 9m 13s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/7ce8f86016c7460087b053e7adb1c053 : SUCCESS in 12m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/383f6241df2c4660ae08ebe61ebfe303 : SUCCESS in 8m 49s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3ed3846baa054e19bcb1cfa5244f7b38 : SUCCESS in 9m 49s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/1367c670b0694e5fbf5146219bd9b7cc : SUCCESS in 10m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/018a910a56824bc2bedc9912a8d8f7d4 : SUCCESS in 5m 14s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/381352ce47384e499757f5230b08c473 : SUCCESS in 5m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0d61b2a6ab494eb697a57b41bad6ba0b : SUCCESS in 10m 04s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/c09bd863f68d4d44bdcee3a5280b5e04 : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e6b2c5b3edad40f4bd24d44b852b2004 : SUCCESS in 6m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/09c5c51cd1b64397b38b7c228bda312e : SUCCESS in 18m 39s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/7abd56fe4bc7448183e18f419829a30f : TIMED_OUT in 42m 06s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3b5ef1b6a86f4777889bf135a1d11c63 : SUCCESS in 21m 07s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/cd4b66f713654f4197a345cf3144ad4c : SUCCESS in 25m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a063bec6ad484ce5b3fcc17d1fbc3805 : SUCCESS in 13m 21s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/109efe483d524483a94cf149cd188be9 : SUCCESS in 13m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/32141c24fc504f0e9be806a7f851908b : SUCCESS in 15m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5bbbf25fd2bf43ebbdbf8b890a2240df : SUCCESS in 8m 34s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/db9642ced9694d68b3c91b82bfe5da99 : SUCCESS in 12m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4dac41219ea4047b5c957fc90d02540 : SUCCESS in 1h 03m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ee2aad59d8d94d78acac3d05065fa372 : SUCCESS in 1h 52m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6914f69b297049438e0723c38f7c6e7e : SUCCESS in 50m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a724a7c7b245062a18bbc47fc0a61fb8f3798eb0","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-09 14:35:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"834cc162eb6983ec432a8eb4adc065803702f1d7","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-09 14:35:35.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"e5c1e9d8110d8e39b76c64f27006aec3fe07213e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 06:34:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61a4318a82e3406e95c68418ebb51ee4 : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e5507915f0474cfc880dc6cdb46c62aa : SUCCESS in 5m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fb5715d62f1e4c588ebad75778201125 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3c7271a51af4a7ba93cad6cb5c43e4f : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45f82e43b2754e939e81056761dd1f16 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ae8909230444fd2aa660e6a4822a5c5 : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1bd46502295a420d80294688299b0aef : SUCCESS in 7m 39s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3cb5f03d91854a9f87450e6bf5800b7f : SUCCESS in 8m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/09996c14026640248401a102cf6aa079 : SUCCESS in 8m 44s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/811c46118edb4059aeaec8b2d91ccb04 : SUCCESS in 16m 32s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/bf955684d3bc4af99e96bc51306a10df : SUCCESS in 9m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/095893d4c0d04e1b92c8ee73cab5a8ff : SUCCESS in 4m 43s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b344127843df4a02af02be7339da276d : SUCCESS in 4m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/69efcb37bc104f04891d1331049fcfc6 : SUCCESS in 10m 11s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/d2b59cdb1d304a3a8e8638983f7d18b7 : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/61b97e0747944024854cc893fe445f03 : SUCCESS in 6m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8d4e747f8b0c4d9f841e9209bf1c932c : SUCCESS in 20m 17s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/11e74fafef2f42a9b71704f624bb3859 : SUCCESS in 27m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2a000c1de18a4be589bc431186d49803 : SUCCESS in 22m 47s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/7c07526270f64f5bb5d28491cd34d5a4 : SUCCESS in 25m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9dee37cf42124b9f8bd27ee227851ac8 : SUCCESS in 12m 07s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/90ae383c28574cb6a62b26832c9a4126 : SUCCESS in 14m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/032277829015444fb6e2543234eb8c36 : SUCCESS in 21m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/37706b2652fb407cbc4b9ac1a589bef5 : FAILURE in 8m 12s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b6a5757f47d34b17aa60fbb5fec4c57c : SUCCESS in 7m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d673b96b74d64917b99c24f78160c54d : SUCCESS in 1h 33m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/af135ffc512448e89964036aeccabbb4 : SUCCESS in 1h 22m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ba9bfe72f28f4b0099f2585e6a3a6eac : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":1},{"id":"96c2b35d7718067b845fa0f336cc58618246bc2f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-10 19:22:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"430c50bbdc75b865e99818f4e4913a40a920f349","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-10 19:22:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"46426bba3d35c2e120c2e41382dcfb8aa302de0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 20:47:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/859ee84b23234043b639be569e463dce : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e631ddb3124149f1a17543f4df3ec3ce : SUCCESS in 3m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8b0321e37f63488284b51c0e2514074f : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d5442ebd18e42e483d5b22610f3a8d1 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93f68491817d45b7970ee1bf4cc5e866 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc555a63bc794de4b2a48869995185b5 : SUCCESS in 2m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fe89567f22d547bab8e9cd01876d479d : SUCCESS in 6m 13s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/c0b96e2353354c6489de84e16ff60a83 : SUCCESS in 8m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2155692b015c4b97b50797b3148d4f32 : SUCCESS in 6m 49s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e0253a2f00394f459ebecf4bb9ad7a14 : SUCCESS in 8m 11s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/69596d6f940e4200ac2a99c6f223d32c : SUCCESS in 8m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/041b8a2cd904469aa8e17560032e3d9e : SUCCESS in 4m 43s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/c1f41b60572244ff88168fef5b266dfd : SUCCESS in 3m 58s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ab611036c355461d87c8d46954fa2bc7 : SUCCESS in 11m 03s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/f2cc25c8dffd498a83f1d446b7c5e28e : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9336325ecde94076a471946d841ca4b7 : SUCCESS in 5m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/45a5a23bbc1b42a5b5e94dc6bc7a32e1 : SUCCESS in 21m 30s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/1b872cb2f45042e3970796fb067c3e00 : SUCCESS in 31m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/74248b3e3c2245d596cf1ca07b32e53b : SUCCESS in 25m 58s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/d2f8183b77704a44ae3b6c35525201ef : SUCCESS in 23m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1fdfc08e96324be4ba92843e1c8e469d : SUCCESS in 12m 02s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/8a3cc379bc634cb2ae84cced27b42f35 : SUCCESS in 11m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0fa6d30b0e9436a9542045ca6b11b2a : SUCCESS in 16m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8965f7a1c46f4577bfdf17184ca83b76 : SUCCESS in 7m 19s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/0c636112fbcc4ab4bab12d94f442b6fc : SUCCESS in 7m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c6b9a295b6f7442daa5b799dd122f698 : SUCCESS in 1h 16m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9cbfd7f997964c9a93ba8f10ea0ba632 : SUCCESS in 1h 18m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e2a4e48f8da1407a93a094c86e99eb2e : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":1},{"id":"788002f6a1c9b1a2d11c0f05193d31a3c557eb98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 20:47:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20753f4e28804adc38c879ee880472c0da044920","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 22:31:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"afe6fde565b3c357238fcb15514dfe1a3e93ddb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 22:31:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2d6cd3f064e4b04b80a9307c041c78e : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e5c837e8e6164074a8ac091e42178cb4 : SUCCESS in 4m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7d7695ac348647038226e80fff038ac8 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09ee2a259c174e45a30eb553e00e5b42 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3fd73d69dc9646fc80425c649ca2f4ef : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ac7dd4407c747729ba88e15b51a65eb : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/40af9db0e52e494da277002c78545604 : SUCCESS in 10m 49s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/75c32e9df1384fbfbca47fc7893aa870 : SUCCESS in 6m 53s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/64aa1dc656ed44c287f1172a08c56382 : SUCCESS in 9m 21s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/a46c601aa8484202a044bb2d2d6d8fde : SUCCESS in 8m 05s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2561832eac4740f9ad45352d4f988661 : SUCCESS in 5m 09s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/10a24ccb780b46c8984caebcc1978f3a : SUCCESS in 3m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ea4e71e0cfcf40019b5b6ac4c6ed8560 : SUCCESS in 9m 42s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/6fbd682ade894d3e90eaa00415b710ef : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b1dea7db14e341cd96f34a4b2124e593 : SUCCESS in 5m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0c2bc1c2f6c94d19b470866a86535961 : SUCCESS in 23m 15s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/ca4f8010ab724f6da2e63d694b3f2ac4 : SUCCESS in 25m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/767e07465fef49cd9a7681b2e8062cf2 : SUCCESS in 20m 43s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/f6829eaa66b4480184d7e85ea492e518 : SUCCESS in 17m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e2ad59b96b1c4b859fb3d86141be18bb : SUCCESS in 12m 15s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/70f83c20fa334ab89160beaa371cdd05 : SUCCESS in 13m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/347584cd615f43e1bc04c609b830ab7e : SUCCESS in 16m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a62f5dcec1ea4885ae77e19ef61e60d9 : SUCCESS in 7m 40s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/dd005533da25410fb161ff72f55f6df0 : SUCCESS in 8m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb29d45b38ea416b9a0328406d98482e : SUCCESS in 1h 10m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5052d63448e24c869870dacc3f8e55e3 : SUCCESS in 1h 25m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3e248ef931404d61b06bdd4c6fef4892 : SUCCESS in 53m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5f6ef94d8ec2cc41ae70fa1d0b2d5fbf7139a77","revisions":{"a5f6ef94d8ec2cc41ae70fa1d0b2d5fbf7139a77":{"kind":"REWORK","_number":1,"created":"2020-10-07 14:50:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/756529/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/756529/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/756529/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/756529/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/756529/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/756529/1"}}},"commit":{"parents":[{"commit":"a1d12a52ae060b471fa5bbfff48f972d0e1b1be2","subject":"update constraint for os-brick to new release 2.8.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1d12a52ae060b471fa5bbfff48f972d0e1b1be2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-07 14:50:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-07 14:50:23.000000000","tz":0},"subject":"update constraint for neutron to new release 14.4.0","message":"update constraint for neutron to new release 14.4.0\n\nChange-Id: Ida57c8559621f42c8fcfcc8bf8a247959e4ecaa6\nmeta:version: 14.4.0\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta:release:Commit: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta:release:Change-Id: I44798b084491a6b381651b5667071c68bcb07a81\nmeta:release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5f6ef94d8ec2cc41ae70fa1d0b2d5fbf7139a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5f6ef94d8ec2cc41ae70fa1d0b2d5fbf7139a77"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
