)]}'
{"id":"openstack%2Frequirements~757571","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I6ea212a06452aa7daa39b19ba697c8803d85f663","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I6ea212a06452aa7daa39b19ba697c8803d85f663","subject":"update constraint for oslo.privsep to new release 1.33.5","status":"MERGED","created":"2020-10-12 13:49:33.000000000","updated":"2020-10-13 03:42:50.000000000","submitted":"2020-10-13 03:42:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757571-1602560570530-75d80b9d","meta_rev_id":"8c8c89b3eec849d03b0f82db2730f5eb92ea6482","_number":757571,"virtual_id_number":757571,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-13 03:42:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-12 13:50:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-12 16:13:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-10-12 13:50:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-10-12 13:50:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 13:50:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-12 13:50:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-12 16:13:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-13 03:42:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e21d4c88c1b741fa6e109465a7d32ca60289ae7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-12 13:49:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d66d3d2b1c3aba46a7d408015f0c2e88b536a445","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-12 13:50:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d754a7e25e9ecff3f6cc31f9ed5dc5c74ac89da","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-12 13:50:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d3cfbd2805bb87b4b91a0b7fa377a0578c43135","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-12 16:13:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d1af1a092521c076ae159886c0fc55ccf1fb1e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 01:09:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0733bda4ab141d8aef8215f22d15153 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab5e6164f1694aa9936ffe0dfb8bd712 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0987b6d6644c40269e72fe91c56df8c0 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/08987a8dd4524302afe2b1a6eabd989a : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12ba3f9f14f747b986b5ae6800ce9395 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4c7ac743377448f399bd8e1fe82d5c14 : SUCCESS in 3m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/670742dfa0cf466397452238da49ea4b : SUCCESS in 6m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/160968577dcd44289b45d23094e10269 : SUCCESS in 6m 05s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/201f73b2285f4e9fad8e9437cf6e1c43 : SUCCESS in 9m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4c97198d448645479a566e479c78d253 : SUCCESS in 8m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7d21acc2ec9649a897fcbb386da8d81f : SUCCESS in 4m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/61666c1b76c04b2cb82fb9f3d977f903 : SUCCESS in 5m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9b61ae14397243aebc2357b097402b13 : SUCCESS in 10m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6a7f47f687084b0bb718d73761ca157f : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b8d97241f8da48bfb18e7f066ad182dc : SUCCESS in 6m 11s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/63aaabd2c93943b1adc82f86b6f5285c : SUCCESS in 12m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a8c7784c01934f26a0bd3c64f790301f : SUCCESS in 17m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/eece8f0bc4f244b5886de3add2876281 : SUCCESS in 22m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0e71c4cff2d8434c867827c003c791f6 : SUCCESS in 22m 21s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6824d7d452fc4ced99012812c78154ba : SUCCESS in 11m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9c1324b3de4e458b9ecb0afabe90628c : SUCCESS in 11m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b75cfb0a846f445fac0e188b9fd372b2 : SUCCESS in 14m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2f0ad0d72e294ca59e6f7618475cadba : SUCCESS in 17m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d073cab6959d45ceb37ff1e0ca92714f : SUCCESS in 5m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d417a2852f0f4692a8614aef534279c8 : SUCCESS in 8m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6138655bdebc4018a46595fca9d3fd02 : SUCCESS in 9m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d8692bef5e943d49b725e00df9fb967 : SUCCESS in 1h 09m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/deca50f6012e435fa515b7813bec2375 : SUCCESS in 1h 21m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/48311e28a8a6437881ddeb57a02996bb : SUCCESS in 56m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ba35f3c91a9e8056448ed025a57338ba0da6e55f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 01:10:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5465be8d3fb2f2b39896f7cf6452036e79c2ad66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 03:42:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"14ff0f4b70663fa65b765d188ebd0e4f03dd853d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 03:42:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e2378c3da14113af2e6d19e76f2c59 : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/470d532704ae48ba9cf99da405dd8016 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e764ec4619a64a90897e863a877cd764 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e563e96bc604ec69ebc6d85dd727d61 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a15478f5e484058971aaaae3f41867b : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/14614c1a5ac1449d94f78dce7438b08c : SUCCESS in 3m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d3bde11d34ab42ce9ff6fe2e02ed44a4 : SUCCESS in 9m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/35192965758c40df830e79668d9ac5eb : SUCCESS in 8m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8510c4fa0758462e87f2836564bae736 : SUCCESS in 9m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ca337268da624a60b77d42185f9ee118 : SUCCESS in 10m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/94456f7b48b241ce82743db3abe19349 : SUCCESS in 5m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/893c235a8de14db2bc88721fbf1d78e4 : SUCCESS in 7m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/572ae612220b4cecac5adbb338f5c70c : SUCCESS in 7m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7f2fd653601b431ca4b73a419abe95b7 : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79bc0cf013e846a89ffc7f4615681e94 : SUCCESS in 7m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4feefbe4bc3e4e87883fc9246b670efe : SUCCESS in 13m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bab9d8c115144bc59f2f972f8d06ece8 : SUCCESS in 11m 57s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/655e7891377c47bfb6fc10c934f3df9f : SUCCESS in 23m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eadad237f91f4ab0acd80c93378a49c0 : SUCCESS in 20m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4a73b00b19c1449cb13994351a833a63 : SUCCESS in 13m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f1809e64a3a446a4bf79ab0ac95a14c5 : SUCCESS in 12m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/47ce71e48ac949889ddec1c682f97be8 : SUCCESS in 14m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/114ebe13e3c0449ab5a065b6ff2aef42 : SUCCESS in 18m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a305e38cd5b44b4ba5e0af7255176856 : SUCCESS in 6m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8f364eec3872454bb13487bfb064c9be : SUCCESS in 9m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aab85aaffe5844cbac48483525cdb129 : SUCCESS in 10m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2e7d34768d14dc4a6997fd66866c884 : SUCCESS in 1h 06m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/759a3f0d9f6143cdb65e5b004e3f3840 : SUCCESS in 1h 32m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1b969fe85b01478fac2a01a44249916b : SUCCESS in 47m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f58f1c01422f07cc2a858831e91c8b433d9f5ae1","revisions":{"f58f1c01422f07cc2a858831e91c8b433d9f5ae1":{"kind":"REWORK","_number":1,"created":"2020-10-12 13:49:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/757571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/757571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/757571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/757571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/757571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/757571/1"}}},"commit":{"parents":[{"commit":"32ffe6bfec0267078d0e3993d5de2ddae46f1ae4","subject":"Merge \"update constraint for neutron to new release 15.3.0\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32ffe6bfec0267078d0e3993d5de2ddae46f1ae4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-12 13:49:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-12 13:49:31.000000000","tz":0},"subject":"update constraint for oslo.privsep to new release 1.33.5","message":"update constraint for oslo.privsep to new release 1.33.5\n\nChange-Id: I6ea212a06452aa7daa39b19ba697c8803d85f663\nmeta:version: 1.33.5\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Change-Id: I0e49a976c5d5c8be9724cd28a5c8d1f0bfd31ec2\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f58f1c01422f07cc2a858831e91c8b433d9f5ae1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f58f1c01422f07cc2a858831e91c8b433d9f5ae1"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
