)]}'
{"id":"openstack%2Frequirements~758487","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I39e6e56995437385ed8b5b6fe0711ad0df324133","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I39e6e56995437385ed8b5b6fe0711ad0df324133","subject":"update constraint for ovsdbapp to new release 1.6.0","status":"MERGED","created":"2020-10-15 19:19:13.000000000","updated":"2020-10-19 21:39:37.000000000","submitted":"2020-10-19 21:39:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758487-1603143577704-61dee02a","meta_rev_id":"6bf7d83bdc97078a2b2e31ad6f019b750249e180","_number":758487,"virtual_id_number":758487,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 21:39:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-15 19:26:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-15 19:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-10-15 19:26:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 19:26:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-15 19:26:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-19 21:39:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40b8345a97c9da270d9c10b0213ffed8ea441ca8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-15 19:19:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50273ee8ba351357a363e70120970a8dcb0cd15d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-15 19:26:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fca912abc6d08543f41826651c8fafa48432e9c6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-15 19:26:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e8adb51226ee3dd7908f03c7afc339db22858d26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 19:36:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2628bf3010dc4bce95a11de6c4eb08e7 : SUCCESS in 11m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ce69d4d2d821b9c3afa1db91fd802781c3a100e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 22:27:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0532bd3bc96443a49f99f663979789b7 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b473ceb59b6b4dad81e4ef737c24e896 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbb5c570d01b4ea6aba75911647a5469 : SUCCESS in 5m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5309d9880b5e458e93ad840c2d0601be : SUCCESS in 5m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71b8f8b9ee2249358bcfa981b5f76b70 : SUCCESS in 5m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cff61e97179144e8b82578154eb8db5a : SUCCESS in 9m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f5003de5c86d4bb5b9b8b0760d2b1c0d : SUCCESS in 9m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bdd6f375d996406f893d73af74c2e383 : SUCCESS in 11m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1001ca7db6864371a03fdb72eb301f0c : SUCCESS in 8m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b07e53d0a3324e329b7849f3ed827c53 : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/be68f4efd9a74095b00eeb9d127ab84a : SUCCESS in 8m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/80277e27dfde4b65b2e82a53e9c10bdc : SUCCESS in 9m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f63b90a56dd64ba79309f4fd90927d3b : SUCCESS in 17m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9bdbd80cb6eb43a6a5ac6507306d18b4 : SUCCESS in 19m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/58a8dd3b74224b779cf2b5f10eaf38a6 : SUCCESS in 22m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d62281c5bad74baca823d18751e35391 : SUCCESS in 5m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a99cd7ab8d4d4eaba426a783aff495d8 : SUCCESS in 16m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4885478f987c4fb9a0145814e434a6e3 : SUCCESS in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a51d0fbc315f477a9b1e165a1d044ae6 : SUCCESS in 7m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2f00314e1de9498b864e3ebc6a8f4543 : SUCCESS in 11m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b285f3c84df04cc08b1f0f4155a24647 : SUCCESS in 55m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38f5150900544f9d916cb095d9cd6fc8 : SUCCESS in 1h 21m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1b7ff5b2575e50ad9560fe242f114333b8665bf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 22:28:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50b38a13121b8910ce303f7e4eaad818babfe454","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-19 15:08:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b5c09693b7c8b700b43cb0db7586f28413b791ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 15:22:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f756224953a545d9a184c509ff197e9f : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a0ddfb1b14318ae66c4fa057d7ba0c73bcafcc86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 17:22:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a188706263e7446b92fe48e72567ad47 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/937355ac6ea0432189cba0b3b0aa706d : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/636b9dec6051468195179109450509d5 : SUCCESS in 7m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/01cb2d84ce0c4a86bf4b3ec4a3d62538 : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e285ab27242d458ea412f47112ba3ba7 : SUCCESS in 5m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/80e13b80415f49258cb242c9ecd3f174 : SUCCESS in 9m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/42f843c0444842b9ac778479ecf398fa : SUCCESS in 10m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2feb313652d54d4989c56ae6c3fea1ff : SUCCESS in 12m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6c319aebe852484eaac34357be9669bd : SUCCESS in 9m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3c0becb1cce34aefaa14f792a9f472c5 : SUCCESS in 13m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/499e1facb5644325804585a1b9ec0b30 : SUCCESS in 10m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c710f09eade6492983a7fb9f9c935db4 : SUCCESS in 8m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/16f4bb36545048df9699c5e258478737 : SUCCESS in 20m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d78341fa1f6f4ed793a86b4e8a075dd7 : SUCCESS in 26m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/610755d083ff4910a94c712e37fc1adb : SUCCESS in 15m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e23e322e30394e2791e7183399d44a48 : SUCCESS in 6m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb03e0a5f69344e29be187e422e821e7 : SUCCESS in 19m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b698f55615624c30817bc92a21592126 : SUCCESS in 8m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c989d3e760914742a7d61b5192e2c3a7 : SUCCESS in 8m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/588b2ef5b0044dd09b0146881d9f37cc : SUCCESS in 13m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9df57f9a92794a2eb145ae238746f213 : SUCCESS in 1h 02m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b274219ae07b4da58855f142bd1a04f3 : SUCCESS in 1h 24m 06s","accounts_in_message":[],"_revision_number":1},{"id":"922f51883d9dd46daf09c27618b8c05cd06df9e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 17:22:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cea0e5ba66698ab23884f95c6a32001e25d1262e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 21:39:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e56fc7a5887c5c224d5ce63e534d94e2222d3123","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 21:39:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5b70de6146e4333b948b0d2e1841c41 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f491973e355340d9a2dd6ba9ed66a926 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa353fda42f541faa8245ba85a8698a4 : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c59ed4b54c44481e85fe038dc813b186 : SUCCESS in 6m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a62e22ccfb8549929cb0badc7f94ad62 : SUCCESS in 7m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3eda021eca744d54a637f1ed0b812280 : SUCCESS in 10m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e2818c081ab742f481359b7fef5079f1 : SUCCESS in 10m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/25ed4e0c75494c6aa786c615b08bf389 : SUCCESS in 12m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9ddfe3ac92fa404f9a65670379f09185 : SUCCESS in 7m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ad45e16e0eef497fb79d29f54f0380c0 : SUCCESS in 13m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bced659e3bb04981b02a5aba308aa149 : SUCCESS in 9m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e51d7209c11a4e169da2d23adb381f7f : SUCCESS in 9m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/188b398395d94c32b082a0dbbfa8b1ba : SUCCESS in 15m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/176e89fcfdef491d92c51f9f07c039ac : SUCCESS in 26m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ce872ec9cdc84648a2d2d2adf128d452 : SUCCESS in 14m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/90c4151ec83546978b4f5d8eefbf8654 : SUCCESS in 7m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24cc178752964ce2ba13f5ea622c7b88 : SUCCESS in 19m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/db30c3729f324478b567a9e9f3ad673f : SUCCESS in 8m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e23b3f6912fc4144ac115834bf012be7 : SUCCESS in 8m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ef726096abc14fa5ad03133f077b2a7f : SUCCESS in 13m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b92e4c597344d1dad696612624aceab : SUCCESS in 52m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e20bbbea8504d528c3270a4fb7da182 : SUCCESS in 1h 19m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4308c0db9172b0d0b60d3a2c0dd76ca3964ad109","revisions":{"4308c0db9172b0d0b60d3a2c0dd76ca3964ad109":{"kind":"REWORK","_number":1,"created":"2020-10-15 19:19:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/758487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/758487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/758487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/758487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/758487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/758487/1"}}},"commit":{"parents":[{"commit":"4bb07a1163756f90cd0d5cb021b6f1259b08de6d","subject":"update constraint for ceilometer to new release 15.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb07a1163756f90cd0d5cb021b6f1259b08de6d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-15 19:19:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-15 19:19:11.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.6.0","message":"update constraint for ovsdbapp to new release 1.6.0\n\nChange-Id: I39e6e56995437385ed8b5b6fe0711ad0df324133\nmeta:version: 1.6.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta:release:Commit: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta:release:Change-Id: I1d661095847898ad0ab4d85819a6f86ca9e155ff\nmeta:release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Code-Review+1: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4308c0db9172b0d0b60d3a2c0dd76ca3964ad109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4308c0db9172b0d0b60d3a2c0dd76ca3964ad109"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
