)]}'
{"id":"openstack%2Frequirements~758541","triplet_id":"openstack%2Frequirements~master~I051e704851c0d4b192795665d807c6f46561d670","project":"openstack/requirements","branch":"master","topic":"upper-constraints","hashtags":[],"change_id":"I051e704851c0d4b192795665d807c6f46561d670","subject":"update os-collect-config to new release 11.0.1","status":"MERGED","created":"2020-10-16 09:29:15.000000000","updated":"2020-10-16 16:25:45.000000000","submitted":"2020-10-16 16:24:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758541-1602865471547-795c03e7","meta_rev_id":"ba70ef14c6c94a2042e2e155128aaed14709c4ba","_number":758541,"virtual_id_number":758541,"owner":{"_account_id":32471,"name":"hanjiangxue","email":"hanjx27@chinaunicom.cn","username":"hanjiangxue"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-16 16:24:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-10-16 14:08:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-10-16 14:08:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 14:08:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-16 16:24:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8721f3d6a60f132cb47298112bf56c9878338961","author":{"_account_id":32471,"name":"hanjiangxue","email":"hanjx27@chinaunicom.cn","username":"hanjiangxue"},"date":"2020-10-16 09:29:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e50ba91286b193609598561ddda0070d2052457a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 09:46:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1a15b4cd7e4148bfa8fb602d09888a22 : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":1},{"id":"edc9b645e7ac00d849a9ffddb0619407e34f3bdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 10:55:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984f4a79f0c54397986a2c4ef5daef93 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/660ec123f3ae4b64b735756372e7514e : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a729f87a2ea4c9392b02f8a2145651c : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dc8478415a0475fb2c28363362837be : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/264c5bf63db34530abcf247461fe0869 : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d8b89fdb1a14b60b894b81ad1c4f4f0 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c12bed90421147f38a668d8c5b7e85b0 : SUCCESS in 8m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5ebf5667dbfa4dbdabefff00280b380e : SUCCESS in 8m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/da4cabe63b214fa0a917473f7bf89bd0 : SUCCESS in 10m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fe939aa5b36c4046968d2fc83a3fdc19 : SUCCESS in 6m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3c2c7d66153f4d4c8cfcb9b509a7d6de : SUCCESS in 13m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b87bd8265208452ba200fb49bac9da71 : SUCCESS in 7m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/40d9898e27274433a5a854843a19b53f : SUCCESS in 8m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9bfc816b75ee4014819857e195c80ad1 : SUCCESS in 15m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/642b87dd10b64dd0b11c0da91d92f0e9 : SUCCESS in 23m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/420c1b92d46549378369bf578a92e835 : SUCCESS in 12m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4e194bc061f54f0eb598ebc32e66e776 : SUCCESS in 5m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7b26293be75443e1a9bbaf26ed790a99 : SUCCESS in 18m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8bffeb485e234dc5ac2a86c847109126 : SUCCESS in 6m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dcec7d59344e4c5aa0dfb3d9a356c6f5 : SUCCESS in 7m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5918d8d8ab984224bbf2aaaa08b56f22 : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0564e8db81c74a5aa129a87c6c5c28cd : SUCCESS in 55m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24089169c989401d9c070ec8b1fb06c7 : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":1},{"id":"cb840c441e058a20ad3b5bb1f86fe1b457c32ebf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-16 14:08:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"007561483cdf21096fb4f177590da1ff85d813fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 14:09:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2b4e9c3464f7f4763d401277a67620be37e279c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 16:24:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c82992a1953e1095993180c456d7b0fa8cb55199","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 16:24:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be12fa206df64c8bb8b6791f5d162750 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fb6529d134b44cc861be6ca9a5cc27e : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03d9899031cd4815ad785c74a002e8b6 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a9bb1ae39bc4fe797786e81866d9c90 : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48ffa3f460c047e6a0312244a1faca8b : SUCCESS in 6m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/80caf48936c943378f70ea4bdb80e446 : SUCCESS in 6m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/58026ecc3b514144bd54858abdd966df : SUCCESS in 9m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5dfc4ab174c9466ca597e78f12391ede : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1bc4efe267b147dfbb1b468c0d041b39 : SUCCESS in 13m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/abc95f4eb97046a39f986bf30febe038 : SUCCESS in 10m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c4fbf3faea5f42eb927072761f81b2e1 : SUCCESS in 13m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6fae241e56e64da29a9c52703da95a7c : SUCCESS in 9m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3d20dd1c89f54714ac1adaef395f2342 : SUCCESS in 10m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7a5f5a22be7e496898fabc506545295e : SUCCESS in 19m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/750b5c18099d43a1b2bc0c26f2254312 : SUCCESS in 34m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ddbb92949412438e96ef2b141af393e2 : SUCCESS in 14m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd6404716c874f14b1c0287d8a6d5039 : SUCCESS in 8m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5f843997384f46ec9d05db64366b16b5 : SUCCESS in 21m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ad9edf860700453489e88bbd914dbc9c : SUCCESS in 9m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/411325694ce04ceaafcaeacadf7a7f5e : SUCCESS in 9m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1e432487eb154777badd49604af522fd : SUCCESS in 15m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a559bd6972ef4261ba4ea89251cd71eb : SUCCESS in 1h 00m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/552c0272fa71497a845c160e0679fd86 : SUCCESS in 1h 25m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4ac8d4999ae20cbef3f8a878efb62aeda3988232","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 16:25:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46b9f867f4ac4900845ab677c1ad3f4a : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bae4b8943b9b9f28c2fcca207eb95b17d4681211","revisions":{"bae4b8943b9b9f28c2fcca207eb95b17d4681211":{"kind":"REWORK","_number":1,"created":"2020-10-16 09:29:15.000000000","uploader":{"_account_id":32471,"name":"hanjiangxue","email":"hanjx27@chinaunicom.cn","username":"hanjiangxue"},"ref":"refs/changes/41/758541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/758541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/758541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/758541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/758541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/758541/1"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"hanjx27","email":"hanjx27@chinaunicom.cn","date":"2020-10-16 09:28:32.000000000","tz":480},"committer":{"name":"hanjx27","email":"hanjx27@chinaunicom.cn","date":"2020-10-16 09:28:32.000000000","tz":480},"subject":"update os-collect-config to new release 11.0.1","message":"update os-collect-config to new release 11.0.1\n\nChange-Id: I051e704851c0d4b192795665d807c6f46561d670\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bae4b8943b9b9f28c2fcca207eb95b17d4681211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bae4b8943b9b9f28c2fcca207eb95b17d4681211"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
