)]}'
{"id":"openstack%2Frequirements~758546","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I8a635621e5669481c5d465045159f663efa4080a","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I8a635621e5669481c5d465045159f663efa4080a","subject":"update constraint for python-ironicclient to new release 4.4.0","status":"MERGED","created":"2020-10-16 09:59:31.000000000","updated":"2020-10-16 16:27:05.000000000","submitted":"2020-10-16 16:27:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758546-1602865625927-332a2ac9","meta_rev_id":"dfc16346cb8d8950ffe6795c554934226c841724","_number":758546,"virtual_id_number":758546,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-16 16:27:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-10-16 14:08:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-10-16 14:08:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 14:08:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-16 16:27:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4aae5e28856f001cb2c49c285f81a71c375fe255","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-16 09:59:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c6054406a1906436d6ef471810aa6e4ae4273ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 10:14:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8cadd6f07d1b4145a3846547f65b4909 : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a68bf698eca353526b4f90a2b828769a1357c873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 11:32:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eaebfc4a9b540988ba8e374d3d9e05e : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17eb927172674ca9bb283424cfb2109e : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b07127d409f44522b2ff4d11f75e6867 : SUCCESS in 6m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a1116e1c0254d488075d0640e62a73d : SUCCESS in 6m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb46c393b2f34096ae131a17f028e321 : SUCCESS in 6m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/56ae54b741a3465692f854fdedec2f05 : SUCCESS in 10m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2ba8fba7d0e046d19b3f0fc70e90782e : SUCCESS in 9m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/70401baaa1544791b3fda6de59483977 : SUCCESS in 10m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4b87de228d6f41b09d4a46293e0777aa : SUCCESS in 7m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/21052245970f4b64819b79235963e210 : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5491280c1b5944adb0358e2d2cc0e731 : SUCCESS in 7m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/68d3a8a2ec984addb9669fdc0a90ed0c : SUCCESS in 8m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/26368e6299ce4e81a5d959e1f6ed4691 : SUCCESS in 14m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8feb3dfed2f04b2fb44d7c4deaf144a9 : SUCCESS in 27m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/62c13506bb4442e09472b8c3502bae28 : SUCCESS in 13m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/96fc7987933e4423834e619cce38564f : SUCCESS in 5m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/202b42dc17714065b56ec05112526807 : SUCCESS in 19m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a867a05125e847bba27c5e1a5237faf9 : SUCCESS in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cfcb4a38eb2e4dd08a33c4684484a08e : SUCCESS in 7m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/072508e8e4a048e78a571a2e72006906 : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b910d75638cb40c29b3220f58e52f077 : SUCCESS in 58m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3c824cfdb2c44e3a882b2af2bfd9320 : SUCCESS in 1h 20m 20s","accounts_in_message":[],"_revision_number":1},{"id":"68db7483dd931a98e73edf7b115adfb807faff3b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-16 14:08:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2fe055335bfc831861e8040ab2856c3f57abaf54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 14:09:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7dc38b8fdf27c00519af880d389171c99cb09fe6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 16:27:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"be135e10585de87a7bb2b39eaaca1f569b870cd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 16:27:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d18d132aeaa1473b9d8838b43eaa31fb : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a2575760b6d47699d93bef0d044e4b1 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c73574d1daa94a9186d97c76c6e81a0b : SUCCESS in 8m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a0dd619c3ad4873895edbe595f64578 : SUCCESS in 6m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fabe2a5cdf3d4985acbfb779cd3c217d : SUCCESS in 6m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aa85fda88513401fa32613b05d183ddf : SUCCESS in 11m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/45993e7493a54c02969905fa7f54ac0c : SUCCESS in 12m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5c94e9ab28b944e6ad15104030539ea1 : SUCCESS in 12m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a25481187e34470da7d965cd17ac7091 : SUCCESS in 7m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/66e1aa21147c42fa9cbab55258c9f6f5 : SUCCESS in 15m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71794d57944a4f189b548a0a094c943b : SUCCESS in 10m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e742c1e6ddbd4131b9f11c65aa324e00 : SUCCESS in 9m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9bba2a657bb3408d8cce6640333aabd6 : SUCCESS in 17m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e85108e68f414f898098dbebbcc0fc95 : SUCCESS in 28m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1360f73e9a29435ba13935b39ace4348 : SUCCESS in 13m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b16580437a9b45c6a5643ab719644e47 : SUCCESS in 9m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/90e2d40c42df4af0b76b94e7452d2961 : SUCCESS in 23m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/593551675b1f4862a85bdec6dc2720c0 : SUCCESS in 7m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c608276f5c5a4a639d6ea331dd25821a : SUCCESS in 8m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5b452a215ae545b9a4a4c94266e18d85 : SUCCESS in 12m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28336919102e4d879a161e73c2775730 : SUCCESS in 1h 04m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1dcd5f5d57be42169c8e2b863adc79af : SUCCESS in 1h 25m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5fcd1eafe70206b58f6896ecb583aa9de360a2c7","revisions":{"5fcd1eafe70206b58f6896ecb583aa9de360a2c7":{"kind":"REWORK","_number":1,"created":"2020-10-16 09:59:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/758546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/758546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/758546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/758546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/758546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/758546/1"}}},"commit":{"parents":[{"commit":"4bb07a1163756f90cd0d5cb021b6f1259b08de6d","subject":"update constraint for ceilometer to new release 15.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb07a1163756f90cd0d5cb021b6f1259b08de6d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-16 09:59:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-16 09:59:28.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 4.4.0","message":"update constraint for python-ironicclient to new release 4.4.0\n\nChange-Id: I8a635621e5669481c5d465045159f663efa4080a\nmeta:version: 4.4.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Iury Gregory Melo Ferreira \u003cimelofer@redhat.com\u003e\nmeta:release:Commit: Iury Gregory Melo Ferreira \u003cimelofer@redhat.com\u003e\nmeta:release:Change-Id: I1a3b9357f8614a6ce2aeb1f092cf5aa3b2b0ef9e\nmeta:release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Boyron Sébastien \u003csboyron@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5fcd1eafe70206b58f6896ecb583aa9de360a2c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5fcd1eafe70206b58f6896ecb583aa9de360a2c7"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
