)]}'
{"id":"openstack%2Frequirements~758577","triplet_id":"openstack%2Frequirements~master~I9d80c5a80672f42f8b321330061c0aa7cb191eda","project":"openstack/requirements","branch":"master","topic":"test-amqp-5","hashtags":[],"change_id":"I9d80c5a80672f42f8b321330061c0aa7cb191eda","subject":"WIP TEST: DO NOT MERGE","status":"ABANDONED","created":"2020-10-16 14:38:57.000000000","updated":"2020-10-27 00:06:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f0064dbe676c76e3469f08a015a5fc7d8e13247b","_number":758577,"virtual_id_number":758577,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-16 16:18:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-2,"date":"2020-10-23 22:25:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2020-10-23 22:25:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 16:18:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-23 22:25:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"535465c7540ddbbb1701a36edc26a088198a62d2","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-10-16 14:38:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65432871d4851932deef9d26bebba2885205aa6b","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-10-16 14:51:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c783a9a67afef1546ea9f07dc7f1b314ba335e61","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-10-16 14:57:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b0f1c099bd8bfad866c8eadd82a3375d346d693","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 15:13:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9883435204c44ba39db5ebc2719b1745 : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":3},{"id":"7507cf5bb8f66e67202ba6ddb6874c355004df4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 16:18:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/276d4fda4dba4326b8dec4d22ec96a09 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/659e33419d694eadaf55757d993c454f : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44ac2095224742688cc7e41fa045bd30 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71f7bce3767f48eea7b7db744a34f715 : SUCCESS in 6m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/093fd0d47ab942959f4c2128d87e4ac6 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/defe03a7b33345e9955e6030eac53b8c : SUCCESS in 6m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/53601ddb7d3b4f398b3954b7cf61debf : SUCCESS in 9m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b315cf35402f4ab2b354aeb2ea8baf7c : SUCCESS in 15m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d686a1e717ea427d80aa6770f57f686e : SUCCESS in 10m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c61fdfc5208343b8b1c812fce3b33f99 : SUCCESS in 7m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1f82684fea0a4cfda72b3d0a46bfcf66 : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb5764f3f7fa41d4aaef0949666f7da5 : SUCCESS in 9m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f04aa05fc1e945788b94a28f4e6547d1 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d69918fa8c514520ab32122d4d0ae5a7 : SUCCESS in 18m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7d74a2c10c0147c5a249821bafa26002 : SUCCESS in 21m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ab93a5dabe854a039367a4c60dc71fb8 : SUCCESS in 18m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d5df0d05a61c41519a8a6ebe899d7d74 : SUCCESS in 5m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/209fd0e213144780bb39c302b61b0cc1 : FAILURE in 19m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b0b9afe029d94b459709e785fe7cd7de : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89799ab6eb1e4b3386786ea25be9c2c8 : SUCCESS in 8m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ed10b2bf6422466c8b58ac5f7341a2c6 : SUCCESS in 11m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6dc4793b115c4f52ac9a1c2b25e43d53 : SUCCESS in 58m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d33609d2c6f44d49c19c5c5e39cca59 : SUCCESS in 1h 17m 53s","accounts_in_message":[],"_revision_number":3},{"id":"9fb0e162b58a4f26e3123ddbf691b8e1c1e2f937","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-23 22:25:24.000000000","message":"Patch Set 3: Code-Review-2 Workflow-1\n\njust marking wip","accounts_in_message":[],"_revision_number":3},{"id":"cff5392aba96df81771d888ae5644f77c58ae7a0","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-10-27 00:06:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d86ee0deb44c60295cd6831f40e8b16d1d3ebf81","revisions":{"30845db46aa06cb7f8ddf722fad11610c9148e25":{"kind":"REWORK","_number":1,"created":"2020-10-16 14:38:57.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/77/758577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/758577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/758577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/758577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/758577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/758577/1"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-10-16 14:38:17.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-10-16 14:38:54.000000000","tz":120},"subject":"WIP TEST: DO NOT MERGE","message":"WIP TEST: DO NOT MERGE\n\nTry amqp 5.0.1\n\nChange-Id: I9d80c5a80672f42f8b321330061c0aa7cb191eda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30845db46aa06cb7f8ddf722fad11610c9148e25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30845db46aa06cb7f8ddf722fad11610c9148e25"}]},"branch":"refs/heads/master"},"67c2405fc1e0a4c7d4ad5424a78cd6e405e9aa75":{"kind":"REWORK","_number":2,"created":"2020-10-16 14:51:45.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/77/758577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/758577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/758577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/758577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/758577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/758577/2"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-10-16 14:38:17.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-10-16 14:51:32.000000000","tz":120},"subject":"WIP TEST: DO NOT MERGE","message":"WIP TEST: DO NOT MERGE\n\nTry amqp 5.0.1, vine 5\n\nChange-Id: I9d80c5a80672f42f8b321330061c0aa7cb191eda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67c2405fc1e0a4c7d4ad5424a78cd6e405e9aa75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67c2405fc1e0a4c7d4ad5424a78cd6e405e9aa75"}]},"branch":"refs/heads/master"},"d86ee0deb44c60295cd6831f40e8b16d1d3ebf81":{"kind":"REWORK","_number":3,"created":"2020-10-16 14:57:48.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/77/758577/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/758577/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/758577/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/758577/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/758577/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/758577/3"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-10-16 14:38:17.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-10-16 14:57:31.000000000","tz":120},"subject":"WIP TEST: DO NOT MERGE","message":"WIP TEST: DO NOT MERGE\n\nTry amqp 5.0.1, vine 5, kombu 5.0.2\n\nChange-Id: I9d80c5a80672f42f8b321330061c0aa7cb191eda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d86ee0deb44c60295cd6831f40e8b16d1d3ebf81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d86ee0deb44c60295cd6831f40e8b16d1d3ebf81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
