)]}'
{"id":"openstack%2Frequirements~758633","triplet_id":"openstack%2Frequirements~master~I09feb3fb710fa7a8b24beb00c7abb43dd3edbb64","project":"openstack/requirements","branch":"master","topic":"upper-constraints-python-ironicclient","hashtags":[],"change_id":"I09feb3fb710fa7a8b24beb00c7abb43dd3edbb64","subject":"update python-ironicclient to new release 4.4.0","status":"MERGED","created":"2020-10-17 00:21:19.000000000","updated":"2020-10-17 04:36:10.000000000","submitted":"2020-10-17 04:34:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758633-1602909296571-81340386","meta_rev_id":"143875d0f292d3f7175ff2ecfb4d1fe3335b9ca9","_number":758633,"virtual_id_number":758633,"owner":{"_account_id":32471,"name":"hanjiangxue","email":"hanjx27@chinaunicom.cn","username":"hanjiangxue"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-17 04:34:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-10-17 03:03:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-10-17 03:03:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-17 03:03:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-17 04:34:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e414b835c4fc52cd47c8d45f8ef09ab8a6943ea","author":{"_account_id":32471,"name":"hanjiangxue","email":"hanjx27@chinaunicom.cn","username":"hanjiangxue"},"date":"2020-10-17 00:21:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f05718f853a22910d5936c78b4a9a0dd9e14792","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 00:37:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a401831845b44abd93d5b8d7cd7d574e : SUCCESS in 11m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5ade030f73567f262b938f2d34190254ef8aa369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 01:40:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d402fdd98748b98ce089f3e5dc4cd1 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fef3313768764fd49cc348ecfcc826e3 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2eab8d51c9e4229a6ef2c827a1533f3 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00df7f219b3042e5afd046247e3f02d3 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7b59ac4344540718ddc73903a915947 : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c529cd7b4f494f289a60c3f226268dc3 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/54e22c887072482bb4fa6a2d6720f522 : SUCCESS in 8m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6ddd75b588164b4f902eb847982fc2e1 : SUCCESS in 7m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5dfee745730d4936bcba80386b09bb8f : SUCCESS in 9m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/52dd239fc2c842918302abe5ad9b2ea4 : SUCCESS in 6m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c8df25b5b6534a0e8c5099c770db25f0 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c5a5fbceff974b148837475861e743c1 : SUCCESS in 8m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/579eaa0e515a40baabe8fb2fccf9bb0f : SUCCESS in 7m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/29df16c37f2c46c2b857a67bbee5d2aa : SUCCESS in 40m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cc7029c896484d279b66c5c706be8898 : SUCCESS in 24m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/53358d74368943579f82d0fc1a9ed730 : SUCCESS in 12m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b6bc9d5b97c6450b9adb6326e4ecfa63 : SUCCESS in 6m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb1f7be3d5534a188d7888c7c8bed308 : SUCCESS in 19m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a001503861c4412785fda31c2a53722d : SUCCESS in 7m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a9acc13fae345f88b5a0e207e0b2793 : SUCCESS in 8m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2113a1cbc8f7492f86221cbc61004ff6 : SUCCESS in 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/915a94516a654ef3b5289abf92de6042 : SUCCESS in 54m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9658664aa45b453ab076e74fc6f47ec8 : SUCCESS in 1h 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4c480175c91b2bf4435446b57c55ddc8d940e42b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-17 03:03:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba62d8c55d6d27df72061e8ca887008378668f0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 03:03:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b16fb18b84a9713859b220ae6f91cbf0c203aa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 04:34:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9a403990b484f6c9dbd3c23607262ef3ae5def83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 04:34:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd5aa20979bf4d17a57589bdbacd74b6 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef44dc4333847129abed63277c47f26 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00fed9ff4f344a0db6327a7c7126febb : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ced1a855540476f99a55db9330b0ae0 : SUCCESS in 7m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0bc922de9ac14603922fcf6bab1da4d7 : SUCCESS in 7m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9eb03ca040614f07af3a49a36559d5c3 : SUCCESS in 7m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c46238284a754702b2062d557a53ea59 : SUCCESS in 9m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fa43f60db6114d729e3269b68e821e13 : SUCCESS in 9m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/68352862fadb412991b60c0f50e06301 : SUCCESS in 12m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f543ebb83fcd4db09b875cc3bbd04f17 : SUCCESS in 7m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/de702891166b4c55866765a2b39eb706 : SUCCESS in 12m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2982ae73a38341f89efaa067917eaa61 : SUCCESS in 8m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c799a0b834484cc08bb0e32cebe47240 : SUCCESS in 9m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/09b4f082f4114bd4a906d5e73138f817 : SUCCESS in 22m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/77f2e904dcdb46acbb39ad7726fa5dad : SUCCESS in 29m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2fb382ae2b554901b5ed2bb99f54ae9f : SUCCESS in 13m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/507d32a7e0b444df862f4d8a112a43b6 : SUCCESS in 6m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0c959a0494ea4a19a5c5cb89b3cde39d : SUCCESS in 19m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a1925aedcb124f54b6c94182d5dbb0f0 : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bcc09d4be99842f8bf299af1280316fc : SUCCESS in 9m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3c3069a425f340a6867bba13c904d674 : SUCCESS in 13m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab40110f1a424eef87a70d00edfa5394 : SUCCESS in 1h 03m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eafc8a2833bf41b890eb5f5d337ec3f0 : SUCCESS in 1h 19m 39s","accounts_in_message":[],"_revision_number":1},{"id":"83d8ff0314fcade51c50609fb590e5939498e99a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 04:36:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28be77ecbaa54346a7578eef4175b5e3 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1b94827388418056609616180cc7c9bf5ca9340","revisions":{"b1b94827388418056609616180cc7c9bf5ca9340":{"kind":"REWORK","_number":1,"created":"2020-10-17 00:21:19.000000000","uploader":{"_account_id":32471,"name":"hanjiangxue","email":"hanjx27@chinaunicom.cn","username":"hanjiangxue"},"ref":"refs/changes/33/758633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/758633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/758633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/758633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/758633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/758633/1"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"hanjx27","email":"hanjx27@chinaunicom.cn","date":"2020-10-17 00:20:27.000000000","tz":480},"committer":{"name":"hanjx27","email":"hanjx27@chinaunicom.cn","date":"2020-10-17 00:20:27.000000000","tz":480},"subject":"update python-ironicclient to new release 4.4.0","message":"update python-ironicclient to new release 4.4.0\n\nChange-Id: I09feb3fb710fa7a8b24beb00c7abb43dd3edbb64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1b94827388418056609616180cc7c9bf5ca9340"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1b94827388418056609616180cc7c9bf5ca9340"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
