)]}'
{"id":"openstack%2Frequirements~758796","triplet_id":"openstack%2Frequirements~master~I4c028c7f1f5608a1c2a60c29195ce2eeae863488","project":"openstack/requirements","branch":"master","topic":"bug/virtualenv","hashtags":[],"change_id":"I4c028c7f1f5608a1c2a60c29195ce2eeae863488","subject":"update virtualenv to 20.0.35","status":"ABANDONED","created":"2020-10-19 11:29:20.000000000","updated":"2020-10-20 01:10:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ff09c5fdbeed8fe7a61aa5d71740ded51aac55ea","_number":758796,"virtual_id_number":758796,"owner":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-19 15:21:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 15:21:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8c08574194dad69a200168fc58b6f968f73e438","author":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"date":"2020-10-19 11:29:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5ddc34bca4d42d7e9265b223c5f7ce57aab280b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 11:44:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/04db7743207d4fe28de3105444f551c3 : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":1},{"id":"521346f512d813837961b5ce8d37d5fbe4057501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 15:21:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c83e1d62ab436f80cec204cdf6d1a3 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e410294dbe341459985db45ace86640 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97bd55d0b30c42bfaef79716b40b1b09 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c959745974c146f69390aed8d5ec8082 : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84cd85a7e4c64b30ad1fe47a52867f7a : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e981cb7276f9447185fba727a8e1d89c : SUCCESS in 5m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f4d4f16d9c224877bf9bb1511565b037 : SUCCESS in 9m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f80e886cfc4940e3bfa26c85fd34f3f0 : SUCCESS in 9m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/105b99f4e92b4861bf52148aa2b1465e : SUCCESS in 12m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2e0cae8724914f6e856b0e105081e5f2 : SUCCESS in 6m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7627211d589149f591841a747b68c84b : SUCCESS in 13m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/41c9b12951ba4fbf9f48033f0de418fe : SUCCESS in 7m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fa1610c77eda48f48c206101f9491496 : SUCCESS in 8m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/03936f9912234430847d23da795e08dd : FAILURE in 17m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/92b17fec3c6b42e3ab52dd89ec75fa54 : SUCCESS in 27m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/06940012374c433287e81c8f43e92c01 : SUCCESS in 14m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c32289919744416864be933131f1adb : SUCCESS in 6m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98849e99df364d16bf41282527689252 : SUCCESS in 19m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/83c292243e6140ff9761540733084bfb : SUCCESS in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b5d48b95d4e4af2a810f7c35e396f06 : SUCCESS in 8m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/52623046ec8b403987bf245967b6eb46 : SUCCESS in 13m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fdeb8083a7f41e39e1d1646d895aca3 : SUCCESS in 56m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9beb898098e3489e8acf5c5246293a1d : SUCCESS in 1h 10m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ac9e144bfd5bf038062806d9c0c1c26384140ce2","author":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"date":"2020-10-20 01:10:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd4ec6200304854740283e759a5d6627f7388459","revisions":{"dd4ec6200304854740283e759a5d6627f7388459":{"kind":"REWORK","_number":1,"created":"2020-10-19 11:29:20.000000000","uploader":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"ref":"refs/changes/96/758796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/758796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/758796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/758796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/758796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/758796/1"}}},"commit":{"parents":[{"commit":"747ff02e2c3b9dc6bcc461731cec8758da7704a4","subject":"Merge \"update python-ironicclient to new release 4.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/747ff02e2c3b9dc6bcc461731cec8758da7704a4"}]}],"author":{"name":"zhangyc","email":"zhyanchao1988@126.com","date":"2020-10-19 11:28:33.000000000","tz":480},"committer":{"name":"zhangyc","email":"zhyanchao1988@126.com","date":"2020-10-19 11:28:33.000000000","tz":480},"subject":"update virtualenv to 20.0.35","message":"update virtualenv to 20.0.35\n\nChange-Id: I4c028c7f1f5608a1c2a60c29195ce2eeae863488\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd4ec6200304854740283e759a5d6627f7388459"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd4ec6200304854740283e759a5d6627f7388459"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
