)]}'
{"id":"openstack%2Frequirements~758844","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I77bb915f76619e3f6ebcda5f0c916dd07890d18b","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I77bb915f76619e3f6ebcda5f0c916dd07890d18b","subject":"update constraint for oslo.messaging to new release 12.5.1","status":"MERGED","created":"2020-10-19 16:50:23.000000000","updated":"2020-10-24 00:18:39.000000000","submitted":"2020-10-24 00:18:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758844-1603498719179-caddc103","meta_rev_id":"80e5680fe7d3f16f3fcf754d4e47441083f0a3d2","_number":758844,"virtual_id_number":758844,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-24 00:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-19 16:53:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-20 02:26:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-10-20 02:26:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 16:53:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-20 02:26:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-24 00:18:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d46c7f28ab26e0828787b4b5c75a73922301656d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-19 16:50:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0940b72ecbbcd908723517ec26b1b25a8451d03f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-19 16:53:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e601364f8a6f30001de882cf4a16e96cab4d89fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 17:09:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f39b83a487f14256bbbedf21143ab0bd : SUCCESS in 12m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d8fa26ef86f52b0c95cea220d869a190588a150b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 19:24:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e730b69fe9742b988d81958030e5257 : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/061e51af83b442ee8f61e334b3922a82 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72ec1a1dd1ba4fb3bac89112a9036567 : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fafade4e1ff34786aeeaea359a5ee5b0 : SUCCESS in 7m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ef9252662944298b72c112e1faf6158 : SUCCESS in 5m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/858eb518b6874c618ff91d19818da79c : SUCCESS in 9m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e09bb1c894ab40a88274a52cc53f3ab2 : SUCCESS in 9m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a126fd7477784b4ca128c1003ff88d5b : SUCCESS in 10m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7fae9b70f4104f2bb4c6f9f1a19a90fe : SUCCESS in 8m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ba2e9596bcc9487498412f08a6b295dc : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b4c3e08f12284784ad06823a95df398a : SUCCESS in 9m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/10efd195bad847e6a165b0ab19b47fd4 : SUCCESS in 10m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d24e4fec0be344bcb8b22325729841f1 : SUCCESS in 55m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e3790d112fee4007982ed36173d5149d : SUCCESS in 27m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cb10184063ba47eeb22cd22ac1124bd6 : SUCCESS in 15m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/886ac91eef04407993c346480ef01807 : SUCCESS in 7m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0cc407ef39a844b1ab65b062d1c64717 : SUCCESS in 21m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/42ebb0500025416e8a91faf3263957f6 : SUCCESS in 8m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/532ce1c2dacd4a758d5e16e3cb41004c : SUCCESS in 8m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9f809827287246b4b963e3fdf691999b : SUCCESS in 23m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d104b3cc11e467da74172ade8d153c3 : SUCCESS in 59m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22bb10fbac44467292f84139a9fdb831 : FAILURE in 1h 10m 00s","accounts_in_message":[],"_revision_number":1},{"id":"16a23ade8c0ac7bbba0f480f470a769fa6d92877","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-20 02:26:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eed31bfcde59195157739fa6dbc4b76d30555bcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-20 02:42:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5aa027c77ef443659e1d469d4db42196 : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":1},{"id":"02fa44fd9b841c415b24b73ef780c1ad80f4b75d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-21 14:16:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"51c85a706a268ad232e7e4b502c898c18cefa660","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 14:34:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"19109ff04f38d5bae56474933607a25c163ce40f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 15:38:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aac6d4dc4fd4481e943b76177294136a : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81bc448f2521444aae4784ad2afca1be : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57d4753f41ff4e6aab8a2ea356f4cda4 : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80c06b7cc9794567bd0d4855cadaedee : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3601209f60f44261b427531abca3831a : SUCCESS in 5m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/826a76272a614fbe92e1d7d6c8ffbab1 : SUCCESS in 8m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d0aaea6770fa482bacee1c3d8aae324c : SUCCESS in 8m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ca886474ea744516a6631e10d100a9ac : SUCCESS in 10m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/49dfba017db34f8abb53393cf1415ccf : SUCCESS in 6m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/860a62f279f84f55865f29694e0919f9 : SUCCESS in 12m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2940ec96455d444fa9bc8237d339c796 : SUCCESS in 8m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cb578950b3094ed597d67e7f803dd346 : SUCCESS in 7m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7ba93c7f2bc141b2bc98d1aec6fb585a : SUCCESS in 13m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/92eaa43026e547b2aef565d3a2755bdf : SUCCESS in 25m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f048e5af62a54d9684d341ac82b5a61e : SUCCESS in 14m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3708c60e9c214bdf81b7f022c2e75549 : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0eb90e76872a4b6fbc9a3465c3fe2f2d : SUCCESS in 18m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3dd9021fc76b4222b0860606bd16849a : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5a104d07d944288a28b7c3e1449e8d5 : SUCCESS in 8m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2e9645e442df4b44a56a1de7c238c9c6 : SUCCESS in 32m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f649aa16cf242bc8964a0a73cceb6b6 : SUCCESS in 56m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa10087bb32045cdbb5280e9835a2012 : SUCCESS in 1h 18m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5b897d8886016cc22b71dff97c7a316610fa067f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 15:38:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d5e86df1469a6bfd11dee9ad9a0b157f7cc0c99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 18:19:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f6c83cffcb44ac82285f426efa0a23 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be2ef85a6ecb40f299f061ce5bbe19d2 : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4df603d9602d40d3af27e35ec4aad2a7 : SUCCESS in 8m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/868632a62258458f8eca3d4aa3fea954 : SUCCESS in 11m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5db58043a7f7455b99ca174180ff2943 : SUCCESS in 7m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/51b6f4f4fd5f4c1eb45dc09b149031d7 : SUCCESS in 11m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4f7712cab08f4a19ba484d68eaf1a913 : SUCCESS in 11m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4472bd114b444e80bc1d276120e0e1a5 : SUCCESS in 14m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f85a7d19e43b44a6b56e877a1c7de711 : SUCCESS in 8m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/58825f9478ef4d5c9946a5889cc54d08 : SUCCESS in 13m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/900137fb25b54bda83a18bd3cb2ab2c9 : SUCCESS in 10m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9a8416f4d8724f23ad8231b0c84f58fd : SUCCESS in 11m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/83dc22662bb746f1aa6dae3f424588e6 : SUCCESS in 21m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8912ec719fd042dd8988fc400ce591c1 : SUCCESS in 35m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1363195061d34bba8d812b0171102b5e : SUCCESS in 16m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/141a32c45c104b5ba81575bbcd73e3a8 : SUCCESS in 6m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1988c1529d3e44d7aba072aef257bae4 : SUCCESS in 22m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/01b47d73ccd44af7865293e0d6d1ee0d : SUCCESS in 9m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/07cf6e4d02d74a0895dc02ebf7ba1a34 : SUCCESS in 11m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/924f23c3ea004c229d665d5321b1741e : SUCCESS in 15m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55fd327b5abd4875acb0f82ab3dae119 : RETRY_LIMIT in 17m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e583e2754a7a4f9288e4533f8aa9e0fe : SUCCESS in 1h 19m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ae3bc7aae368455a451af70963af0d3838d6d52d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-23 21:57:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8b22fe08f3f9a217e337a49a8ea4b36e3e99fbee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 22:15:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dc568f47fe5a4f23999e349f28e576e9 : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"dcee44677f86548c86950ed82934de9b93ca0094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 23:09:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/550744aec8ae4916b46f3f08794cc0c6 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/595c16b00ba143a1a25081a3c1ad1f5c : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac77b1474de24a469200e78c5bc6252e : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/745b82dd446a49398d34751a527ea451 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b4ab057d0384b2888e10dd6878917d7 : SUCCESS in 5m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/698ee76ee54a4045adc8e2b0c013142e : SUCCESS in 8m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5dba9a0e32754d48bae2d3967767dcbb : SUCCESS in 10m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aae3f39efd3e42b19d3fa9ffa8fb8db2 : SUCCESS in 9m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/079c1be295f147168c289643d3efa89e : SUCCESS in 7m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/07e375e8d7614045b4b88dfd2d68cbd8 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e8f04605c52e41e3b5544cc27f139fab : SUCCESS in 7m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4ab4d2d9b6fd4ec69629c8f1cc129c60 : SUCCESS in 8m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b7180820d2b44bb1bf4f02bf79a496bf : SUCCESS in 13m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/524579de20904967bc390e96e1ea3545 : SUCCESS in 24m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7b457288fd5d45a4bc0a8b1046a4ed15 : SUCCESS in 25m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ca291dc59544465e8dee6490ed3f808f : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f7fddf10c5a4e4392c8f4b8846b890c : SUCCESS in 17m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/880a4298fb9144fa99a72b6639f20fc8 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2bc219d61a74b7a891a304c4cc34547 : SUCCESS in 15m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bef2b4f35deb4a2bb585dcdd8f88cca0 : SUCCESS in 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d516b25a5f94ca39ca797b49677d82b : SUCCESS in 52m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1291a52588f4beab58f6afa2d57da69 : SUCCESS in 1h 08m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e70513f7431a3597470ec9222863594ed65d821d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 23:09:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27a0498e3b6ba20b00efa20abc9d075f7e4de8de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 00:18:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8158a762cbd14472a5b3c56786359867 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9611dc8c275542d490dcf46c9944e29d : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac196e42d4564459bd63b60e5b56304e : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35dea6d4a0c14d878971059027182725 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f837a2426a6649fdb5394082aa362c9d : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7e6fb64c243f4ce3aca348a0275acc43 : SUCCESS in 6m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4d43625eac4248029d8f472b28a06532 : SUCCESS in 6m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7e0b6d7ef44245fc9b8fa7b133ccf8dd : SUCCESS in 9m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/77c993db4ead48b29d226b733ada1625 : SUCCESS in 4m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/763094e2c7e246998b387ab1823d32ff : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e9ddd2ac2bd547f89fbbcc5ceaf41433 : SUCCESS in 6m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/035688fc79bd47dcb6d3d5b7949389a5 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/10ff6a63af6f4d8ab7adb6791ab7c3d4 : SUCCESS in 16m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4018b7cc3e2545669626be07b27689b5 : SUCCESS in 24m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7198eba512a8470a978841eb843b37c9 : SUCCESS in 11m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/833c99e8c4d44687b8d9acfdbf7f7c09 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a82725465db94b87a9379a89efc734ab : SUCCESS in 18m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cabe975903364f5c967a2ec92a523f4d : SUCCESS in 4m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd74478f983f4278afd9aa7341233909 : SUCCESS in 6m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4184efb59baf4989a97a5dcbe92352c4 : SUCCESS in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fd9fcfb6fc94b6c922b9e54a784c56d : SUCCESS in 50m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/211b4b1231b74ed3961b083e143ddddc : SUCCESS in 1h 07m 32s","accounts_in_message":[],"_revision_number":1},{"id":"57667fa937361eda42c98758730f5b6ce40998ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 00:18:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c324cf01c85265041b3837c7b4a5ef39c1c7e56b","revisions":{"c324cf01c85265041b3837c7b4a5ef39c1c7e56b":{"kind":"REWORK","_number":1,"created":"2020-10-19 16:50:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/758844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/758844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/758844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/758844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/758844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/758844/1"}}},"commit":{"parents":[{"commit":"679941a9f4f9d2053d830ea44c58baadd03b3a11","subject":"Merge \"update constraint for python-ironicclient to new release 4.4.0\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/679941a9f4f9d2053d830ea44c58baadd03b3a11"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 16:50:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 16:50:21.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 12.5.1","message":"update constraint for oslo.messaging to new release 12.5.1\n\nChange-Id: I77bb915f76619e3f6ebcda5f0c916dd07890d18b\nmeta:version: 12.5.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Change-Id: I7bce53ff7bc87ae19ec80d83d788dcfcb853a5c8\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c324cf01c85265041b3837c7b4a5ef39c1c7e56b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c324cf01c85265041b3837c7b4a5ef39c1c7e56b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
