)]}'
{"id":"openstack%2Frequirements~758886","triplet_id":"openstack%2Frequirements~master~Ic422fa58de885633f305cc8c266c5ffa837fe119","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic422fa58de885633f305cc8c266c5ffa837fe119","subject":"update constraint for os-win to new release 5.2.0","status":"MERGED","created":"2020-10-20 00:57:19.000000000","updated":"2020-10-24 04:30:48.000000000","submitted":"2020-10-24 04:29:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758886-1603513748581-5b795edd","meta_rev_id":"40f27f059ee4008d3ccf5f186f8d20f095e442e6","_number":758886,"virtual_id_number":758886,"owner":{"_account_id":32471,"name":"hanjiangxue","email":"hanjx27@chinaunicom.cn","username":"hanjiangxue"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-24 04:29:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-22 11:47:47.000000000","_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-10-20 02:26:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-10-21 14:40:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-10-20 02:26:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-20 02:26:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-21 14:40:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-22 11:47:47.000000000","updated_by":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"reviewer":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"state":"REVIEWER"},{"updated":"2020-10-24 04:29:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6d28c6ba9aa64f4e204b73a387585e4c595ff3f","author":{"_account_id":32471,"name":"hanjiangxue","email":"hanjx27@chinaunicom.cn","username":"hanjiangxue"},"date":"2020-10-20 00:57:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"392e258b6535084256e1ba60e07a771d91a4156a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-20 01:13:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/29307b1ec40c4ef4aebacf849dff0ca3 : SUCCESS in 12m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d7c8acd877f36b0502528755b030392f6305b838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-20 02:19:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c8eddadf3f047e5befc268147d9b136 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63354223a1e34553926d3412afe36f89 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a39786a2d344139909d6195ac3cd6db : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/449ec4feb3324f27b1462a6b9aa4effb : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b230031c0723456197b2eec89e00a0f4 : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7085ef9c30aa4e13ae21b49268979d23 : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/365b3b1b1466400190f98e0b8f5b8734 : SUCCESS in 8m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/92072a02970d48aca378d7421e1dd7ba : SUCCESS in 7m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c7149ad7d0234c62b5f1eb974c41fed8 : SUCCESS in 9m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/264b7ac8ef9842879a91af0fb805459a : SUCCESS in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4373db814ab749b29afdd76b32f068da : SUCCESS in 12m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34cdd54d0190400786a3549ae0fe6b04 : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ffc6b10768d34d11b6d9d445570fe1e1 : SUCCESS in 7m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/56bc05e7d4b646e4866d5ce2fc7fe83a : SUCCESS in 15m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7b677ca9118f49658371be4d10b5947e : SUCCESS in 17m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8a2fa9bd735b4c02b99b2fdadcf133a0 : SUCCESS in 20m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/85a0fe4b8d3449a8bc3ab3add30544dc : SUCCESS in 5m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/babac97b312e42e3a05cc21a9c8ca7dd : SUCCESS in 20m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/24a6892298254a33a0624cf8d344df4a : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/94ccae4239164018a1c64cd427d59e7d : SUCCESS in 7m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0f607b2c15e949218e3e6eac262d5cda : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba47992b166a44fd806648fd1bef3a18 : SUCCESS in 1h 00m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d965ff44bac44c738923fd60c0e3294c : SUCCESS in 1h 18m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1744bb17d35743618df5d755a6b57877d935b097","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-20 02:26:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a36ace1f3725177b6ae84d75ce5cef4e5e156697","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-20 02:26:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6ae2f44aac5c7e0c850b9735889182175b03973","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-21 14:40:38.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck\n\nLost in a restart?","accounts_in_message":[],"_revision_number":1},{"id":"e53ee1dc5d8e9634b139ff29f29519d92e9e4e8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 14:43:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"cec2ca64a827464836a3ae6673b7034cbcbb727b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 15:51:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee5f251ea1c9426e99daab3255fb7a1b : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/202e5fdff97244788de7b11c8a221888 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ae4eaa4afa14e13af3038886312cc2a : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51930a3b84af49f089488cf4ecd61e0d : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9bb6bf7c9c5e4122bebc344322f37980 : SUCCESS in 5m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/07416d145ad6435ca0137b9edfe55741 : SUCCESS in 5m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6344a41bb36949d59ae0e70c3351a58f : SUCCESS in 8m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bbe0775cb7c4411cb27b5ff13998bb7a : SUCCESS in 9m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ce4c34b3d757425c9b253f945da27217 : SUCCESS in 12m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e19a59d601614fce9b8f6069938e2e1c : SUCCESS in 6m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b6aa369b4e4c4ad6abb998b777411ec3 : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a0c1a82329d47ada59bc980fe17ab48 : SUCCESS in 10m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fcb05e2c83c74f5289c665276c4bf515 : SUCCESS in 9m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e676c9aedcdf4c248d302aeae28ad9bd : SUCCESS in 17m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/df0d638d76b24d1ab8b1685e10d05e48 : SUCCESS in 25m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0675c2a4c37b4eeb8cb49c785aa43cc2 : SUCCESS in 13m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8b3e1614984e4c2089988a69e2bf46ec : SUCCESS in 7m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1d7e0335276479cbc3de34186c9651d : SUCCESS in 18m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a57dc095cbd1482996c779c09ec1ac6e : SUCCESS in 7m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c32c8cc892a4d469578222a30d0da78 : SUCCESS in 8m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/225475e305944cea80e8649c7883fe7d : SUCCESS in 13m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1edc0ec9b08b43cebff16b6ec3d0887b : SUCCESS in 1h 03m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b68401063c224a4a807b7102a6a9f672 : SUCCESS in 1h 08m 11s","accounts_in_message":[],"_revision_number":1},{"id":"baa5bd41fb6b9f50b228c19804058572177a7031","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 15:51:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c31154b53ee841153b9d986df997ed7afd2cb1c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 19:07:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc8d4a1296544c86b194ea3f29b2473b : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87974438c634901a629a1be64cd31fa : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c977f88dd14b48beb687f2348ae2c123 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3a7890b67754a15898aa7e8539bfbbf : SUCCESS in 8m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c4b8e31dd4254bb09d1d4c56b3ee3002 : SUCCESS in 6m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/762f85f2fdda4089adab2377fa4e22e5 : SUCCESS in 14m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/49c12fe4347b4476ba15ed4eb60a3fd1 : SUCCESS in 9m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9a86e71b7ee64080b319b0157b938300 : SUCCESS in 10m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/159a9a0dbfc845dd9cdcefb62428e42a : SUCCESS in 12m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6ff5ce10fc8a4f3b9d7ff996b4c7905b : SUCCESS in 8m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ba1ae00d3bf7442383e1846dc61d7660 : SUCCESS in 13m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c1eadae275f0423abb45da8c37dd4f8d : SUCCESS in 10m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/881c4c9b44424b8982bafa0c3f945133 : SUCCESS in 10m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dfc0835007f241c69e44eabec3fd8feb : SUCCESS in 17m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/58f0c8f338194b93bffd20e6eb9b6e59 : SUCCESS in 28m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/deae54a9027b46c2a13fa2a359a0775c : SUCCESS in 15m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/da999e08803d4de9b8a9bf00fa3b20dc : SUCCESS in 6m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a83b14e185f944c3a4276db00fa273e5 : SUCCESS in 21m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/77b36476464144ada232ac07c45139b4 : SUCCESS in 8m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a37eca760a441b8a73d0a071293e45d : SUCCESS in 9m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dc7056a140c64265b17476a092e31666 : SUCCESS in 13m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23d5e7ad941940e6b5e50531c05caed0 : SUCCESS in 1h 03m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/964271c81446403abbfc7c148f933814 : FAILURE in 1h 13m 09s","accounts_in_message":[],"_revision_number":1},{"id":"34f1708f6679057d6e2d15893212c2346f0acca4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-23 21:55:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"96ebfe3e35c2779665e6dff4a23ef03116f3d914","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-23 21:58:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1abd7efe498cc9378a833556d23f46e6e7423a6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 22:15:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/563be47a1a1d4ce9adab29cf567e749c : SUCCESS in 15m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c1d4a7ef34c9f7b995a2444e502f4a3c49c636f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 23:16:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7a8b7a3a30d4ab1a4506ddfe3bcfb2e : SUCCESS in 18m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ad1b05d114543cea13349f8d84a8537 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b67d058f7c24984bed4c0d340561f35 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/430f5854f0c24e20b4bdbc1f496eeb73 : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74f78aad3b16436ab1d5aa3c25d11b56 : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b85087e4d174f7c9e5298089e1f6d5a : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7ba7aac25c464a78a73bb6cf598aff15 : SUCCESS in 8m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0c78727be61e426880a8abbbc23bde3a : SUCCESS in 8m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d965bf6d4df44db4977044f1a6c21e03 : SUCCESS in 10m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/668e7801843f405c8d21b3b9a8f93fa4 : SUCCESS in 17m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f9fc89d11dd64b9aad48bdcb525ec279 : SUCCESS in 21m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea522e516ab442658dd757a56449fc77 : SUCCESS in 7m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f3bdcd8b65d842a8b219d00ddf691278 : SUCCESS in 8m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/77abc83beff44db999f9750f67180944 : SUCCESS in 12m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/324aa9bc5eaf46d6821f9c1b5c365b82 : SUCCESS in 34m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f749617ab14b41cfaf575ce2b9dfd41d : SUCCESS in 12m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/13ab30f47063497f9febf24a32844147 : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a99e3b48f514200ae667a7cd68113df : SUCCESS in 32m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ca253bb3415245f4aac886dab8642b72 : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/241083d0dbfb462f8c0b8a6d672aeaa7 : SUCCESS in 7m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/088f9f5be4a241989d4a2c3ef53fe3e2 : SUCCESS in 18m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bf957ee158c4fbabe471a1fb3fb98c9 : SUCCESS in 57m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/078808493c56442abb6fc635ebee81ee : FAILURE in 1h 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ce0c8f367cde26034a6a4bcc5a58996e33324757","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-24 01:49:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"40adac2c0ae5f8d7b0028d7195a8eaea7bc66e42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 02:11:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e603a8d337114a218c5e09c1efa088cf : SUCCESS in 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"5865995c3aa9c8b797c44a220b82e00d7270352d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 02:58:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/508f4ede584b4ef99cd6f45798914781 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3ad1b4170094b0dac60f811e2d462f1 : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b0195d1e75c4329a6708315e5bb8817 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc8090e46e9f4760b3d346a4b9a1f606 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/967890e33a8f4167ac3bb408a47b74c8 : SUCCESS in 7m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bceea6003460487f9ae122e9b4487c56 : SUCCESS in 2m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3f1ffd4c504a48bf9a969cd3ccbf814e : SUCCESS in 7m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c97c5bd095784bda9fab02aac99aff4d : SUCCESS in 11m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/05535e392760471287e34057c7867c80 : SUCCESS in 12m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fbb74c9515d44bf59068711cbe86dfcb : SUCCESS in 6m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fad2b66d85d444c981813e6e394dc54f : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c30786d693564dc7ab5fe4d890ca883f : SUCCESS in 6m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fd8c32e4308a4e56a09ffb8e06955249 : SUCCESS in 8m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d5672bc54fe648a1a26982f732cfa7fa : SUCCESS in 14m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2f40c89650744fe890bd54adadad3a66 : SUCCESS in 22m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3c463d079e8d47e182bf694fd72bacb3 : SUCCESS in 16m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7998ee93d005498aaefe34d5e1d40d41 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0cc002504663410d86421945902a1cf7 : SUCCESS in 18m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b5edcfe32ad64bc7ab72d80be8285bf3 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4612928c8d4d4762a54e97dd72013def : SUCCESS in 7m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/211886128ad74c3893182fddfee467b2 : SUCCESS in 8m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1824456c36c04050b1120da07a3573c8 : SUCCESS in 48m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e3093e420584584a5c48ac50133dae0 : SUCCESS in 1h 06m 57s","accounts_in_message":[],"_revision_number":1},{"id":"69c389e576d8fba7e5df0e56251dbd9032620ecb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 02:59:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2901fda50f71316f0776fedc93a321a475d746e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 04:29:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fc497cdcdfb998c1143550b3c91e2b9755a156fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 04:29:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a293e6509d549d4b206bb4afdf4f21a : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43b969920b2240928390f5d9e436f167 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd37c2155b404cf4b742d0b8249a74b2 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e20a2cd9ec154381a51bda245ce93e5a : SUCCESS in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51815b69c9c44958aa895dfc8dbb0260 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/424a5ac4fd7440948ecb1f393dc2e84f : SUCCESS in 2m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/09b4fd281b094f35adbe64a5da3d1552 : SUCCESS in 6m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/64c1a0d7de1a4b01ae09f8c5f8a4d5a7 : SUCCESS in 6m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eefb019abd3643cf8d3bb8abd124c78c : SUCCESS in 9m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/967f639626aa44c8848d204ca9371559 : SUCCESS in 4m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/378ba560b8d641cca36fe63e97737e75 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de739a776efb40cd90e71c48580d66bc : SUCCESS in 5m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1f6da9e00d8a475588555d65fed6d050 : SUCCESS in 6m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cd076c80721645568bef16f9ed9ed592 : SUCCESS in 12m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e2523a9e5938495d9cf6c970ba90aefa : SUCCESS in 25m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9de274ef5b904693b367cf787bf21069 : SUCCESS in 11m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac8b53614dc64772a969c70a13813910 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4dc32824f74f4a028b0170a7e69ef177 : SUCCESS in 16m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cbf6171b386f4f0fb9b7930923798976 : SUCCESS in 3m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c386005a7354cf6b364e1dca4abe05b : SUCCESS in 5m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e0fa567b27db4e1bb500a729cc224044 : SUCCESS in 9m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0447aa23aed14ee0bce128dcaf77aa88 : SUCCESS in 57m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1db9dc3dbc9c42bc869b817df105f975 : SUCCESS in 1h 23m 43s","accounts_in_message":[],"_revision_number":1},{"id":"c3f4ae28b70e6d118d942f0085b408bae44646ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 04:30:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd86fb8fa4a4bfcb54ef46c55ed38cf : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29f4bd8248f7d64c5cf1155f0506e55a6faa10da","revisions":{"29f4bd8248f7d64c5cf1155f0506e55a6faa10da":{"kind":"REWORK","_number":1,"created":"2020-10-20 00:57:19.000000000","uploader":{"_account_id":32471,"name":"hanjiangxue","email":"hanjx27@chinaunicom.cn","username":"hanjiangxue"},"ref":"refs/changes/86/758886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/758886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/758886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/758886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/758886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/758886/1"}}},"commit":{"parents":[{"commit":"d447ce59fdd0c5ca992486af35dbd5d1008181a7","subject":"Merge \"update ceilometer to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d447ce59fdd0c5ca992486af35dbd5d1008181a7"}]}],"author":{"name":"hanjx27","email":"hanjx27@chinaunicom.cn","date":"2020-10-20 00:57:00.000000000","tz":480},"committer":{"name":"hanjx27","email":"hanjx27@chinaunicom.cn","date":"2020-10-20 00:57:00.000000000","tz":480},"subject":"update constraint for os-win to new release 5.2.0","message":"update constraint for os-win to new release 5.2.0\n\nChange-Id: Ic422fa58de885633f305cc8c266c5ffa837fe119\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29f4bd8248f7d64c5cf1155f0506e55a6faa10da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29f4bd8248f7d64c5cf1155f0506e55a6faa10da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
