)]}'
{"id":"openstack%2Frequirements~758947","triplet_id":"openstack%2Frequirements~master~I32eb5479e7eef90eef5967891c57f16f60cf0c1d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I32eb5479e7eef90eef5967891c57f16f60cf0c1d","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2020-10-21 06:21:52.000000000","updated":"2020-10-23 22:09:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3742dc85017b1c3d41182d8af9bca13579c08090","_number":758947,"virtual_id_number":758947,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-23 09:11:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-23 09:11:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ecaefd442245b94ab2059ff7908fcbc77d8b8a69","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-21 06:21:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"762e3193e1a86aedf93785e91b4274222d5a5d5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 06:25:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"976216af46e718371e3b85a05c99ef93c5419045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 09:35:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a17dbc1386554094b047232a5f06f8f3 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/043edb5de3704dcda169db49c7d07676 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b58bcfa4af9c42aa8513d9a7ad9a4d3c : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57ea5d88eed441d89d34626ae96145e8 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61a3ef2772f44ceb8ac9fbc486458d0a : FAILURE in 6m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ce7177257c74da0814f985c0c8b5146 : SUCCESS in 5m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d8274a3922364d1bb73544947f72bd69 : FAILURE in 7m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3fb71695bb5d43f89b806b9adf969867 : FAILURE in 7m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2d0630a50ec3404f892d10f81c2d1679 : SUCCESS in 11m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2ba4558936b04b869a6ddd4decce281b : SUCCESS in 6m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e9e3f650f371451faf9cc73d92c39620 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7ffe289e36f84485af88a6cfa2671a7f : SUCCESS in 8m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/14ae2001845a4a5aab1cc982cb2f5f97 : SUCCESS in 8m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/794474faffd24fafade438d6686b4a20 : SUCCESS in 17m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b712f72ebf864b7f8a82d3ccced72674 : SUCCESS in 32m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b3c4782c016b434e8f3ad2159a6f43cd : FAILURE in 14m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e63971479c97406d8c38b36e68455ad5 : SUCCESS in 7m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d32a5a3792f24297b9e1f812e3819878 : SUCCESS in 23m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a17db7b9c69e4fdaa53c3be185d761cb : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4a9a283b41514fb8a584d61cb7a44484 : SUCCESS in 7m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/323cba19418146b494d873b9985a751a : SUCCESS in 15m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60e27e073afa43e9af6c8e4e8eab0b86 : SUCCESS in 2h 28m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24c33562014a41729cdb4f959c8eb726 : SUCCESS in 1h 16m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a824a10608075300c09d9265683fccf3d20d1964","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-22 06:18:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9db3039f36f123f2bb412acf237b1e0c4f00d05f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 06:30:20.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"4a9c0b35bd65a52cc3fce1d15bc8f992afb998e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 08:48:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d290ec39b5480f842eae434019d0e0 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/434c7022081e41c49ed26d95a87313eb : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02b27a5bfa924f3baaef32a0b9b293a7 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5e4c9874baf4e499a062c9b8f29c6bd : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97422b168afe47e688903998f3f98543 : FAILURE in 5m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81f35bc50a024d6fab85fae92cbdf99b : SUCCESS in 5m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/89a7beea345e490e9db20753165e2159 : FAILURE in 7m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5230251b34b1442497df4dc8b4e9a7c2 : FAILURE in 8m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6335ff5612654422a0b5821b0208c88f : SUCCESS in 12m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2c0ef92deb644062bd87b175b58df385 : SUCCESS in 7m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/de458ae1fd884257b1d6b712583f039b : SUCCESS in 11m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/341771c7ca0946fa81610ebb79b94d72 : SUCCESS in 7m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/14c3b3ea60b54e9a867a1b2d6443d786 : SUCCESS in 8m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9e05476981624ca9b5d92f7691a75805 : SUCCESS in 20m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/43e976ee403a42e290073bc78afac669 : SUCCESS in 31m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/07e5b5d777ca4d3fb3ffd57102d12b17 : FAILURE in 17m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f2e4b1fc5c714d71bc5851d88424a847 : SUCCESS in 6m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40912a2a13fb44a7830e37a08f87f411 : SUCCESS in 21m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1ed110b239694f1a8d390271dfbd3984 : SUCCESS in 6m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f51cda0ceca2440ab4d225687858ee1e : SUCCESS in 8m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4bff6048f1754c57ad0827b6505622ad : SUCCESS in 12m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d28d45a20b564578a4270ed5510cc2e1 : SUCCESS in 53m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07fe97e23f6445fd991d920299fa23d4 : SUCCESS in 1h 17m 51s","accounts_in_message":[],"_revision_number":2},{"id":"b1e048ca532a8ae4253240f978e11cebce7d2ddb","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-23 06:18:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7b8d28d161334f6871547a93176f570d0acb36ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 06:31:49.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"e9ee0d4021d72b587efa8e56658b8970b9016291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 09:11:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d2a85935a44ce1b2199fd001319ea2 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cd7e394806945d987a2c5162e429378 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28c03bc2d5c94962a5791683b073d6e4 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c4e696b6a814f839e2c8f6e65a9d362 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d2de4669c34462d8f3d87c8a766b80d : FAILURE in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a8f95790cff436b89aa93bc5c2ca8b6 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e5dec7297f1f43b089165fbfc2cb31b1 : FAILURE in 7m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/65c56eb91e704601b1e3f7184ef929a2 : FAILURE in 7m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ab1ffb61490a4ac685089c203ff05613 : SUCCESS in 11m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fe33105cb0204867a5d88d9d3fde249d : SUCCESS in 6m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9a82f08af2c84cbca4953b20e0cb1488 : SUCCESS in 13m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3a19ed3c1da94967b9548c1d9db68443 : SUCCESS in 8m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d729e6a634b64aa6b675a4bdfe4ee86f : SUCCESS in 9m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d968711a1b82426294777ebf964a2c12 : SUCCESS in 15m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b200b07d89e840bd90d83bfc384ebb2e : SUCCESS in 24m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2a5eef09661a4a448649aeaffe2c1add : FAILURE in 21m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8bf7f8ccd5814aaaae7e35f51b9b84fe : SUCCESS in 6m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a38eb008c90243dba5914b6fcd1e86e7 : SUCCESS in 19m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d1199a0e45a94832a5a8086d572b482f : SUCCESS in 6m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c3601823b8294c6b912ac3d348667d8b : SUCCESS in 8m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7d004982ce0d44c990783550f54920a9 : SUCCESS in 11m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d3ae80b4f7d40b5b7790ccdb891ef3e : SUCCESS in 56m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f348512ee384583a8b97cf7a35a7bc7 : SUCCESS in 1h 28m 26s","accounts_in_message":[],"_revision_number":3},{"id":"14471eb38c4fa9e9f07e520683475e8b8fad9915","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-23 22:09:24.000000000","message":"Abandoned\n\ngonna remove for a reset","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"183271bfff085bc1215ab863923343fae66659d1","revisions":{"0668e9916ec10a63b1874bf752376a8b9bf15a95":{"kind":"REWORK","_number":1,"created":"2020-10-21 06:21:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/758947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/758947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/758947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/758947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/758947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/758947/1"}}},"commit":{"parents":[{"commit":"d447ce59fdd0c5ca992486af35dbd5d1008181a7","subject":"Merge \"update ceilometer to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d447ce59fdd0c5ca992486af35dbd5d1008181a7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-21 06:21:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-21 06:21:48.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I32eb5479e7eef90eef5967891c57f16f60cf0c1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0668e9916ec10a63b1874bf752376a8b9bf15a95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0668e9916ec10a63b1874bf752376a8b9bf15a95"}]},"branch":"refs/heads/master"},"624b7b63c23af3b7b48d43a5c9e6b226de579f7e":{"kind":"REWORK","_number":2,"created":"2020-10-22 06:18:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/758947/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/758947/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/758947/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/758947/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/758947/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/758947/2"}}},"commit":{"parents":[{"commit":"fe2af414297dbf8d6112df7c3440a868e6dd9746","subject":"Merge \"update oslo.messaging to 12.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe2af414297dbf8d6112df7c3440a868e6dd9746"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-22 06:18:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-22 06:18:32.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I32eb5479e7eef90eef5967891c57f16f60cf0c1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/624b7b63c23af3b7b48d43a5c9e6b226de579f7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/624b7b63c23af3b7b48d43a5c9e6b226de579f7e"}]},"branch":"refs/heads/master"},"183271bfff085bc1215ab863923343fae66659d1":{"kind":"REWORK","_number":3,"created":"2020-10-23 06:18:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/758947/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/758947/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/758947/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/758947/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/758947/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/758947/3"}}},"commit":{"parents":[{"commit":"a24ca0b232d8feb938bb5989e9f870680804ea7f","subject":"Merge \"Update ovsdbapp upper constraint to 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a24ca0b232d8feb938bb5989e9f870680804ea7f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-23 06:18:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-23 06:18:27.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I32eb5479e7eef90eef5967891c57f16f60cf0c1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/183271bfff085bc1215ab863923343fae66659d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/183271bfff085bc1215ab863923343fae66659d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
