)]}'
{"id":"openstack%2Frequirements~759101","triplet_id":"openstack%2Frequirements~master~I85b2b28d77704914c88137c63770825e588fc6d9","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I85b2b28d77704914c88137c63770825e588fc6d9","subject":"Update ovsdbapp upper constraint to 1.6.0","status":"MERGED","created":"2020-10-21 16:26:15.000000000","updated":"2020-10-22 16:12:23.000000000","submitted":"2020-10-22 16:07:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759101-1603382848377-8f9d6ad0","meta_rev_id":"1f9339b13297b4c40fa4bd14537170dc336002ba","_number":759101,"virtual_id_number":759101,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 16:07:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-21 16:51:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-10-22 14:07:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-22 02:53:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2020-10-22 02:34:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-10-21 21:43:15.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-10-22 14:07:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-21 16:51:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-21 21:43:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-22 02:34:27.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2020-10-22 02:53:47.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-10-22 14:07:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-22 16:07:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e6f658df14dd7cf5ac71d57ccc097f2b70ee95a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-10-21 16:26:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96f0a97efbb48d54146b2e8ee0588dea6df9f656","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 16:37:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"40d19e6ec856f3666c8ee8ca16fc3123f0d1eeff","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-21 16:51:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"913075e7ae40412e6ca5e54c17f0680d6e272aa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 21:36:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a579ec71ab3a4deb88dad86e439df925 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50c908e809244377a8244f9b658dc0d1 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33dba757917b4c31a81edcef018ffca9 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/683b1400f7514441bb79df411421b15a : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38a4f21d18184ca9843608a8bd2893c4 : SUCCESS in 7m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab0c03baa1e04b28a070dcaadf70341f : SUCCESS in 5m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bdc9b6e8292842019f75b07c33361552 : SUCCESS in 9m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/401d50f042fb4007a2da42b0a0872b5c : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4174a063596747238f8a2d29a0f35b86 : SUCCESS in 12m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e6eb5474f49d49e2b5a86b24d1f1a405 : SUCCESS in 6m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a6fbf32fbe084214a163ea618fba6086 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f680c7237c247ffa5dc50d888aaf091 : SUCCESS in 8m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7a41ea0ab61b4f518f5ede2f5e708a2c : SUCCESS in 8m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5b7e86c3660d4dfdb5983895e21c5e33 : SUCCESS in 20m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/76aec743ea87492e9abe9c138c7e5906 : SUCCESS in 28m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1c4bfd13a4f1434da6cdb7b11c1f4a55 : SUCCESS in 13m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a465d7e4f5d5455dbb01bef42684096d : SUCCESS in 6m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a62eea8914943fa96bb35b5c931c64c : SUCCESS in 15m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9849f7d824304d6fa3084f66bf5c0c0b : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/430941a821a54a28bbab4fbcf0e68126 : SUCCESS in 9m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/166e8a54c97047fba02a35c1abde40ef : SUCCESS in 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71a9d8fd5f5b47808fd6397682b092e0 : SUCCESS in 1h 05m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77a3996871584a77a15c7fd3f983aa96 : SUCCESS in 1h 28m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7a2efbdeaaf78858cf53ec9cb49517fbbe07ec91","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-10-22 02:34:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e81b72cd1476b313e62e3c2a8d33abc9161792d4","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-22 02:53:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c762c07797149d192e6f57abd122d1659b3d9a7d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-22 14:07:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM https://opendev.org/openstack/releases/commit/8779d7453c67816e2c50313793d6a74e191a5223","accounts_in_message":[],"_revision_number":1},{"id":"789d25314a95be3e2f707cee1466f0cd9880cee0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 14:08:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb2ded50352b493c797518d18a590e26a12073c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 16:07:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39520bb758dc4ab5b3e9fa1152e2da55 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75f513a33114452db97b2a377b41c0ac : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a5372796621420f9e3c9d29db5b4e4f : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f1461268f0342e6ac219949fbe08a30 : SUCCESS in 8m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e07074e6d5c24950a2d152fd79abfbbe : SUCCESS in 7m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00bdf03538d54c05aaa413a167162e44 : SUCCESS in 6m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a6db07ad198247c0873c1bf03e9bda2f : SUCCESS in 10m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/87d610b340414075b2ff3f8b2af45e5c : SUCCESS in 10m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d89fdfe9f0b240d79532040136f80a4a : SUCCESS in 11m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/56958036c86c40c8a7a001de2ededd86 : SUCCESS in 7m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/40d6a91cd06340eea3d3c2f2061fdf88 : SUCCESS in 14m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d8e5955979494457b87826ddc99419d8 : SUCCESS in 14m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b9289dba6f17407a9b9e66733e5fd544 : SUCCESS in 14m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/03081de7a63b44d6b6a52d6d3ef3b223 : SUCCESS in 17m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3b17ee7bded9429d84f2dd526c5d5762 : SUCCESS in 31m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/668fa06ca3d84389af8a44b82b2506a5 : SUCCESS in 16m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/527f0a6386e24a90996869bad923b9cb : SUCCESS in 7m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55649307e70b40aba4672d530ec93b75 : SUCCESS in 20m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c644f075b67a4c658a97b8046bff3d18 : SUCCESS in 8m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ea60432cb79043b5aff166756d60c93a : SUCCESS in 9m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/15be111652244c338bfa583de3531017 : SUCCESS in 14m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55edd379d9d64e53b4ba3d04d51f7bdc : SUCCESS in 59m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25391935954c4f1d99e9f3044cd014ec : SUCCESS in 1h 25m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c384c14912f3d860160d2d4533ab252a297c28de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 16:07:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1f9339b13297b4c40fa4bd14537170dc336002ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 16:12:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b37a079c0e41f985000e454e1f8d7b : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7eb5e4c1fa7e9a1ec114e2d662638d9085dfe4d","revisions":{"d7eb5e4c1fa7e9a1ec114e2d662638d9085dfe4d":{"kind":"REWORK","_number":1,"created":"2020-10-21 16:26:15.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/01/759101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/759101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/759101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/759101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/759101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/759101/1"}}},"commit":{"parents":[{"commit":"d447ce59fdd0c5ca992486af35dbd5d1008181a7","subject":"Merge \"update ceilometer to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d447ce59fdd0c5ca992486af35dbd5d1008181a7"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-21 16:26:03.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-10-21 16:26:15.000000000","tz":0},"subject":"Update ovsdbapp upper constraint to 1.6.0","message":"Update ovsdbapp upper constraint to 1.6.0\n\nChange-Id: I85b2b28d77704914c88137c63770825e588fc6d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7eb5e4c1fa7e9a1ec114e2d662638d9085dfe4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7eb5e4c1fa7e9a1ec114e2d662638d9085dfe4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
