)]}'
{"id":"openstack%2Frequirements~759550","triplet_id":"openstack%2Frequirements~master~I9bc0d31660ad37a95a0fc5d8fcfedd502aef2bd5","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I9bc0d31660ad37a95a0fc5d8fcfedd502aef2bd5","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-10-24 06:16:11.000000000","updated":"2020-10-24 21:22:45.000000000","submitted":"2020-10-24 21:21:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759550-1603574496455-c854e23f","meta_rev_id":"514b355dcc35cb42c4d33e1724e7bf78cf434e3e","_number":759550,"virtual_id_number":759550,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-24 21:21:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-10-24 18:55:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-10-24 18:55:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-24 18:55:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-24 21:21:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6863c72b3cd1a548fbca06af6eeb574f40633915","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-24 06:16:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28534e95fc2629d1d1e02fe317d7458ec8dabcc0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-24 06:23:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0379fdbd190f3bb49a03fc90dd67b8bd3bf5fa7f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-24 06:23:53.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"64ef408eef90ca522044a29ef14131013cb75173","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 06:46:14.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/60a935d7fd4345c3aef80cb8be698430 : FAILURE in 15m 41s","accounts_in_message":[],"_revision_number":2},{"id":"f99930a9df6973873c45e5fa1a3069c898f1c33c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-24 07:28:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"36e73f44ed36bc01d12d78c8c077bfe090ad7b74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 07:50:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4a3edcf460e24f768578d00980b91c9e : SUCCESS in 15m 43s","accounts_in_message":[],"_revision_number":3},{"id":"187ac0a41beceedfb50a95786e9065d0c3f84438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 09:08:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8fb14906c734b678289b8d775fc9c81 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eabaaf2ac4e43c5b75e472011e44638 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a6b1a5aa5124f9ca75d89ae15c40007 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd4acb7de617409d9cc4c3f9f1f0a294 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af57f8c1be9a4fcbb2aaa10059a22a8a : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f7ed3ea1ed33428aaba554086f8201b7 : SUCCESS in 3m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/23ad3bb11e554094aba3a9c256ea65d9 : SUCCESS in 7m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8ca3750f1dd64c6095dd6b952fdd4c02 : SUCCESS in 7m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d6739dfd78b2496589327316449711cd : SUCCESS in 9m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fbda66e054ff4a9b891c98a9f5ae4b32 : SUCCESS in 4m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fc96413801d7488b8621ac3edb833316 : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c41247cd343b4be39f6b22404aeeaef9 : SUCCESS in 6m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6052dbec0dc948039c27e1c08d6c30f3 : SUCCESS in 7m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/65b0f871a92c4db69e695fdac46ce917 : SUCCESS in 39m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fffa92fc03a640fe9387814f224ba5c4 : SUCCESS in 21m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ee0f8ac3fb9740ba9c5479905f15e2c4 : SUCCESS in 11m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f355fe4f56324be095ca9ba638137a85 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4f6aa8b47f74f0293c3017025601715 : SUCCESS in 21m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/af168ae2d9154bb7b7f3a97b25f9a5bd : SUCCESS in 7m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9fedb132b9f44cda8e7fa1681778a471 : TIMED_OUT in 31m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4f88880d251b486a9938a3fbdb8ef0f7 : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d1ebec37f6840ecb07de0204e8e85ee : SUCCESS in 1h 33m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f304f59d96f3474fbf56e11bbbdcad86 : SUCCESS in 1h 13m 13s","accounts_in_message":[],"_revision_number":3},{"id":"766e64dacab55d55776341f0d76ffc1101e327ab","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-24 18:55:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"23be52952b05658d4cd9ecde0e3ce06e20093477","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 19:17:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cd63949ce6344cbdb526365fae34dda1 : SUCCESS in 18m 06s","accounts_in_message":[],"_revision_number":3},{"id":"f66d9a5919337a8854acc9708998cf7214f28a81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 20:08:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/311aabefedf84e09af631a68fb5e750c : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/524c3d083f3746dc934109b36ee1e312 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c77f3df56b6413697eb2968ca972968 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/712b13809194495082cfb18aa4830852 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc47e2c3e8344b88ac33869e05b0ef1d : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fe9331a91da14b67b4bbe7e9b0aec3b3 : SUCCESS in 3m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6e1288fd3c264522b84e5a9c73e46028 : SUCCESS in 7m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f83233e2761048c686fbeba8e8c3bd61 : SUCCESS in 6m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cc83cbbfb3174bb08bdf3914b25070cd : SUCCESS in 10m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4a2a868069d54d47a9e4718f5d4b8116 : SUCCESS in 4m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/03dd678debdc4985b3e1c6ca333a6b07 : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f98eecd0583474c87620732346760a8 : SUCCESS in 5m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b82d110f410242d0939f7eb0ebe36ae1 : SUCCESS in 7m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9a92b5230c804e3cafa36ec12b7f2e3b : SUCCESS in 12m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6d96877eeab143e08adccf5c56439707 : SUCCESS in 23m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1f4be6a4ee0849f89da19c0afa9af615 : SUCCESS in 12m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ad6558efe363424b846d1c48a070c8b9 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6078b917c23c4b45b0a279b961c879b2 : SUCCESS in 18m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ba984023a994464abdc5ca02c04a93b7 : SUCCESS in 4m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5721481b83ab4aadad2d91927cfce0de : SUCCESS in 6m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8b34e7d6df514f05b511a758cf3c8f51 : SUCCESS in 9m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/618ee8925cc14635a134c00134e167b0 : SUCCESS in 47m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ea67a7d9f2f44778729dff6e4c07e81 : SUCCESS in 1h 12m 17s","accounts_in_message":[],"_revision_number":3},{"id":"bc674597bd83fa8209c356e0b1c0ac495a69bb0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 20:09:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7da12d4e90fad896cc3040e302ed76208084d26b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 21:21:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b8f044e7bffbc92e6cb46dace5fcce580ff3b4d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 21:21:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce29891fb6b418780ef05eabde982f5 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/948ce40544104476841ffdb47e782567 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43a0f4989dae405296ccc3328e473be7 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f48a72b6b0243aa8a3e1c57f60bd2bd : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9bbd6f03f76349aeb9702f4156b5e0c8 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/631867ff090c42baad9722c2f4c9ab0b : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5561903f4444454f8af781923371a418 : SUCCESS in 7m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a385c38e54744a559a0dfd553cc206bc : SUCCESS in 6m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c38b352175db446d9f08065957250c97 : SUCCESS in 8m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/46d1bb201516452f84678d1c1a747171 : SUCCESS in 4m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/192b1d04f2d64ef6b96276c1586e01cf : SUCCESS in 6m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eebfdd7f10544e0887b1a42d0df387df : SUCCESS in 11m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a838da27c9664ebe88fbde6521e3f5dc : SUCCESS in 7m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/981143dc2ce44a55bb678c6c5210cc35 : SUCCESS in 34m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/deeb23f293724d08a64d3234288ca816 : SUCCESS in 28m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a37f0146813043ec8eedc1dcbc774c16 : SUCCESS in 12m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/306491f1da2f4a068d56b03edbb10342 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bdc76998f6994d9fa8d642498a445539 : SUCCESS in 18m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/324155b482294c5d96dd200a232fab2a : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0f4688fe546c418da71e731d7bfb216f : SUCCESS in 5m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6b18868f556743528d218a7825c5cd09 : SUCCESS in 9m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2b790ad88cf4c169b2e0bb06142dad3 : SUCCESS in 54m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5229052a27604f208b654f19004ae605 : SUCCESS in 1h 11m 00s","accounts_in_message":[],"_revision_number":3},{"id":"514b355dcc35cb42c4d33e1724e7bf78cf434e3e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 21:22:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8072f067b30e46b382a9cd1f46fa812f : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"11ac0414d8c7b5c94b5dcc9c183c091a1926e9d0","revisions":{"0144a30b54913621fa213656ccee9bed876649e9":{"kind":"REWORK","_number":1,"created":"2020-10-24 06:16:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/759550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/759550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/759550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/759550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/759550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/759550/1"}}},"commit":{"parents":[{"commit":"47a2a17179d95f1b9bfc5a90b78868c7a1312d5b","subject":"Merge \"update constraint for os-win to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47a2a17179d95f1b9bfc5a90b78868c7a1312d5b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-24 06:16:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-24 06:16:08.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9bc0d31660ad37a95a0fc5d8fcfedd502aef2bd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0144a30b54913621fa213656ccee9bed876649e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0144a30b54913621fa213656ccee9bed876649e9"}]},"branch":"refs/heads/master"},"2fc016234f1d33973bf0fc3606ca70ce950c881c":{"kind":"REWORK","_number":2,"created":"2020-10-24 06:23:51.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/50/759550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/759550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/759550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/759550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/759550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/759550/2"}}},"commit":{"parents":[{"commit":"47a2a17179d95f1b9bfc5a90b78868c7a1312d5b","subject":"Merge \"update constraint for os-win to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47a2a17179d95f1b9bfc5a90b78868c7a1312d5b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-24 06:16:08.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-10-24 06:23:36.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9bc0d31660ad37a95a0fc5d8fcfedd502aef2bd5\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fc016234f1d33973bf0fc3606ca70ce950c881c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fc016234f1d33973bf0fc3606ca70ce950c881c"}]},"branch":"refs/heads/master"},"11ac0414d8c7b5c94b5dcc9c183c091a1926e9d0":{"kind":"REWORK","_number":3,"created":"2020-10-24 07:28:41.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/50/759550/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/759550/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/759550/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/759550/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/759550/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/759550/3"}}},"commit":{"parents":[{"commit":"47a2a17179d95f1b9bfc5a90b78868c7a1312d5b","subject":"Merge \"update constraint for os-win to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47a2a17179d95f1b9bfc5a90b78868c7a1312d5b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-24 06:16:08.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-10-24 07:28:34.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9bc0d31660ad37a95a0fc5d8fcfedd502aef2bd5\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11ac0414d8c7b5c94b5dcc9c183c091a1926e9d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11ac0414d8c7b5c94b5dcc9c183c091a1926e9d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
