)]}'
{"id":"openstack%2Frequirements~759552","triplet_id":"openstack%2Frequirements~master~I5990b00f94305813780854698fef2d3e1e721e79","project":"openstack/requirements","branch":"master","topic":"bug/1901383","hashtags":[],"change_id":"I5990b00f94305813780854698fef2d3e1e721e79","subject":"revert libvirt-python upper-constraints bump to 6.8.0","status":"ABANDONED","created":"2020-10-24 06:55:09.000000000","updated":"2020-10-27 20:14:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"004362cdbefe049120c8f3cae7709114d32d4040","_number":759552,"virtual_id_number":759552,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"date":"2020-10-27 09:03:36.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-26 11:46:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"all":[{"value":-2,"date":"2020-10-26 10:31:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-10-26 10:40:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-26 10:38:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2020-10-26 10:33:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-24 18:55:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-26 10:33:06.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2020-10-26 10:38:28.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-10-26 10:40:09.000000000","updated_by":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"REVIEWER"},{"updated":"2020-10-26 11:46:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-27 09:03:36.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"}],"messages":[{"id":"2f04dcd9ce5998105a3906e3b6b1833d87a7f962","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-10-24 06:55:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9b9ff81eccce5b2ce9d7e037773e3dbc630e752","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 07:13:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/58df927fbff146e59a585f432e86c12f : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f52c0bbb7cd52d0b0c9b7701ab5f7c18d68a0124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 08:21:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e8781d29d924b06924b10fc6b26318f : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66bdc018fb3b42f0b8e586fc10b5f321 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3187e03838444278c12584a3f84013b : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a1a3d61aef245ca9ea9de46f6da80e1 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f194fc07c2cb42f5abde828169793138 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a8b69485bc4a45b39dbc2ba2a65131e9 : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/73725341bc7148aa808c571cf7738837 : SUCCESS in 7m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/029903a8088a48cea61157fa5285259e : SUCCESS in 7m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b571833a5add4a93bc448d49da4c8d9c : SUCCESS in 8m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/67da018354ba46f1b574b2c61c726f0b : SUCCESS in 5m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f28376076142494fbc95539edf1a0fcc : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3773f6a749464e1ab48312c6517f6d5d : SUCCESS in 6m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e4946d5382d2476786edbe1fb66216e9 : SUCCESS in 7m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4bdd831f2d2c4975b1f861fc411f1494 : SUCCESS in 12m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/32983a08363248b18462b9c9fad5a6a3 : SUCCESS in 23m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e774a569e2094c339a37e20bde34ef47 : SUCCESS in 13m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c139c16a3cf34965819bd5dd6b245bce : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55960714a6bd4011bc52cf144ba467e6 : SUCCESS in 17m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e6dcfecfb50c4f5c9aa0091582853dbd : SUCCESS in 25m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7c714a765af42d1958215c45e4d01c8 : TIMED_OUT in 30m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a3e512d9318a4b76b975f886aa5edf43 : SUCCESS in 11m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/056df233d7ff40d2ac7041979e2d182d : SUCCESS in 49m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b46972c6a9fc4b03b3401058d2d63826 : SUCCESS in 1h 23m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3b60a3c307ba980ba7bf718211a4a00615f78317","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-10-24 10:26:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3df68638a77ef6204428c9ed3da3fd7cec8880ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 10:47:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5e470cd93ec64b9f8a0823ea568b1db2 : SUCCESS in 17m 16s","accounts_in_message":[],"_revision_number":2},{"id":"8f5ea4358b0e325cd966735e76247a9e54dab7d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 11:43:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf674aad01ad494dbabf81a13b3e1d1d : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff91599214ee4c5c98302fc0372fddc6 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c5c417441264efab3dce49c0c7b0498 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bdecec0300f491488eaebf26191631a : SUCCESS in 5m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a1250b69cd742aabbb8a25c541beb86 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0346cb4a75348908cf86923a378ee10 : SUCCESS in 10m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/844dba47927b46e8a99ef6ae403626b5 : SUCCESS in 7m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8ed6af2ccea24f00838b0361c2284f43 : SUCCESS in 9m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f974c4eace914fda93e2d4ba3d8ef1be : SUCCESS in 12m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f673daf9f99b44b9a97ad4611c0fe606 : SUCCESS in 5m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f6c8d9e832714fe7a299c4c40f348c98 : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e9df0b943ad048d49b6eac4285dcdc6a : SUCCESS in 6m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6c05184a0f1944c9a12b4d2db3cca18f : SUCCESS in 6m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5ff7c4393dd7438e84187bf07bd3eea3 : SUCCESS in 19m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1d5d8ee6f6c141d297a8f33df9727acc : SUCCESS in 23m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/44960867fcb1441aa5b38b47303e7196 : SUCCESS in 16m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2e54970c0ecc4cf79200352fe1c4d2cf : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/101c1004655240999df65acecc842b47 : SUCCESS in 16m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cd2af8085b07472ca0fd28d0e852e438 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f134b654239f41958867b8f92c138907 : SUCCESS in 5m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3794cfba5b66419da1abc45b93392ecd : SUCCESS in 9m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c321aeacb434bc5a2c851075dd13d36 : SUCCESS in 49m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d39ec2bda0844ce18053c0ae0c7f3d00 : SUCCESS in 1h 15m 41s","accounts_in_message":[],"_revision_number":2},{"id":"bf6083b5adf310546c326a1d8b413fc38d2fb086","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-24 18:55:24.000000000","message":"Patch Set 2: Code-Review-2 Workflow-1\n\nDNM\u0027ing","accounts_in_message":[],"_revision_number":2},{"id":"30dd12ce452fa016128b281071dee21327f3657c","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-10-26 09:52:04.000000000","message":"Patch Set 2:\n\nIMHO given this unblocks the gate I think we should push ahead with this revert, adding Related-Bug: #1901383, while we also file a bug with the libvirt-python folks. I\u0027ve pinged danpb and kashyap about this downstream.","accounts_in_message":[],"_revision_number":2},{"id":"754cd246486a4a0d77ae4a478d6857b4534dc24b","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-10-26 10:31:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f410720d204b82f56f88ef3deacfbb5ea545e2d8","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-10-26 10:33:06.000000000","message":"Patch Set 3: Code-Review+1\n\nI agree with this.","accounts_in_message":[],"_revision_number":3},{"id":"ca86a3f5d27ec0223e72440db3527aaed3c01abc","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-10-26 10:38:28.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"67ddb2a453ff9bb6b5a4d136c7e06845133beb58","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-10-26 10:40:09.000000000","message":"Patch Set 3: Code-Review+1\n\n(Talking to Dan Berrangé from libvirt on IRC, he guesses it might be due to the typing hints added in the libvirt APIs.)","accounts_in_message":[],"_revision_number":3},{"id":"6e7f4cbee127b2fa530aefebaadda5776959351e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 10:49:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2c3408b6dc8a42ae9289c000fee1fd9e : SUCCESS in 13m 52s","accounts_in_message":[],"_revision_number":3},{"id":"200243f90a3a936bb9457b2b18f016da257853a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 11:46:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24a0b63c65d0444bb763cb04c9b64341 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34e6a3bea3e949e3bd1396a6a9a6c6a8 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e39f6d0ee034ad3a4966ceb7279c7a4 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eff711c3ce5546dc9148ba4ece16b36d : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e83b22567b3f4732a84075a28c204c10 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3a22cb3b006243b7adfbafd7a4fb471d : SUCCESS in 3m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9eb64d9573824aa5939b2c8e4368d4e1 : SUCCESS in 18m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/99502bed3c70413ab7170b3b0946cd67 : SUCCESS in 10m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4464d2d911384536a9ca4881e4939ffe : SUCCESS in 11m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/974df0792ec64b40b32bd09c9d5b29a4 : SUCCESS in 5m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a50b06e6e3094e0d8a379d969fe8f3e4 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da91a43988094281b93dbdc089abb236 : SUCCESS in 6m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a82d87cf0b91484a9f861d8e20fbe1d2 : SUCCESS in 6m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/15d0656fff3c4a8da09862b3193c7a0d : SUCCESS in 12m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2e2acc9f7fd84ca1a5fd81dd4e4aadf4 : SUCCESS in 29m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/afc964ad64f64d26a83b9d75270a13a6 : SUCCESS in 12m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7dfad99be9c4496894db13eef45ad35c : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/883ab7d890de40f99a07a2778efa7e7e : SUCCESS in 22m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c8571c391702436c93e1a42df26c9f5f : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f5ebce86614472eae88c78b05d1c522 : SUCCESS in 5m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c2737b68e21147d4ad749f23b15a034c : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ebc6f7fa017441688ab11cf6c1592a7 : SUCCESS in 49m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c67e0145c9f74537a5728bbf752ea0a6 : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":3},{"id":"ff4da64ec3e1f4f47419445b8b8f0f0543d8eb5c","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-10-26 12:35:40.000000000","message":"Patch Set 3:\n\nSo, Dan Berrangé further narrowed down the root cause:\n\n[quote]\ntpool + Python 3.6.x + type hinting is fubar\n\nIf I add \u0027tpool\u0027 into my [libvirt-python] new unit tests, then it crashes \u0026 burns with the error msg Mel got\"\n[/quote]","accounts_in_message":[],"_revision_number":3},{"id":"06e1bbfbbfb6fbad721b0cab5d486af3e0712593","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-26 14:05:43.000000000","message":"Patch Set 3:\n\nso, is this a tpool/eventlet issue?\n\nthere are a couple of reviews related to that\nhttps://review.opendev.org/759691\nhttps://review.opendev.org/753239","accounts_in_message":[],"_revision_number":3},{"id":"5030c2cbb5c08359dcc1130c0d047ba1b79ba3bf","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-10-27 09:03:36.000000000","message":"Patch Set 3:\n\n\u003e so, is this a tpool/eventlet issue?\n \u003e \n \u003e there are a couple of reviews related to that\n \u003e https://review.opendev.org/759691\n \u003e https://review.opendev.org/753239\n\nI\u0027m unclear on the link tbh but I\u0027ve pushed I568b0c4fd6069b9118ff116532f14abb46cc42ab based on some feedback from the libvirt devs in the bug so Nova shouldn\u0027t attempt to proxy these typed classes anymore.","accounts_in_message":[],"_revision_number":3},{"id":"776b58915998dbcec7139b57f389da3e18c076a4","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-10-27 20:14:50.000000000","message":"Abandoned\n\nWe were able to solve the gate failure with the following fix in nova:\n\nhttps://review.opendev.org/759831","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"45c3291a9675be1e8ddaab2b49e033cb12698180","revisions":{"7e84892ab52aec662e822397b807bf9cf1fae459":{"kind":"REWORK","_number":1,"created":"2020-10-24 06:55:09.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/52/759552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/759552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/759552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/759552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/759552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/759552/1"}}},"commit":{"parents":[{"commit":"47a2a17179d95f1b9bfc5a90b78868c7a1312d5b","subject":"Merge \"update constraint for os-win to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47a2a17179d95f1b9bfc5a90b78868c7a1312d5b"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-10-24 06:52:43.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-10-24 06:54:59.000000000","tz":0},"subject":"DNM revert eventlet upper-constraints bump to 0.28.0","message":"DNM revert eventlet upper-constraints bump to 0.28.0\n\nPartial revert of change I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\nsuspected of breaking nova-compute with:\n\nFile \"/usr/local/lib/python3.6/dist-packages/eventlet/tpool.py\", line 149, in proxy_call\n  if isinstance(rv, autowrap):\nFile \"/usr/lib/python3.6/typing.py\", line 1162, in __instancecheck__\nreturn issubclass(instance.__class__, self)\nFile \"/usr/lib/python3.6/typing.py\", line 1148, in __subclasscheck__\n  raise TypeError(\"Parameterized generics cannot be used with class \"\nTypeError: Parameterized generics cannot be used with class or instance checks\n\nChange-Id: I5990b00f94305813780854698fef2d3e1e721e79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e84892ab52aec662e822397b807bf9cf1fae459"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e84892ab52aec662e822397b807bf9cf1fae459"}]},"branch":"refs/heads/master"},"e858a8f855dbb1e8251eb0c3e873fa05bc3e7114":{"kind":"REWORK","_number":2,"created":"2020-10-24 10:26:43.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/52/759552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/759552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/759552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/759552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/759552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/759552/2"}}},"commit":{"parents":[{"commit":"47a2a17179d95f1b9bfc5a90b78868c7a1312d5b","subject":"Merge \"update constraint for os-win to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47a2a17179d95f1b9bfc5a90b78868c7a1312d5b"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-10-24 06:52:43.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-10-24 10:26:22.000000000","tz":0},"subject":"DNM revert libvirt-python upper-constraints bump to 6.8.0","message":"DNM revert libvirt-python upper-constraints bump to 6.8.0\n\nPartial revert of change I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\nsuspected of breaking nova-compute with:\n\nFile \"/usr/local/lib/python3.6/dist-packages/eventlet/tpool.py\", line 149, in proxy_call\n  if isinstance(rv, autowrap):\nFile \"/usr/lib/python3.6/typing.py\", line 1162, in __instancecheck__\nreturn issubclass(instance.__class__, self)\nFile \"/usr/lib/python3.6/typing.py\", line 1148, in __subclasscheck__\n  raise TypeError(\"Parameterized generics cannot be used with class \"\nTypeError: Parameterized generics cannot be used with class or instance checks\n\nChange-Id: I5990b00f94305813780854698fef2d3e1e721e79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e858a8f855dbb1e8251eb0c3e873fa05bc3e7114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e858a8f855dbb1e8251eb0c3e873fa05bc3e7114"}]},"branch":"refs/heads/master"},"45c3291a9675be1e8ddaab2b49e033cb12698180":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2020-10-26 10:31:55.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/52/759552/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/759552/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/759552/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/759552/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/759552/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/759552/3"}}},"commit":{"parents":[{"commit":"1f550ffc402181aeaba2886bc7a0d4aba185a08b","subject":"Merge \"update constraint for pbr to new release 5.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f550ffc402181aeaba2886bc7a0d4aba185a08b"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-10-24 06:52:43.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-10-26 10:31:26.000000000","tz":0},"subject":"revert libvirt-python upper-constraints bump to 6.8.0","message":"revert libvirt-python upper-constraints bump to 6.8.0\n\nPartial revert of change I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\nsuspected of breaking nova-compute with:\n\nFile \"/usr/local/lib/python3.6/dist-packages/eventlet/tpool.py\", line 149, in proxy_call\n  if isinstance(rv, autowrap):\nFile \"/usr/lib/python3.6/typing.py\", line 1162, in __instancecheck__\nreturn issubclass(instance.__class__, self)\nFile \"/usr/lib/python3.6/typing.py\", line 1148, in __subclasscheck__\n  raise TypeError(\"Parameterized generics cannot be used with class \"\nTypeError: Parameterized generics cannot be used with class or instance checks\n\nRelated-Bug: #1901383\nChange-Id: I5990b00f94305813780854698fef2d3e1e721e79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45c3291a9675be1e8ddaab2b49e033cb12698180"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45c3291a9675be1e8ddaab2b49e033cb12698180"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
