)]}'
{"id":"openstack%2Frequirements~759691","triplet_id":"openstack%2Frequirements~master~I524fc271797cfc24a284b9f1f65ecc641771f914","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I524fc271797cfc24a284b9f1f65ecc641771f914","subject":"DNM revert eventlet upper-constraints bump to 0.29.0","status":"ABANDONED","created":"2020-10-26 13:25:14.000000000","updated":"2020-10-27 09:15:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d9cbd563fc3cce64fa7fab92a5d1eb59f51cd2eb","_number":759691,"virtual_id_number":759691,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-26 14:56:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2020-10-26 13:25:14.000000000","_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-26 13:25:14.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2020-10-26 14:56:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e373d9b98f80eaac694104b59c8f9d8c7a110501","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-26 13:25:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9049d45e8ff5347329709aefa525fefbcd115610","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 13:46:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a88bcfbb45404665897539498176d7ed : SUCCESS in 17m 42s","accounts_in_message":[],"_revision_number":1},{"id":"96f7b94ead2fb2ed4afd0c247b198ed476636d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 14:56:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af69b863d7074770a121448a32c1baee : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/011dda0efc6a4e539b1d1b84b2cacdfb : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2872ef63a289421f94330969bc6d94af : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d2b5dadb5ab44ebb158fbb72807c668 : SUCCESS in 7m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e994b8301a54441ac5cc5bd344fa14d : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f77e4d5338ee452d9585e8690b5290d4 : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f155c498e3484ded9c6b18d3a7d1ba76 : SUCCESS in 7m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0f2bda7201d844b0bcc94afcda864bed : SUCCESS in 8m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d0715cf55c00432f8c180e4f8e85861b : SUCCESS in 10m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/02019df731254d20a3674fe25e17ed90 : SUCCESS in 5m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3b6d0e1ac5cc435799a36c5bb6b82333 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e87b7d087454506991954a78a0339da : SUCCESS in 6m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/91eb768093c14b768b1e765ead253404 : SUCCESS in 8m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/64c4e50bed014f59a39b97d0b0c94773 : SUCCESS in 15m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/51066b2cefe6426cbbc6bb7eae8ceb6f : SUCCESS in 26m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d05346989dd54631947c57e3644d598d : SUCCESS in 13m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/30f7a6f26a8947e2b0d2e0ebfe52f972 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/51b7bcf09e8340daab30687d625711d2 : SUCCESS in 17m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/06708f0458af47b1b73543f1749ee58f : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/345f83c00d2c4a9485bcb965bc2f2ef8 : SUCCESS in 6m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0e2cc0fbcf864304ac5dce884559cb17 : SUCCESS in 9m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/044ba77fdb7c4e04a46dbd9b9fdd8b41 : SUCCESS in 55m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/479eed4d905b4d50acd586ff40464682 : SUCCESS in 1h 15m 12s","accounts_in_message":[],"_revision_number":1},{"id":"69f2b267cb6d12786e757688a2cf5b0c6cf13363","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-27 09:14:26.000000000","message":"Topic bug/1901383-patch1 removed","accounts_in_message":[],"_revision_number":1},{"id":"d9cbd563fc3cce64fa7fab92a5d1eb59f51cd2eb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-27 09:15:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3e5dd4b2b3a8a03f1c1920e4f2294923f1693f1","revisions":{"b3e5dd4b2b3a8a03f1c1920e4f2294923f1693f1":{"kind":"REWORK","_number":1,"created":"2020-10-26 13:25:14.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/91/759691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/759691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/759691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/759691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/759691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/759691/1"}}},"commit":{"parents":[{"commit":"516107fe5e90e56eddcf00516d3206ca0a68f357","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/516107fe5e90e56eddcf00516d3206ca0a68f357"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-10-24 06:52:43.000000000","tz":0},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-10-26 11:20:12.000000000","tz":60},"subject":"DNM revert eventlet upper-constraints bump to 0.29.0","message":"DNM revert eventlet upper-constraints bump to 0.29.0\n\nPartial revert of change I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\nsuspected of breaking designate\n\nChange-Id: I524fc271797cfc24a284b9f1f65ecc641771f914\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3e5dd4b2b3a8a03f1c1920e4f2294923f1693f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3e5dd4b2b3a8a03f1c1920e4f2294923f1693f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
