)]}'
{"id":"openstack%2Frequirements~759694","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I9e682ce062680f825b242390679ec3f732786bb6","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I9e682ce062680f825b242390679ec3f732786bb6","subject":"update constraint for tripleo-common to new release 13.0.0","status":"MERGED","created":"2020-10-26 13:43:03.000000000","updated":"2020-10-28 01:03:15.000000000","submitted":"2020-10-28 01:03:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759694-1603846995526-693ff990","meta_rev_id":"a71adac6c9a502368554288f434b1c5ba7f568ac","_number":759694,"virtual_id_number":759694,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-28 01:03:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-26 19:04:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-27 15:33:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-10-26 13:59:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-10-26 19:04:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-10-27 15:33:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-26 13:59:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-26 19:04:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-27 15:33:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-28 01:03:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d046594c6c4f7305278613acadd226dfdbdd14a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-26 13:43:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48acf234fb61d5b8452bf557191da2e557d4bccf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 13:55:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f4d7e65915f848639e0829c47137defb : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":1},{"id":"bdbf6f1a02581848fbece83e135115e7106b1b85","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-26 13:59:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8a704d55b9873ed5b8639b100f925b829d1a3af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 16:12:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d6f8ce42da14709abec8743920a3ff4 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/722dd4c746064d45ae35dab3977f9faa : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d75dd685dc7543dfbcfed3592deb2547 : SUCCESS in 27m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd2c369b383046fb90a4a18473d6ba70 : SUCCESS in 18m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fcf215d2fd8d4755a47911e66faf530f : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e6a3f0a1586648c4b33e0cf4ee080b04 : TIMED_OUT in 40m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/82374e436e2a4b6fa835ccd21a5790e6 : SUCCESS in 6m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/db9530b553714380a5f32e4431b43dbc : SUCCESS in 8m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/969fa5b29ca34e21ac47e2b2ca0fed3e : SUCCESS in 5m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d7ca57ed7d6a4fbdb0e4c0915edfd047 : SUCCESS in 24m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c7189addb444d299bd400ac1d054e61 : SUCCESS in 6m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a8cd7ea1e86f4a38ab76543f61bdc29a : SUCCESS in 7m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9d0c9e442851445b9c84d3d1eec1cd36 : SUCCESS in 14m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b0c746bec0264a26a8cc3984ed1e89d6 : TIMED_OUT in 41m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f6f6e42805f647d385f4dd89cc76dc2b : SUCCESS in 12m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/528b01a3998e4b5180f49688e486a6c5 : SUCCESS in 10m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cff52219ab784a6c821e3180349b5b14 : SUCCESS in 19m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f64fccc6d16746f5837b642912a1c255 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce2a5af45f5c4dde83b42c2bffa3a464 : SUCCESS in 5m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/de4683f5bf60466fad8a2251c80c9162 : SUCCESS in 10m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b2be268a2d14b0f95fff29d3b262b27 : SUCCESS in 55m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/daf603836c114d138945ab050a5aa989 : SUCCESS in 1h 14m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3bab7637514a63ca100fb5e6b1adacb21f5f925d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-26 19:04:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ee696436dbea592d3988b36246dc26dabf53e047","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 19:17:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6b169d9728d84fa7a7c40af5743629b1 : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e06331b4547286b34a4a677c8fdebdf96cc2a2b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 20:58:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce68894327b144f8a5efbe7fd8a6fb07 : TIMED_OUT in 31m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61447b7a39ab4c37933f066189ce2416 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f4f48b2a5594c27b7d53fa59d9dffb8 : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d35f920581f4a8bb02719ac4689e0e8 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6155721d40742909c2b770cc2cb73ce : SUCCESS in 3m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d896f12a2cef4fcba40d025bc0de8c3d : FAILURE in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bc9ade8024394544837a206a626b08c6 : TIMED_OUT in 41m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f1f6ac472e3c4fd6a88054e668586f31 : SUCCESS in 8m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d08ba56f2def4929966fccb52e9e90c5 : SUCCESS in 6m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dddfd51278244ca3b398767183f73391 : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d0044f2774d4ea292c643baaf77f5d3 : SUCCESS in 7m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e4826f73ddb7409d9baf5bf8c64d6f03 : SUCCESS in 6m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e0eb0af3ceac4117afea7e7e316d6b9f : SUCCESS in 13m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0c57c97f6ffb4915b95512f562ee65b6 : SUCCESS in 26m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a0c1c4c01b5543bfa7e2f7abbe63e89f : SUCCESS in 11m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7c9fa1a530f44bc8a4f6d103c16451c8 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d787c48a0d504815820efe08425f06e5 : SUCCESS in 19m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5c47764258d04c9da767125ca84599b1 : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32685bdc03e940c0bc8acce482ed5c64 : SUCCESS in 6m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/44033899132d42a6b935376e22087f39 : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c68ce32a3044937b1528a2079be49f5 : SUCCESS in 56m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9bbaea1e60044fa989294e61da08f1a3 : SUCCESS in 1h 10m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c784ed1ff325062e4882408cde850854be6aa0a2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-27 15:33:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"28533fc0adb20a592c85508ee3ae688f090799b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 15:49:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/af78672c30f34853859b4db233b21867 : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d0f1806e56b1b09c140482589104d40b50ed49f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 23:42:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f6a078116ea45a5b558ee504d22d33a : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d24ceb4ccc6446b6bfc705c7e2dc2eb5 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed1ab5b36e1a46cdbbf03454a3568b97 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dda97326a804492aad32edd920d32dd4 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ba76a23e01b41e6a0f5da65ae59d508 : SUCCESS in 3m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/15990f1e5793452bba9fbd3b3b1b3811 : SUCCESS in 6m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c2fb9cb0ec144d4f9684596f46e2064e : SUCCESS in 8m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/85134e46388846fa9c4c8a8c5cff11e9 : SUCCESS in 8m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/036b947b65e544329108f0940d36ab18 : SUCCESS in 4m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2419f71c3c3b48108c938c77425a66d3 : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7553d70b834a436e921acfed14563e0d : SUCCESS in 6m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5b11fd3e27584165bfdfce3459363495 : SUCCESS in 5m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4b8440551fea44189e5af0aaf39cb646 : SUCCESS in 15m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5a90bded8fdb4562be2c4cec4556bc51 : SUCCESS in 28m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/add17c185b814dfbb9cd5b4e935b4368 : SUCCESS in 11m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4ae7599f2a26488d95ca55f97f19b20e : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca2dc3568ac34255bf68c8ca04ce3a20 : SUCCESS in 21m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e349141a727242b19cf921288bb527ed : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08023099455c40288ead6dabe9a6c6bd : SUCCESS in 14m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/93c19c90a04a43648c3c4fb53face497 : SUCCESS in 15m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32b3ac0037ee4dffb98f1a3c23adc077 : SUCCESS in 54m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/474af6959106477aaae4ddf02198a859 : SUCCESS in 1h 12m 19s","accounts_in_message":[],"_revision_number":1},{"id":"c08ff4358ae634c01329ee7a3686f58fb61c11ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 23:42:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd649d8839084807fb947783c9db1f6cc02562bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 01:03:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9c253fff5d7d30107f4da4bae35b8fa6b22e3a27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 01:03:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57387dd2133849aebf51b46de33228c5 : SUCCESS in 17m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed56184b53d34a1f8cd2c9daf676c9b8 : SUCCESS in 18m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfd9716d01b648aaa366ae0b173ec11d : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c14d6e80b8d74c5d83252dc56a2ef1aa : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/190701d0740946fc895ece71fa6b79a4 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e5ac7b3640d74bd599add3dbed0a8449 : SUCCESS in 8m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cef6a7e15d2d4490b06354ff72d875da : SUCCESS in 6m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6fd7fad1f93846a998985c67064608f8 : SUCCESS in 11m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/948c8246f7c84c839cf17d21d845b7e9 : SUCCESS in 4m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/30ac3fdf665d49ca84c69c3d170dfb14 : SUCCESS in 9m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e96286f6d905496091b51551b5a07109 : SUCCESS in 6m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/70a7c5b2a9b545adb3ff83c8cda79c61 : SUCCESS in 5m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ef7068b5265f474382123c2a6cd612b8 : SUCCESS in 13m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bf7aaaa27d4f46d3875463c378aee23b : SUCCESS in 24m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3528e6a7899645fdbf9eb8060b6bca33 : SUCCESS in 11m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8f9368f6bc9f46d5b4cf5e0ddb182ad0 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/21861e2de77e48989bb504c9655240ff : SUCCESS in 16m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/61826dc704484a5295a9e4e70104d11b : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08b61b42ecfb462d80670e5d5324d48c : SUCCESS in 5m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/31aa2743c4be48b58e97ec2fb2c3153c : SUCCESS in 9m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/482c988afe604695a62965ab92d07966 : SUCCESS in 52m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/494cdeabae604637b892afe3f2cc71ff : SUCCESS in 1h 10m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c8ca542d7608b2cab9342ef244515eb0650069b","revisions":{"3c8ca542d7608b2cab9342ef244515eb0650069b":{"kind":"REWORK","_number":1,"created":"2020-10-26 13:43:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/759694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/759694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/759694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/759694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/759694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/759694/1"}}},"commit":{"parents":[{"commit":"c77e706f5f7a5bc1918b5dc31731a482bd57372c","subject":"Merge \"update constraint for oslo.messaging to new release 12.5.1\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c77e706f5f7a5bc1918b5dc31731a482bd57372c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-26 13:43:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-26 13:43:01.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 13.0.0","message":"update constraint for tripleo-common to new release 13.0.0\n\nChange-Id: I9e682ce062680f825b242390679ec3f732786bb6\nmeta:version: 13.0.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: MSA \u003cmarios@redhat.com\u003e\nmeta:release:Commit: MSA \u003cmarios@redhat.com\u003e\nmeta:release:Change-Id: Ice03f717faac78c8ca0a57f9839cdffd31e7d1a1\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c8ca542d7608b2cab9342ef244515eb0650069b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c8ca542d7608b2cab9342ef244515eb0650069b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
