)]}'
{"id":"openstack%2Frequirements~759705","triplet_id":"openstack%2Frequirements~master~I7335ae674aa5f230bfb740fc96a42b50e7e7fa18","project":"openstack/requirements","branch":"master","topic":"bug/tripleo-common","hashtags":[],"change_id":"I7335ae674aa5f230bfb740fc96a42b50e7e7fa18","subject":"update tripleo-common to 13.0.0","status":"MERGED","created":"2020-10-26 14:27:01.000000000","updated":"2020-10-28 05:46:34.000000000","submitted":"2020-10-28 05:45:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759705-1603863924604-332a2ac9","meta_rev_id":"d64983e5767dddc03d6cb6b5b465b37a4e69fab2","_number":759705,"virtual_id_number":759705,"owner":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-10-28 01:04:21.000000000","_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-28 05:45:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-26 19:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-27 15:33:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-10-27 15:33:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-26 19:01:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-27 15:33:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-28 05:45:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"545a31b731ac9f73e209f0c0f4889529245edbc1","author":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"date":"2020-10-26 14:27:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5e681e88157385184c19636a04cc89dbe4d9e62","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-26 14:29:21.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/759705/1","accounts_in_message":[],"_revision_number":1},{"id":"18c996e3c19ceff7d1b7e5992796652d78c2fbba","author":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"date":"2020-10-26 14:40:54.000000000","message":"Patch Set 1:\n\n\u003e Abandoned\n \u003e \n \u003e https://review.opendev.org/#/c/759705/1\n\nExcuse me？","accounts_in_message":[],"_revision_number":1},{"id":"1997ac14ec043ce89c6ded3555b17c76704c6a5a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-26 15:02:14.000000000","message":"Patch Set 1:\n\nAh, sorry, wrong link. I saw the other bot proposed update, but that one I now see is for stable/victoria. This would get picked up with the nightly requirements update, but no need to wait for that to happen.","accounts_in_message":[],"_revision_number":1},{"id":"4557ed95cefd70f1a4ec6bcc4186d7b515805f0c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-26 15:02:16.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"715a5118bbc48a3118179e2eec25d7dc15a788ee","author":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"date":"2020-10-26 15:05:30.000000000","message":"Patch Set 1:\n\n\u003e Ah, sorry, wrong link. I saw the other bot proposed update, but\n \u003e that one I now see is for stable/victoria. This would get picked up\n \u003e with the nightly requirements update, but no need to wait for that\n \u003e to happen.\n\nyeah！thank you","accounts_in_message":[],"_revision_number":1},{"id":"b2232ac2390c775854e31635676d203ef0d2a5f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 15:19:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2896200ffe42477ea9a3170b88c3afc1 : SUCCESS in 16m 23s","accounts_in_message":[],"_revision_number":1},{"id":"cd193ebd02a53cdb253957176064293fd15f7e0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 18:31:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c78d952faf12438683e3416e47fe34dd : SUCCESS in 21m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bbe75f3677e457aa2c0b35f5cf84785 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef429fc25306405d8922dd7145b69627 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f94ea4c761374922acdef60f4571eeef : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c12fb552a344c08870349f815ee7f76 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/147f9e1abf9f4db69370520980daf9bb : SUCCESS in 20m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/63eca779064a48c29073b0ac1f70f02b : TIMED_OUT in 40m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eb776b923b5846d7aee14ca4781df74a : FAILURE in 3m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e778b79f059a495ab2f0821fb0e9c414 : SUCCESS in 28m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/847816db42d44571941ba990de3ac01e : SUCCESS in 18m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ba644fa2400345368a09046c7276c693 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a4519e04358643dbb1b2e3101a9dbf24 : SUCCESS in 6m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5842636504e249cfa4eac37ed49fe604 : SUCCESS in 6m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8b5f9872fe4045e7b4af561f6bbd5496 : SUCCESS in 12m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8c38b2cd8b004909904ce5c2014aea3f : SUCCESS in 27m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/15535fbad8c4488383e79e0e5a26baf0 : SUCCESS in 12m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3eecf51c1ae84dc69407d30a3c3e0406 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d14c87448da7489db74df0128ade6dc7 : SUCCESS in 18m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/eb86ce2296a3490f9b0c84b8dd3ae2a4 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09946e4420c747eab47a2f5b8b508eb7 : SUCCESS in 6m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/953691ed3ece432fabc39aaa124fbf7e : SUCCESS in 9m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15ccebfc2fdf42548dcaabdc401eca35 : SUCCESS in 1h 01m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc1b92598fa4485cb85d0982a6b16126 : SUCCESS in 1h 13m 26s","accounts_in_message":[],"_revision_number":1},{"id":"30d0c9887df8e91c96326c883cb7b606157ba397","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-26 19:01:42.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3cf4a2c26ecc4d674770e843ebb9cf4fe483324e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 19:20:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/28e07d9634d84333811824523e57b6c5 : SUCCESS in 14m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a7584a85468e52bc48c24a09f79b77347b8b8b54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 20:37:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a0e073f4584a6b931b51c339da599c : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39373dea5e45422bba3da087cbb275e0 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab02724a22c841dabfcf76faab81fa6f : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dfa4d7c464e47bab38a96ebfc55c87e : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6987873ebdc04b879ae0b996d8fc2bf6 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5cacee26ec54962832f55ff451056ee : SUCCESS in 3m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0c0c186723854a1e9c21d2b93ed53d7f : FAILURE in 3m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/71f01ad3c3884852af7b7a8262adf3ff : TIMED_OUT in 40m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4c9932ae54af4f3e95663a2704ea9080 : SUCCESS in 8m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b88dffb5689b4843ab5004cdbe5c1ccd : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6cd8fe989db64de28269392dbaa8f01b : SUCCESS in 8m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3d3bb7690d46474ebc79f79af96d1e8e : SUCCESS in 7m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/585fb87e7f054585acabdddff4d3a890 : SUCCESS in 7m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/27e65424e12d4cc880015792b56bb14f : SUCCESS in 13m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/37463b1c50a64aec9332f58a29b189f7 : SUCCESS in 24m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6d5e32625152425e8e8b9e54117cb463 : SUCCESS in 13m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d1e6b7b7d0564eeca57da9c5d5228277 : SUCCESS in 4m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ec85e5e08cd043f6b91d22be9062f8ff : TIMED_OUT in 40m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0d01e5b073154dfd88d0aa803543d999 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1536651369544a6cae172b1cd6c3e14d : SUCCESS in 6m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c33f803c764d4ba59fe3f47cc01acef5 : SUCCESS in 9m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3995d0c9e014446e93b47dc33a255c7c : SUCCESS in 53m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1387249ab8b44b21b2b746bf85550b3e : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1699b8144210b6dfcc474b5bc277dc970e8cdd1b","author":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"date":"2020-10-27 00:53:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"52075c9ff9eaaee3a89392df801c3e13518b1ca0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 01:11:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3c3fe20616c94f40ad1a2624ddcbaa0a : SUCCESS in 15m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3c3ec8891b59961da759c1331c78279afb89e3e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 01:49:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da1315345da4c7a9c62b7139fa2995b : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b198ac1aba5c43669554d25a0c53002f : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18b66aa83b60425393a0587d2a213f4a : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16308005259642c6a5fdc21ce1357cfb : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa5ca5049de944e8b15e43bc4b5e7797 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8795fd1dea1947518d870de1d64228f2 : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c955ee5883e947209acd1b66de543486 : SUCCESS in 8m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/62cf7fe4aa594dedbae5a02b7340a4bb : SUCCESS in 8m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/87ffbce08b4b4222bea67e07ee6cd5ea : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4b17b38ef0754a5da392d7411c6e1bf8 : SUCCESS in 4m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5458473c13864acdb722b7dedf17c233 : SUCCESS in 7m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6841880b10294a36b98320d2b56bc980 : SUCCESS in 6m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/437db1f466e34e51a2f123716092f995 : SUCCESS in 9m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bcf6bc50b18c4d4cb28e28ab6f6b07bd : SUCCESS in 13m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/828400d326854064aa2d087b8b0736b8 : SUCCESS in 27m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/513f47cad0794c1293ce66288785bd80 : SUCCESS in 13m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/55a8bed210e14c2c87613c505ba3c3a8 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a46c14e644643b093a6464e4a8d1fcf : SUCCESS in 17m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e9facf996be848e692e7c8581d1ced36 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d465876c9a324f79b02c463d7da9b6b7 : SUCCESS in 6m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/25d66a55e2f4483686dc2c7f0cf13685 : SUCCESS in 9m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef61e12d3926442495f69a0fa55223de : SUCCESS in 47m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/824c3c8cadaf4e398b68b006fd644c91 : FAILURE in 53m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e8f201109b0b67e9c3b4f950e18112615920f43e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-27 15:33:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9641962582f0fce43a08c3c21c20f8301ae96c7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 15:54:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c583ad7036ac4235b131de9ce15f4bf6 : SUCCESS in 14m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8e4bc5db99175c3587ae8fceac7c74efe073d398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 00:32:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37e64be56e4d4f8db8791ff8bf0caecf : SUCCESS in 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d53167bb486f47d29211ecd4cae3227e : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c868f89cebd84d1ea45ce1db8ec66c82 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0157f6b9e4c4a89afa54d215d64e5e7 : SUCCESS in 21m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ba8cd732c4b4689b023a19ec9e1c12b : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/488949e1dcf84cb0a7c7ebeab4a246fa : SUCCESS in 20m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6842f0632c9f4cc6ac01fcd5c56a03c2 : SUCCESS in 6m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b80a484b5dc64423b7da2bc854412ad8 : TIMED_OUT in 40m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/78ac6317b605421f96e4a02c440ce6b8 : SUCCESS in 8m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4987d9d7c0ad4f2d9f17f3ca831d22c5 : SUCCESS in 5m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6ff5d8ffb5e143bb86f5ab438ca0d3f7 : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/785b7c27c68040e2b1bf9b2ffec0bc23 : SUCCESS in 6m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5395edfa6bbc4196aa519982c2b1fd15 : SUCCESS in 6m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5513fba54e5a40d9951172015db47c05 : SUCCESS in 12m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a2b19ca23f604455a646e8c1e3d9c3af : SUCCESS in 31m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/36c3365378654eb0af1fc6d15745a46e : SUCCESS in 13m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/169864e5dcf649889e4e139189c5fd7a : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9914c99fc2ae42d9a698aa9744a91a2d : SUCCESS in 17m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0c0f2af5029648e390cadf35b69cb584 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/613cfeeb6ddf4ce6aef94e2dbef2a285 : SUCCESS in 26m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bcc13a3f6bd44e5b9548ae215b0f2560 : SUCCESS in 10m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c60212c82cb41788e3cd45de9e92c11 : SUCCESS in 1h 41m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f31e36e65a654586a889f0097f1950af : SUCCESS in 1h 59m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3b9653a20d6ffb3dce20d933666f0256ace85d7b","author":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"date":"2020-10-28 01:04:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"359480e1ad079d25007cef089147628a9b2fb727","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 01:19:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7aa3255e32ae4dffba854cdabbde8376 : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9ac49ee903d44d46f7fbb544a049d517c411f8e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 04:21:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98e0555fb1b54ebea0c7e696bdaa5a28 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e3985002817493ca7fd883e971dc390 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6ce0d0297b44d7bb4bc70e97d9ccd76 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf2cefa1e321473fa8c49d62cea520c4 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10322b9b83a840feb4066bed2580ecae : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93a56f117b904b4fab1ad141bc4cfb21 : SUCCESS in 3m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/12753c3f58f84ead9fc1d54138f26277 : SUCCESS in 9m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1a545c465db84b919e67acdee5ba167a : SUCCESS in 6m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c89f40108afb4d41ad742ba02a688201 : SUCCESS in 9m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/60661c5cb0c24f549969f81eaa9928e9 : SUCCESS in 4m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/56e89f9fd2f94574a753805f333fef9c : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ccc1ae1f2a7a4cb593c6b1733473b703 : SUCCESS in 6m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f2e9854853d34b8bba5a449a6985fa78 : SUCCESS in 6m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7c646796dab4402283b5b7e77400d320 : SUCCESS in 17m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/88cf328cde514e05b3fc7f5112daa2b7 : SUCCESS in 23m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a7794829d7cc440590962e4b5758166e : SUCCESS in 14m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5d3a4c6f5adf4be880b3be6d28cc9305 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0582d1cb4f0f443c93df711469082395 : SUCCESS in 20m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/73b8fe2b2fb44a63b3dcac43330b3063 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96a3b4fdba2f4ecd8e109dca54a4f96d : SUCCESS in 7m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b186ebac1c944683b64fe49ee19ad63c : SUCCESS in 9m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8eb788d46bcb46439e0440fc0c89317b : SUCCESS in 54m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d213159e86049caabb19b845d0e3b80 : SUCCESS in 1h 21m 52s","accounts_in_message":[],"_revision_number":1},{"id":"209886ba435a62bf04bfae93ceb2cb2f33070992","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 04:21:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c950795142c67500f6a08f5c58545459fbe4db78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 05:45:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"33c8d35750f4428a63ccdfc63ce0e07211a12ae9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 05:45:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e9e15b7ea984148b899359df4f8b263 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25d94abd56f94219afeb2f1897d0009a : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3342155b4230490e99ff6d18f6ff525d : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85158192e0da4dc2b1373ae3b446076f : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7594ea90d590417bbe291bd22fbfaa5a : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79947731a39a41b3bd02cccce989d010 : SUCCESS in 3m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/37d72457a7374f9297ea2da8d8815ad8 : SUCCESS in 7m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0494a5029ea6469fbe45f30f52d3d12a : SUCCESS in 7m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/084ee869d8e64547acad81600868160a : SUCCESS in 8m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4aa1ce994c1f4de3918672af5498b2c9 : SUCCESS in 4m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/88a4305ee63249209db56256bc9212c9 : SUCCESS in 8m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87338bffdc464bfb8b926d093cfc5936 : SUCCESS in 6m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/88ccba570fd64c0b8fe6b97792615652 : SUCCESS in 7m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e74b2368643e4c9487ec7f3dd749245c : SUCCESS in 14m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f678b340d6f248329c70783a43915ead : SUCCESS in 23m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/84c75a6f32a64f3ab2aa4dc3b729a1a4 : SUCCESS in 16m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/12de534bfc6b4e50b3f1123133114f4d : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/15bae154129e44e6a5ad73e5b6981bc8 : SUCCESS in 23m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0878918211fe444bbd1a0096104b842b : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e13a5d8c5b2a459cbd235917a85761bf : SUCCESS in 6m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a904b3a2d7184e12b2123b7b24b8ca03 : SUCCESS in 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c96183e4d8fa406a82980b3c7a0d0a15 : SUCCESS in 58m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eca6f9fabf5c48b5b6dd6e6ea6ffb516 : SUCCESS in 1h 19m 46s","accounts_in_message":[],"_revision_number":1},{"id":"47396767a067c91dd0b72e5e062bcba6cc9b35a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 05:46:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce7d19f5f4bb44d79ff5ae248eb8e58d : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50288bd8a3c1a9e8ea77e7a39d2eab25174d6e77","revisions":{"50288bd8a3c1a9e8ea77e7a39d2eab25174d6e77":{"kind":"REWORK","_number":1,"created":"2020-10-26 14:27:01.000000000","uploader":{"_account_id":30533,"name":"zhangyc","email":"zhangyc257@chinaunicom.cn","username":"zhangyc"},"ref":"refs/changes/05/759705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/759705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/759705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/759705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/759705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/759705/1"}}},"commit":{"parents":[{"commit":"747ff02e2c3b9dc6bcc461731cec8758da7704a4","subject":"Merge \"update python-ironicclient to new release 4.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/747ff02e2c3b9dc6bcc461731cec8758da7704a4"}]}],"author":{"name":"zhangyc","email":"zhyanchao1988@126.com","date":"2020-10-26 14:26:14.000000000","tz":480},"committer":{"name":"zhangyc","email":"zhyanchao1988@126.com","date":"2020-10-26 14:26:14.000000000","tz":480},"subject":"update tripleo-common to 13.0.0","message":"update tripleo-common to 13.0.0\n\nChange-Id: I7335ae674aa5f230bfb740fc96a42b50e7e7fa18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50288bd8a3c1a9e8ea77e7a39d2eab25174d6e77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50288bd8a3c1a9e8ea77e7a39d2eab25174d6e77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
