)]}'
{"id":"openstack%2Frequirements~760057","triplet_id":"openstack%2Frequirements~master~I95369d2abe393b05003cc8013837c943e372856d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I95369d2abe393b05003cc8013837c943e372856d","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-10-28 06:30:32.000000000","updated":"2020-11-01 11:34:42.000000000","submitted":"2020-11-01 11:33:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760057-1604230403746-8dccae41","meta_rev_id":"00ff54773870a129a64332b66f4a5a6afeacc4b5","_number":760057,"virtual_id_number":760057,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-01 11:33:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-11-01 03:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-11-01 03:38:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 18:08:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-01 03:38:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-11-01 11:33:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3a58b5e724e5ab09f43232f0fe58cbe34430fc4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-28 06:30:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd55e24b025b054266ddacdcc226cf88cf0dc538","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 06:45:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/001ac56e07444257bafad6c5f57cdbdb : FAILURE in 9m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4bd04de2d7317fa7ee644580fcd7e0a380872639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 09:40:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51b5155483046529b7ebf597e72246a : SUCCESS in 20m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0507ae4814d04bb59efc2e6289832f89 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39e2ad9705cf4410ab9deb40c9946dd3 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf58f3502ed04591bafdc2c73d3b13c3 : SUCCESS in 22m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2816855683754de9bd07fed13000d85e : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8d0e2d1589d4adca7ee8589d3f8cb50 : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d0e03fea9e484d709174ebb32cf76992 : TIMED_OUT in 40m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a53e7ae73f55439f87dbb122ac23b38a : FAILURE in 4m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5b27d2031c2340c3931fd12a7d1e543d : SUCCESS in 10m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/af03a72227024334a03039f49bac39d5 : SUCCESS in 17m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/774ed94a492145bc98951f8d9cb2f34e : SUCCESS in 11m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/071150d93767438f8e9a85ff83fdd7f1 : SUCCESS in 5m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8beda1bcdf654c1fb4f98c845d7b065a : SUCCESS in 23m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ec12099b876c4522ac406b359558ab20 : SUCCESS in 16m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/069aaba84cc44616ab7626b5a6c59ddf : SUCCESS in 29m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f1fae02c44b442069d2760141f7bea84 : FAILURE in 35m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a8529614bcd64de0a0aaefc5f7994d79 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f3d7ca6b3484b88896649d7e4ab6034 : SUCCESS in 19m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0922efa26c74478586d9566845639875 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b7b8212d94cd457cabb8884703c0451c : SUCCESS in 6m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f9e1e092968c499cb9c42c7b744db57e : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a9ae5889a39477c854e835e0bb876bf : SUCCESS in 1h 06m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d472693b5374b3d8a2bdf06b2acf16b : SUCCESS in 1h 17m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0ca06c297809bb546424c382885e7990bf52395b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-29 06:47:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bb01ac44ef4038f61539610042f9faa0a08db548","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 06:59:14.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c1a69ee4a5d44a50b36a62cae10b5a6e : FAILURE in 7m 03s","accounts_in_message":[],"_revision_number":2},{"id":"45c51d30a7601488978514c291a5ab83eedfa8a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 15:01:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ec0740a9dd747fa9044ef5c88c67cab : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/285c4d79488f4777ac20197f52da095e : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/120e92121d2a4aee8e4387b4be5e5004 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ff9fc7dcc634f0f9ead7788aea5d1f3 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2414aa4d20545ea95f70f10e71e594b : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e87ed04f66e644c6ad7f1981c5cfe450 : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5f0b3ef2ee6941399c53bcc64b7c3896 : FAILURE in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e606965bdeeb41bfa9f9a50b6f869315 : FAILURE in 7m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b1dbaee4e21e4150b45a3321741b5349 : SUCCESS in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cd0b67de66d84118b7b186a26383756c : SUCCESS in 7m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/834286523609466399b29b22f9580b30 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4af941a2707947909f5e39785898c36a : SUCCESS in 7m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0468ea7d214446bb853749a4719b751f : SUCCESS in 7m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/99efdbfb72984ec98da77e0ab03999bd : SUCCESS in 16m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2f4f062a4d6a4816ab88fa0d60cd4132 : SUCCESS in 25m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1daae2d0848f458dbbe85d5ae56a6a7b : FAILURE in 13m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f777405e694a4027b41bb3e89d4d2a3c : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f3110e90b67d4cbbbae31578146659a0 : SUCCESS in 17m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8f2bd91a41604e9cbd09b575d58e6d88 : SUCCESS in 7m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc7f08cd4d0a41a3a5b1bc3e7bdba14a : SUCCESS in 7m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6d2dc0dc096d4d379b84bc53a8a19f9a : SUCCESS in 10m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc69804b4767462e8f3bc6325fd2cd32 : SUCCESS in 53m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/993852a8f337454a9a7724f72451c317 : SUCCESS in 1h 20m 20s","accounts_in_message":[],"_revision_number":2},{"id":"dd98adaf63fd5d6da3432865a9314149978eef83","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-30 06:20:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"660e16e665ac6d2951366a17495706ca0bf51744","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 06:39:41.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ea0e18d747224de58a7e22329b61424d : FAILURE in 12m 16s","accounts_in_message":[],"_revision_number":3},{"id":"1fc2b7fe517318a1c436566b00c26b592ca48415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 08:27:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e01160d6ed5544a09768b707dc369f13 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe9e515a139468a958d3c4843c0f3b4 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cba894cf85e646f3a74df108970dd4e0 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2886b3a7caa84fcea7a2976b7117fdc9 : SUCCESS in 6m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4875fa2ed6774b959ab6775d2603557f : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/26e4647d5db64c458c94ced9f9d1abba : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dc2fd456b23e464787cfabbce98b45b3 : FAILURE in 4m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/901b14ae96cd46bebd2ab24a264d942d : FAILURE in 11m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/104c30d1b5544e03ba8aba611869ef41 : SUCCESS in 9m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c06a4feb09524c9bbf6879ca6ad2e578 : SUCCESS in 7m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/617b3594314440f5b2a2dcc8cdae6161 : SUCCESS in 11m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f117f24ef9734f22b774a17090271046 : SUCCESS in 11m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2223a3936ce64bc5961c1a8de69d8200 : SUCCESS in 7m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b6411bb7e1034cc19fc59a872df2a7b4 : SUCCESS in 13m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e65fdf0fa86b4cbd847dccc79d03b167 : SUCCESS in 27m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b106373cd5f248f5a57ad39a91517b52 : FAILURE in 12m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5e270252df884354a6e3c6b493377b96 : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac0f7ec8397c44138291fc161db4ba9b : SUCCESS in 12m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3e15b2b75acd47a88de43918862cc277 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/749fa3d35a734c16bd21f81f9e7b8633 : SUCCESS in 6m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8f986db42e9244efa780600c9a93a5ca : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb2368dba4b14dc49284d43b42a19ff8 : SUCCESS in 1h 01m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98a4571db86b4c2fa5e4e5ce765c7e06 : SUCCESS in 1h 16m 38s","accounts_in_message":[],"_revision_number":3},{"id":"33fefec12dcad884a8fc1f78884dac0145d818b7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-30 16:09:23.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"86bf8b277af94b2c2d4f804aeabc3c126127e10e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-30 16:11:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"80fa3ac23faa9582307f68326a8a454f8eae2243","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-30 16:11:43.000000000","message":"Patch Set 3:\n\nkeeping mock\u0027s update in, want to see what breaks","accounts_in_message":[],"_revision_number":3},{"id":"ae4a2a450ae9394c96bfedf8a0af4995256b10be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 16:32:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d228357402644c489f57fe51e0f9cc9a : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":4},{"id":"fb2814f108f2332dc80bfe614e0dd2f547c6a03c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-31 01:28:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51c47a77a77d42209561b3a909c3a360 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c0125cc3ac24db29491e6640a1184af : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96064f4c7d8d4b0b9d591b16aef62303 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82ccf842b34c41769af10d1f0e63402e : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/002e6670afc541749b07f7b34d48eba2 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc8d22fd008d476eacc1b1dc22688d91 : SUCCESS in 3m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d62790565b844f2f915e28b970f1e118 : SUCCESS in 29m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6f628d0423af44a29e02ee029cd58774 : SUCCESS in 11m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d2b71a19bafb4ad0a7bd3cf3af4c24e5 : SUCCESS in 13m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8776800c549f4b58a96f69454d01f500 : SUCCESS in 5m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f0ddf229a66b4068bf075d980b96f730 : SUCCESS in 7m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/15b8de55e0704a6f8ad9cf25831b1260 : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b6d55e8861324a52b30e1239813b13a2 : SUCCESS in 6m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/18f5c41e1e6a4c6ca7fb2a3322388cc4 : SUCCESS in 13m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/690dbaa6d6484ee08b765a982f504ca1 : SUCCESS in 26m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b613b70ea8764029bca337103ea5e4fd : FAILURE in 13m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dbc2ced222754294880dbf6002491801 : SUCCESS in 5m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7d9d6613cd4642408752a6a2f846371b : SUCCESS in 16m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5d4c3e20ed0142038cbc9dc295154ec0 : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/40bcab0da5fa4d9c9d34f788466a20e2 : SUCCESS in 6m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/668db88141d64a2fbbe6ceb87ec1f086 : SUCCESS in 10m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a68b5830c3874a13807ff7129d94da91 : SUCCESS in 1h 03m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d1bb827ba7b4b6085fb0eaa4d4fadef : SUCCESS in 1h 26m 07s","accounts_in_message":[],"_revision_number":4},{"id":"916d1521baaf4f4b2f5cafa779dbb07f4ddeefa7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-01 03:04:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"deaa618de3a636b6d487082f360d662a5af4661a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 03:24:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7c3a040c3e5b4204b0fc4b0cd3956086 : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":5},{"id":"60daf9882307b25e9e6e759a600c1fa3eb14a594","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-01 03:38:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"301a71a64ce8d1e12bfe2f5683d51204de141bc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 04:25:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/048047e6066c4cc892e887398cd74b76 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5cc60db9a99496fab26dc075ca69527 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87b55e364394434cbedb76525b6213bc : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15512d3c0b5343658dcc0cbbc012d7c1 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e4353d894a142fdbb6b4d97a453d07e : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6daf2792982f468d8cc6bb044cf56f31 : SUCCESS in 3m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1fc919916ad8461fb78721c7705f668a : SUCCESS in 7m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9121817dc4ee4951a38a31dc8750227e : SUCCESS in 7m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d91410166b4349248572db68e6cc0585 : FAILURE in 8m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1f5c2640868348288fc648ab525ba0e0 : SUCCESS in 4m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c63d790482404fe09436a99a93e47e9b : SUCCESS in 7m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/69dea5242c81438eaece60f76940e419 : SUCCESS in 6m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5508668001af4ad6ac1837a849e4cc1e : SUCCESS in 6m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c6d3c167b53c4cb4a8bea36f613562aa : SUCCESS in 13m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0f7daa3d9fec472491a89e131a1574f1 : SUCCESS in 29m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e2ddb5f56f7d41cf9421fb4863e3b452 : SUCCESS in 12m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eadfc365d0654ec19d154bc38a10bec2 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4eeb79579810468a92bfbb4544ee346c : SUCCESS in 15m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d9e9c3ab02694d87b3098a11f543575f : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2063d1a2b9844e21892ed3d062b5df7d : SUCCESS in 5m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/295202975d3e48d499781e9ad51ff9e4 : SUCCESS in 9m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90908781c62648eb99ab1d4eda9bef66 : SUCCESS in 54m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d22299a46be74d9f9bf586a8f027dfce : SUCCESS in 1h 18m 04s","accounts_in_message":[],"_revision_number":5},{"id":"2db1bfcd0526e8dc60f490249d60143e1c95b721","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-01 07:45:04.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"961902d5a4956340a01727cb2708c62901080dc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 08:02:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a728c2bd49024bb98acec78ce5d38afb : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":5},{"id":"985420e427202d87c4193a8e58feb6870639d6f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 09:14:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a920b59c82e457c8aeb2b4407ae5dec : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/315a97f666eb4a6bb4620a01d31280d9 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ffacd595a84652aaf451de51219483 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7f903b4c8654f918dbf0aa3a4e83065 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/abe9f92ae150406daa815426d6decebf : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/188d76f66d9d44c0835d0e793a22d572 : SUCCESS in 2m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5fc0b08a5a3243f2b9b15cb770d868a8 : SUCCESS in 7m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4f4b48d95ce94ab3b1db7211372619ae : SUCCESS in 7m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ad051e9ba5ad4e19b8317d7aeceff7f6 : SUCCESS in 9m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c80cddf62b5e43f88e478da559b1c04e : SUCCESS in 11m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d212ab30ce81479a965523b410a9791e : SUCCESS in 9m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c81f56f256c04aa099b99ab82353ba0e : SUCCESS in 16m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c6a90fef98cf4a2f91541d7d9113bb0f : SUCCESS in 7m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0b38425c68f141f4bd36a89ca86835f9 : SUCCESS in 13m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0d072793f36e48e5b7f2c2b76a9a48d3 : SUCCESS in 27m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4cea1c3ea0c14981b71346b0683d5b66 : SUCCESS in 11m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0330da109e4040c7803c14038ffe7322 : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5f45c25bb4a3488c87eb984ab99cf77a : SUCCESS in 16m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/986ad8d5c72a43afbc0ff3500de6c3df : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2dccdfa4c2f141c38dd77730e8207d4b : SUCCESS in 5m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/74d0e481948f40589b68d789dc70ce9a : SUCCESS in 9m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1014ff3b076b43ff86677c66d67ad3e9 : SUCCESS in 48m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b42924d1cea46799373fcbbdd890ad7 : SUCCESS in 1h 23m 24s","accounts_in_message":[],"_revision_number":5},{"id":"12d17b9470995418d5a89dea89732844e43429fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 09:15:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"401d954e7d6cb5dcddb1488a2e014853475ef178","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 11:33:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"a235151e1269d365f91dfa80ea312d2c80517192","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 11:33:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d9dec2b98b3472db74436543839ed1e : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fefbeda060a24eb685c0246488b0f7ea : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07f53dd141d24187afcdfbe730d7303a : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adad0046c4e84b2cb61b9789e36cde05 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/785b257714604e46b908da3264cb2d91 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d3ced65d49e4218ad7d1525596fa806 : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/46625c75e9fc495f9d20b539debd08cb : SUCCESS in 8m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2950eabe072640948daa9a56a6ad0aed : SUCCESS in 8m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/369e5fc86e5b4ea5ae71ecc149a53602 : SUCCESS in 8m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/68486546befe4a4288c342ce12387ebe : SUCCESS in 4m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/35a9808d67ac40d19c1e68f23b85d422 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f78ca43baa6746b6b35ea527156bf75e : SUCCESS in 8m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/974c89c0450141968b13e6e8f63b9490 : SUCCESS in 6m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ea750a68355046739bd57059f043eac9 : SUCCESS in 12m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c2d4e25800414bd9aeb898b998e4acef : SUCCESS in 16m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/32bd64ebefa843478368682d598ed298 : SUCCESS in 13m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d804faac1a724625ae4603dc58b01673 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/61c09ef7ef314d4595977660ceba1632 : SUCCESS in 15m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d23d82eec2eb441eb51e80af4770ce29 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f1c2a89a4c142c4abb816049dfa7524 : SUCCESS in 6m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d8b29c57aba14b08bb61a5733e6309ed : SUCCESS in 10m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8784b18b80004103b07001a5e488176b : SUCCESS in 47m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3fa0d2edbe046f0aaa5f5fd82994a9e : SUCCESS in 1h 20m 27s","accounts_in_message":[],"_revision_number":5},{"id":"00ff54773870a129a64332b66f4a5a6afeacc4b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 11:34:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d92ac7adf3744695bba88ff0f23a645c : SUCCESS in 58s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"113441bf8f688ab7037f954bd50a12d6fbf4acab","revisions":{"4770fa76d227aa05f5a1cd7a82d3d4d6ffe33a8a":{"kind":"REWORK","_number":1,"created":"2020-10-28 06:30:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/760057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/760057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/760057/1"}}},"commit":{"parents":[{"commit":"2cf34dcd1dafe0a5976b406e782944eb1bdcd036","subject":"Merge \"update tripleo-common to 13.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2cf34dcd1dafe0a5976b406e782944eb1bdcd036"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-28 06:30:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-28 06:30:30.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I95369d2abe393b05003cc8013837c943e372856d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4770fa76d227aa05f5a1cd7a82d3d4d6ffe33a8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4770fa76d227aa05f5a1cd7a82d3d4d6ffe33a8a"}]},"branch":"refs/heads/master"},"004f1da3b12828db60ab5828c20d6fb49b5722b1":{"kind":"REWORK","_number":2,"created":"2020-10-29 06:47:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/760057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/760057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/760057/2"}}},"commit":{"parents":[{"commit":"f7d8626cf847dbac6a9bbf3e565d41a74db11e45","subject":"Revert updating version of pymemcache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7d8626cf847dbac6a9bbf3e565d41a74db11e45"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-29 06:47:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-29 06:47:27.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I95369d2abe393b05003cc8013837c943e372856d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/004f1da3b12828db60ab5828c20d6fb49b5722b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/004f1da3b12828db60ab5828c20d6fb49b5722b1"}]},"branch":"refs/heads/master"},"92cb3906f31dca2255af3ab61625ed2ee60705bd":{"kind":"REWORK","_number":3,"created":"2020-10-30 06:20:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/760057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/760057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/760057/3"}}},"commit":{"parents":[{"commit":"f7d8626cf847dbac6a9bbf3e565d41a74db11e45","subject":"Revert updating version of pymemcache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7d8626cf847dbac6a9bbf3e565d41a74db11e45"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-30 06:20:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-30 06:20:56.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I95369d2abe393b05003cc8013837c943e372856d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92cb3906f31dca2255af3ab61625ed2ee60705bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92cb3906f31dca2255af3ab61625ed2ee60705bd"}]},"branch":"refs/heads/master"},"3ff5ffe93a6438a4778bf04a2e62c89afef351e0":{"kind":"REWORK","_number":4,"created":"2020-10-30 16:11:26.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/57/760057/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/760057/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/760057/4"}}},"commit":{"parents":[{"commit":"f7d8626cf847dbac6a9bbf3e565d41a74db11e45","subject":"Revert updating version of pymemcache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7d8626cf847dbac6a9bbf3e565d41a74db11e45"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-30 06:20:56.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-10-30 16:11:09.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I95369d2abe393b05003cc8013837c943e372856d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ff5ffe93a6438a4778bf04a2e62c89afef351e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ff5ffe93a6438a4778bf04a2e62c89afef351e0"}]},"branch":"refs/heads/master"},"113441bf8f688ab7037f954bd50a12d6fbf4acab":{"kind":"REWORK","_number":5,"created":"2020-11-01 03:04:51.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/57/760057/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/760057/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/760057/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/760057/5"}}},"commit":{"parents":[{"commit":"f7d8626cf847dbac6a9bbf3e565d41a74db11e45","subject":"Revert updating version of pymemcache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7d8626cf847dbac6a9bbf3e565d41a74db11e45"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-30 06:20:56.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-11-01 03:04:43.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I95369d2abe393b05003cc8013837c943e372856d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/113441bf8f688ab7037f954bd50a12d6fbf4acab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/113441bf8f688ab7037f954bd50a12d6fbf4acab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
