)]}'
{"id":"openstack%2Frequirements~760103","triplet_id":"openstack%2Frequirements~master~I155024516774949d9438d4f0b918f023eebcc6fc","project":"openstack/requirements","branch":"master","topic":"revert-pymemcache-update","hashtags":[],"change_id":"I155024516774949d9438d4f0b918f023eebcc6fc","subject":"Revert updating version of pymemcache","status":"MERGED","created":"2020-10-28 10:33:56.000000000","updated":"2020-10-29 05:51:15.000000000","submitted":"2020-10-29 05:49:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760103-1603950579642-05be731c","meta_rev_id":"f46e9dcba1d56825b682fd6c4a1de64d2234f0b7","_number":760103,"virtual_id_number":760103,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-29 05:49:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-28 13:48:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-28 13:56:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-10-28 13:56:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 13:48:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-28 13:56:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-29 05:49:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a2c315746bb04e24ae6dc9503409eac543aaf17","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-10-28 10:33:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c065b39bb325ceb233c63334bce43a8e779b685","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-28 10:44:42.000000000","message":"Patch Set 1:\n\nCan\u0027t we simply black list this version directly in tooz as it was already did for 1.3.0?\n\nhttps://opendev.org/openstack/tooz/src/branch/master/setup.cfg#L67\n\nI don\u0027t think we want to decrease the version like that.\n\nAlso we could ignore this version by using our global-requirements:\n\nhttps://opendev.org/openstack/requirements/src/branch/master/global-requirements.txt#L206","accounts_in_message":[],"_revision_number":1},{"id":"63b7c7cdc55142a3f36f6c93548a7abdf40d5e5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 10:51:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/21008ab247a54af09f756d00e2f8084e : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"00a02d440853c3ec50b8dbd41ed3bb98eb028b9d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-28 12:08:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a92055198bc14550d238c426393bef69f7d7a84c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-28 12:10:57.000000000","message":"Patch Set 1:\n\nWe should get it blacklisted too. The next release will raise the upper constraints again. Users should nearly always get that new version, but making sure we block the bad version will ensure no one ends up with it through some dependency constraint combination.","accounts_in_message":[],"_revision_number":1},{"id":"bce69b68fe94e3c61cfb9fd836a511e4d3b4665a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-10-28 13:08:41.000000000","message":"Patch Set 1:\n\nHervé, Sean, if you prefer I can convert this change to a blacklist in global-requirements, but I\u0027m not sure this will block tests that use uppper-constraints, so we probably want both ?\nAlso I\u0027m totally ok in blacklisting that in tooz, but I spoke with people from oslo and they are going to have a look, so probably there will be a fix soon-ish, which I guess will be valid for the next version of pymemcache as well.\nWe need this mainly to raise awareness and unlock our CI as currently we\u0027re excluding pymemcache 3.4.0 in a non-ortodox way :)","accounts_in_message":[],"_revision_number":1},{"id":"d9a82b439a21f0b322b604f1236d07630486902f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-28 13:22:09.000000000","message":"Patch Set 1:\n\n\u003e Hervé, Sean, if you prefer I can convert this change to a blacklist\n \u003e in global-requirements, but I\u0027m not sure this will block tests that\n \u003e use uppper-constraints, so we probably want both ?\n \u003e Also I\u0027m totally ok in blacklisting that in tooz, but I spoke with\n \u003e people from oslo and they are going to have a look, so probably\n \u003e there will be a fix soon-ish, which I guess will be valid for the\n \u003e next version of pymemcache as well.\n \u003e We need this mainly to raise awareness and unlock our CI as\n \u003e currently we\u0027re excluding pymemcache 3.4.0 in a non-ortodox way :)\n\nBlacklisting in global-requirements should prevent any projects from using it. So if you can update this patch, I think that would be best.","accounts_in_message":[],"_revision_number":1},{"id":"bed81eac279b1a06231fb4d647a8a51f61123d37","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-10-28 13:36:54.000000000","message":"Patch Set 1:\n\n\u003e \u003e Hervé, Sean, if you prefer I can convert this change to a\n \u003e blacklist\n \u003e \u003e in global-requirements, but I\u0027m not sure this will block tests\n \u003e that\n \u003e \u003e use uppper-constraints, so we probably want both ?\n \u003e \u003e Also I\u0027m totally ok in blacklisting that in tooz, but I spoke\n \u003e with\n \u003e \u003e people from oslo and they are going to have a look, so probably\n \u003e \u003e there will be a fix soon-ish, which I guess will be valid for the\n \u003e \u003e next version of pymemcache as well.\n \u003e \u003e We need this mainly to raise awareness and unlock our CI as\n \u003e \u003e currently we\u0027re excluding pymemcache 3.4.0 in a non-ortodox way\n \u003e :)\n \u003e \n \u003e Blacklisting in global-requirements should prevent any projects\n \u003e from using it. So if you can update this patch, I think that would\n \u003e be best.\n\nwill do, thanks! :)","accounts_in_message":[],"_revision_number":1},{"id":"ef97949cf4ee6ad63f8a28968bc4db8ffdd56cca","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-10-28 13:39:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cfc8e100d5b07b45fa62b14475e71525d2173948","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-28 13:48:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8d507ee81082fbc3e2c69785a434d65952c4d9e0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-28 13:56:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"314de8d38582f26c6192f59d8dc85d1b879d38b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 13:57:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e882a588cb01466a87170ad4756c23a9 : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":2},{"id":"6d39bd7c99bbbd91e0e77e3416b387307f2e4b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 04:29:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e316fc732d84af3be28b7c3e8cccc83 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0894dfd997874d5f899cf95e09352420 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01895e001aab4ab184e5db1c718aa7be : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/556d41b6ea904c29b21059284bf1d451 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff4e069380ed4f13bc8e0e9e0d4665da : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f855e28b68d7462fb04b93fa036e97cb : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fd5f19696f814bcda357b64293099063 : SUCCESS in 7m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ac6950eda9a64211a9ff017fb3e5ff33 : SUCCESS in 6m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/411d5a6a35914be788e0d99a09fe319a : SUCCESS in 9m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0ddcb38e901446bdb759a39e7ec14464 : SUCCESS in 6m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dd009809628d435288d04ce567a02d32 : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87267d5726cc46e5a73085306c73786a : SUCCESS in 5m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c5008314d2394907b7ca8f7ddb6b305d : SUCCESS in 8m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9f3fe372a80546a294284367de044b0b : SUCCESS in 11m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3750a964d93b4e82b7cabdb7825671cf : SUCCESS in 24m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cd9c92dfbd204d3e8258f16f49c9d9fb : SUCCESS in 11m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/116a0f3fe3ae4572ab7e9b1a38126ac6 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a49451b463a42d2b57b19a63189fc07 : SUCCESS in 20m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/11cdd7fa41064895b2622a1c40a6ac38 : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e1ec214d53fe4e369c97dc8b48b265ea : SUCCESS in 6m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ae993fc06e664572a59f958bdcbed837 : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa0440d9fe414c04822b625a855326a6 : SUCCESS in 1h 01m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/971cd688d65f45c3b61a69feeb0aaf61 : SUCCESS in 1h 25m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b1b1f66551ac00963d9bccc5f52c5af0b58cc81c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 04:29:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8fd6c5159e191008cc2f15cea3039b22f0330fdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 05:49:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f85d5c3416499a9d4c0d9d295e5173fa913122ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 05:49:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c402ece0b2b44638beb6fca51ee03d0 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db60139c8f254fc5881d8bdff126e49e : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e23e16cf44d4f698e15a8296a64a07f : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2af8db2e2129437fa67145854217c26a : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e6f5945dd434f2f9291b9dc4e2ae178 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/07a67c9732da444cba3bc8cc2fc3d2cf : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9ee0d2dde75646d085ec5f215bc66b7d : SUCCESS in 7m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a37b6c3f32a042839cb1f76800924cdf : SUCCESS in 8m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/67408a301091473dab9124d9c3df7c00 : SUCCESS in 13m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c278c60c7f23426a92772e260941db85 : SUCCESS in 4m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aed5f8dcc75b4081be569f1a26845ffb : SUCCESS in 7m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/743523f40a1c459fb05659905a2bc447 : SUCCESS in 6m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/08fcdc95f3ee46d7bee065a9fc98c41f : SUCCESS in 6m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6de44da0baff44e89507f1a0c6610c21 : SUCCESS in 12m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f8d54c28236147e7beb0662f0751bfda : SUCCESS in 26m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/51dbdbd0a38942609867baaa393365d0 : SUCCESS in 11m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0e8b2ae7f66f4907940bc2cbff73f58f : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7c131816f593481bad130f2774aa273f : SUCCESS in 18m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/eb20e5b5eb864a4cbe2cc09178ed92d4 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90ca393c17494b888622896dfc413328 : SUCCESS in 5m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/595b429bf85041478b279cf6fb361682 : SUCCESS in 16m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d7818df0b06474d9de3aeae14934722 : SUCCESS in 45m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc697573704246ecbaca0edce995ed20 : SUCCESS in 1h 15m 53s","accounts_in_message":[],"_revision_number":2},{"id":"04b446d6bfd5934300ccf8a0aadfb119af775339","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 05:51:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30e8954b261c4eb79c29f67cac227912 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f7d8626cf847dbac6a9bbf3e565d41a74db11e45","revisions":{"ded4877d17b8c862e65cf1d502f463ab96bf11be":{"kind":"REWORK","_number":1,"created":"2020-10-28 10:33:56.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/03/760103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/760103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/760103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/760103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/760103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/760103/1"}}},"commit":{"parents":[{"commit":"2cf34dcd1dafe0a5976b406e782944eb1bdcd036","subject":"Merge \"update tripleo-common to 13.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2cf34dcd1dafe0a5976b406e782944eb1bdcd036"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-10-28 10:31:20.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-10-28 10:31:20.000000000","tz":60},"subject":"Revert updating version of pymemcache","message":"Revert updating version of pymemcache\n\nThe latest version of pymemcache (3.4.0) breaks the tooz memcache\ndriver effectively breaking all the projects that rely on that.\nAn example is provided by the ironic-inspector-tempest-managed-non-standalone\njob in the ironic-inspector project where we had to force the\npymemcache version to be installed to 3.3.0\n\nChange-Id: I155024516774949d9438d4f0b918f023eebcc6fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ded4877d17b8c862e65cf1d502f463ab96bf11be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ded4877d17b8c862e65cf1d502f463ab96bf11be"}]},"branch":"refs/heads/master"},"f7d8626cf847dbac6a9bbf3e565d41a74db11e45":{"kind":"REWORK","_number":2,"created":"2020-10-28 13:39:03.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/03/760103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/760103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/760103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/760103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/760103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/760103/2"}}},"commit":{"parents":[{"commit":"2cf34dcd1dafe0a5976b406e782944eb1bdcd036","subject":"Merge \"update tripleo-common to 13.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2cf34dcd1dafe0a5976b406e782944eb1bdcd036"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-10-28 10:31:20.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-10-28 13:37:59.000000000","tz":60},"subject":"Revert updating version of pymemcache","message":"Revert updating version of pymemcache\n\nThe latest version of pymemcache (3.4.0) breaks the tooz memcache\ndriver effectively breaking all the projects that rely on that.\nAn example is provided by the ironic-inspector-tempest-managed-non-standalone\njob in the ironic-inspector project where we had to force the\npymemcache version to be installed to 3.3.0\nWith this patch we set the version of pymemcache in upper-constraints\nback to 3.3.0 and we blacklist version 3.4.0 in global-requirements\nto prevent the use globally.\n\nChange-Id: I155024516774949d9438d4f0b918f023eebcc6fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7d8626cf847dbac6a9bbf3e565d41a74db11e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7d8626cf847dbac6a9bbf3e565d41a74db11e45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
