)]}'
{"id":"openstack%2Frequirements~760109","triplet_id":"openstack%2Frequirements~master~I5d102e14bcf42470770c54297b9af320045413a9","project":"openstack/requirements","branch":"master","topic":"image-encryption","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2021-11-23 00:51:05.000000000","reason":"Change was submitted"},"28271":{"account":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"last_update":"2021-11-23 00:51:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5d102e14bcf42470770c54297b9af320045413a9","subject":"Add python-gnupg as a requirement","status":"MERGED","created":"2020-10-28 11:26:59.000000000","updated":"2021-11-23 00:52:27.000000000","submitted":"2021-11-23 00:51:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760109-image-encryption","meta_rev_id":"87b678f4788b8b12d0add81884d49271723facf3","_number":760109,"virtual_id_number":760109,"owner":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-23 00:51:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-22 22:05:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-22 22:05:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 11:45:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-28 21:11:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-29 02:42:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-11-22 16:37:39.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"}],"messages":[{"id":"7902d50851aceb6fa307cc3c84b1f90831c326ad","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2020-10-28 11:26:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62ec07f6a3ac5ba17dcc5c4bc581538e54038853","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-28 11:44:10.000000000","message":"Patch Set 1:\n\nHello,\n\nFirst, thanks for these changes.\n\nPlease can you answer these questions [1] by updating your commit message.\n\nAlso I think you need to add few other changes, please refer how to our doc and especially how to add a new requirement [2].\n\n[1] https://docs.openstack.org/project-team-guide/dependency-management.html#for-new-requirements\n[2] https://docs.openstack.org/project-team-guide/dependency-management.html#adding-a-new-dependency","accounts_in_message":[],"_revision_number":1},{"id":"cd8aa08dabf0f1a54b6105344eae8f5c3b394be5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-28 11:45:04.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"868a1602a1fea596a3bae3b252ffc86b5fd9ca7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 21:11:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e4eefabc704eba827aa7a0b87f02d7 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20e0177f84e3474886c33450c5455285 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f808c08b50643959db377380260dae2 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cc52814a7a64786bec8238fec8f15bb : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c3f8d95cf3f4f118061a8eac96f125b : FAILURE in 3m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8f6930a8f3f48d4a610138f85a4e52a : SUCCESS in 59m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36646fc8d4254a6185e1e9a5c58d28b8 : SUCCESS in 1h 18m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4e1f5161cbd8343c3dc8d58966feecc92b90ecd6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-29 02:42:36.000000000","message":"Patch Set 1:\n\nany reason this over gpgme or pyGPG?","accounts_in_message":[],"_revision_number":1},{"id":"bf35989fe722a4b0879303d329842cd67e67defc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-07 01:22:43.000000000","message":"Abandoned\n\nno response in 4.5 months","accounts_in_message":[],"_revision_number":1},{"id":"0f03e644948cf1e2be9ecfebea204e92c97e47ae","tag":"autogenerated:gerrit:restore","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-11-22 16:34:04.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"fb8ed941786449ff4e9e31804bb5666865ac8c67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-22 16:36:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e91b706b2ce0f2f4f7e68749a22299515337e662","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-22 16:37:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bff7a870d5ca25c75280d6176c02c262a4c92961","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 16:53:55.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cae063ddb25b4e529c27e77f25305039 : FAILURE in 14m 53s","accounts_in_message":[],"_revision_number":2},{"id":"403cb271cd4022cf5fe147f851642c1c8eb825c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 18:04:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e9d433ae24343d594da559d5ddb8aef : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a97fd5b9dca1413e91398f72ab38a2a4 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50f03d583c754f6082934b6ee27f12b4 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d02061de57d74db79eafb76452a61160 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41d838a919fc4d0bb3a64ddd924f8df3 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0dd0fd1a710e4ca484680f487528478f : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/46e3a997074540c8811e23fb3c41df31 : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/018bb875055642caadccfda8f18c5f81 : SUCCESS in 10m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/613ce4adb4214ac3813476bca49080a2 : SUCCESS in 9m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/68618603dbe843f5bf66690415e279bc : SUCCESS in 6m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4418d86bbb3941d0a8644b9c500f893b : SUCCESS in 10m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0c7643b3f1414e2bb380f40fc8110aec : SUCCESS in 5m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cd0c1315349c4a4fb6574cf1baad8bc9 : SUCCESS in 12m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7473317489b4301a8458b19b776d176 : SUCCESS in 8m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d523ddc51ebd4b23bb84e4065f70b4ec : SUCCESS in 8m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/71f5e24d60f444e08cfd45c178c91b3e : SUCCESS in 19m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4b0cd3e07f044c2aa03c5674736ba7f6 : SUCCESS in 3m 54s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6a22a9928c7a4a17a27725705aafc08d : SUCCESS in 11m 58s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/049d4719f98743ce81913fd0cfbe0ff6 : SUCCESS in 5m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f18a2ea5f6b1460b82d3ae730689edcc : SUCCESS in 37m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3cf479814280492ab16a2b670bda6e52 : SUCCESS in 12m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/576b4e995e964799bf5c59297151ed28 : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fd38bccebc604cb3a4748fe52b7948ea : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0137cbf6f0e14ab5b2369c43094e6142 : SUCCESS in 17m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/eec7c5ab557e461a91ff5b7bfc70bc80 : SUCCESS in 6m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e67e6ff256de46c19fa1e6d23e0a5e97 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9fad2ea12fd74eefa5852ca430af69a1 : SUCCESS in 7m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/931f8a8ee87145fb9b9b094681bffa0e : SUCCESS in 11m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b70e7395ddcd43fbab6f4501dc58ac15 : SUCCESS in 51m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31258ca8125444b8afce214ffad99d0e : SUCCESS in 1h 21m 11s","accounts_in_message":[],"_revision_number":2},{"id":"31cee187c8239ea70d092fb41fbe7e18eae1b60a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-22 21:53:54.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"244b5951a2efbf2d579cb6819293617d766d24bd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-22 22:05:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b8ca83935dcb04fac7a1d49a333beb09d4344974","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 22:12:11.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/060cfd4babb04409b39fa4147a995897 : FAILURE in 13m 31s","accounts_in_message":[],"_revision_number":3},{"id":"a5118d23d652ac65f92a508241146b151e06b2a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 23:37:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d8c98e2751249f1a35885b354d411f0 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1317262f53b4cbab46ddf513da8637f : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68a7a15d612c43238cefb459827abc70 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c66a47ab806448e9f974b30641f9e71 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c385ff5265d64c2eac258581273b6535 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d58fa5953524264a4158936525db768 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6de1279429174e109591c87a14c1e364 : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/011f2d161dfc45b2ac2d8f19f03feb5c : SUCCESS in 8m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5c0534c07a0c43eda7b62fcc2153a807 : SUCCESS in 9m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/566a3962614f449689e394427a3fc332 : SUCCESS in 5m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fdf8a76efcbb4262b805ece120edbf9e : SUCCESS in 9m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b0dbff1f83fb4d598d8e7be9ba95b7e5 : SUCCESS in 5m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7142eb220d5b4b45b2a832932349a51a : SUCCESS in 9m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8275b010090f4cffbf1a203adbffa2c7 : SUCCESS in 9m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d6c521ccb47b4812874f60e99aed9f9a : SUCCESS in 7m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/73ddda04259a41c68a4c9f73fe143de9 : SUCCESS in 15m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/707d335b8d2f4fd0b05eb41111bcc89a : SUCCESS in 3m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1f13b9d1d4584d2490113e8e70e6965b : SUCCESS in 10m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c423fe3404c4471086b65dd0d66e228e : SUCCESS in 5m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/65b48b68997f445d982e6e410e699d5e : SUCCESS in 33m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b1c67df9e7374050be99e1c75b45caf5 : SUCCESS in 12m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a0328db3b6e54142a9bbe629d4a8a724 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3078a67131b94439aeeefd1ce4b019b5 : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/21300da325b24288a1fc61b5b731421b : SUCCESS in 20m 21s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/11c34289563e4725a41a6b16837e3572 : SUCCESS in 7m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/108cf840efe9451caa52dfc48c4e3754 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4b67c38cb845448a90991a477ddb0d7f : SUCCESS in 6m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/084a02ef6b5247d2a9fa342e7ac802ff : SUCCESS in 12m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e77e88f141854b1389bfdbb4d0e085d7 : SUCCESS in 46m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fab8383b20a4a6fa4698d0bf69fd6a9 : SUCCESS in 1h 36m 27s","accounts_in_message":[],"_revision_number":3},{"id":"90bb85776802d0c4852dcf87dd55d1f7b098c751","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 23:38:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"13c585c2b13e4dcdf8b67acc8da5cd12bcfde7e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 00:51:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9cdf3b0ca7e4ee384b62434fc96988f : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52ab147bba2b48dfb9ea3b8b4e297053 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67f5ffe9f15a43cab8d6e96ab5d3ca33 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7096775008d646b093007fa8b2c8264f : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/39ebd9b66635421e88fcb5117bd3ef65 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cbe0978fa83b4ed5b52007530ee3c710 : SUCCESS in 3m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/756aa61eeb8247d8bcac94de6b847e8e : SUCCESS in 11m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/50b8e8d18666420f9fa66bcdb9715dbc : SUCCESS in 8m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ddf7fb253ef24a37bc504cf230b17ed9 : SUCCESS in 9m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/63700bc46c6e406499509e4bc71a5e51 : SUCCESS in 6m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cb39b909319c4c828961b0601464ceff : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e2e04ebfc7964258a533b0df1a11d548 : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/20171cbcc80f43d187f433362aec64b8 : SUCCESS in 8m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0c7fd19d9a5b45369431b45348af7d37 : SUCCESS in 13m 45s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0ea73256f1014f159fef940fc1229065 : SUCCESS in 5m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9565bdad77244d37bb3ed76afcfae1f6 : SUCCESS in 9m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f8e0613f095e4f8d8caaa06bca346975 : SUCCESS in 5m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e86d724c7af7421490f682535e9dd003 : SUCCESS in 32m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/137114ded2f44a518d10a95c3253105d : SUCCESS in 10m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/31465105760e491b95802789ac30f2f7 : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/164a36a91afd4d7395fe3d5c0eaf6e5f : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2831bf3fa7c7450f9721fbc69cecc35f : SUCCESS in 18m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/dfa1ad3a1fb6491a9eef9f7c41b5a6b9 : SUCCESS in 7m 10s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f3e1d430e67d4b47bea65556dac9c220 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2cff605654dd43638ebe053439ecb83f : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3f47ff1b87424874b822117df4a0d804 : SUCCESS in 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31b0f3cde18a42b5ad87b10e4865a349 : SUCCESS in 40m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c1fc24639ff4812a8cf318b74af0f93 : SUCCESS in 1h 10m 39s","accounts_in_message":[],"_revision_number":3},{"id":"9fd1c529bb5114826ccd4174bf2e6297e9f899ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 00:51:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"87b678f4788b8b12d0add81884d49271723facf3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 00:52:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1969a47bd23489aaeed056e01986932 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"88ec401440dc48cd675ad5ddd2d04147eaff2335","revisions":{"ab41526c584a1d95a0470a3dde354a7074fa8e57":{"kind":"REWORK","_number":1,"created":"2020-10-28 11:26:59.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/09/760109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/760109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/760109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/760109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/760109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/760109/1"}}},"commit":{"parents":[{"commit":"2cf34dcd1dafe0a5976b406e782944eb1bdcd036","subject":"Merge \"update tripleo-common to 13.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2cf34dcd1dafe0a5976b406e782944eb1bdcd036"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@secustack.com","date":"2020-10-28 11:23:49.000000000","tz":60},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@secustack.com","date":"2020-10-28 11:23:49.000000000","tz":60},"subject":"Adds python-gnupg as a requirement","message":"Adds python-gnupg as a requirement\n\nThe image encryption needs python-gnupg to encrypt and decrypt images.\nEncryption and decyrption will be implemented in os-brick.\n\nChange-Id: I5d102e14bcf42470770c54297b9af320045413a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab41526c584a1d95a0470a3dde354a7074fa8e57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab41526c584a1d95a0470a3dde354a7074fa8e57"}]},"branch":"refs/heads/master"},"6be37fa9c16e6e5c1dda5a5be2fffea19826f7bb":{"kind":"REWORK","_number":2,"created":"2021-11-22 16:36:09.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/09/760109/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/760109/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/760109/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/760109/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/760109/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/760109/2"}}},"commit":{"parents":[{"commit":"a73224328ad16363f139ed21134ce1e5b23fde5a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a73224328ad16363f139ed21134ce1e5b23fde5a"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@secustack.com","date":"2020-10-28 11:23:49.000000000","tz":60},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-11-22 16:27:49.000000000","tz":-300},"subject":"Add python-gnupg as a requirement","message":"Add python-gnupg as a requirement\n\nThe image encryption effort [0] needs python-gnupg to encrypt and\ndecrypt images.  Encryption and decyrption will be implemented in\nos-brick.\n\n[Q] Is the library actively maintained?\n[A] Yes, latest version was released 2021-03-11\n[Q] Is the library good code?\n[A] Yes, The docs are thorough [1] and the code looks clean [2].\n[Q] Is the library python 3 compatible?\n[A] Yes, 3.6 and greater [3]\n[Q] Is the library license compatible?\n[A] Yes, BSD-3-Clause [4]\n[Q] Is the library already packaged in the distros we target (Ubuntu\n    latest / Fedora latest)?\n[A] Yes, Ubuntu [5], Fedora [6], Debian [7], CentOS [8], and PyPI [9]\n[Q] Is the function of this library already covered by other libraries\n    in global-requirements.txt?\n[A] No\n[Q] Is the library required for OpenStack project or related dev or\n    infrastructure setup?\n[A] Yes, an image encryption effort that crosses Glance, Cinder, and\n    Nova [0]\n[Q] If the library release is managed by the Openstack release process\n    does it use the cycle-with-intermediary release type?\n[A] N/A\n[Q] Do I need to update anything else?\n[A] No\n\nCo-authored-by: Josephine Seifert \u003cjosephine.seifert@secustack.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\n[0] https://etherpad.opendev.org/p/image-encryption-weekly-meeting\n[1] https://docs.red-dove.com/python-gnupg/\n[2] https://bitbucket.org/vinay.sajip/python-gnupg/src/master/gnupg.py\n[3] https://bitbucket.org/vinay.sajip/python-gnupg/src/6979164ad4a90037077ec5d4c1d62f0ef7c36a7e/setup.py#lines-41:44\n[4] https://bitbucket.org/vinay.sajip/python-gnupg/src/master/LICENSE.txt\n[5] https://packages.ubuntu.com/search?suite\u003dall\u0026arch\u003dany\u0026searchon\u003dnames\u0026keywords\u003dpython3-gnupg\n[6] https://packages.fedoraproject.org/pkgs/python-gnupg/python3-gnupg/\n[7] https://packages.debian.org/search?keywords\u003dpython3-gnupg\u0026searchon\u003dnames\u0026suite\u003dstable\u0026section\u003dall\n[8] https://centos.pkgs.org/8/epel-x86_64/python3-gnupg-0.4.6-1.el8.noarch.rpm.html\n[9] https://pypi.org/project/python-gnupg/\n\nChange-Id: I5d102e14bcf42470770c54297b9af320045413a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6be37fa9c16e6e5c1dda5a5be2fffea19826f7bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6be37fa9c16e6e5c1dda5a5be2fffea19826f7bb"}]},"branch":"refs/heads/master"},"88ec401440dc48cd675ad5ddd2d04147eaff2335":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-11-22 21:53:54.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/09/760109/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/760109/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/760109/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/760109/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/760109/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/760109/3"}}},"commit":{"parents":[{"commit":"a73224328ad16363f139ed21134ce1e5b23fde5a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a73224328ad16363f139ed21134ce1e5b23fde5a"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@secustack.com","date":"2020-10-28 11:23:49.000000000","tz":60},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-11-22 21:51:44.000000000","tz":-300},"subject":"Add python-gnupg as a requirement","message":"Add python-gnupg as a requirement\n\nThe image encryption effort [0] needs python-gnupg to encrypt and\ndecrypt images.  Encryption and decryption will be implemented in\nos-brick [10].\n\n[Q] Is the library actively maintained?\n[A] Yes, latest version was released 2021-03-11\n[Q] Is the library good code?\n[A] Yes, The docs are thorough [1] and the code looks clean [2].\n[Q] Is the library python 3 compatible?\n[A] Yes, 3.6 and greater [3]\n[Q] Is the library license compatible?\n[A] Yes, BSD-3-Clause [4]\n[Q] Is the library already packaged in the distros we target (Ubuntu\n    latest / Fedora latest)?\n[A] Yes, Ubuntu [5], Fedora [6], Debian [7], CentOS [8], and PyPI [9]\n[Q] Is the function of this library already covered by other libraries\n    in global-requirements.txt?\n[A] No\n[Q] Is the library required for OpenStack project or related dev or\n    infrastructure setup?\n[A] Yes, an image encryption effort that crosses Glance, Cinder, and\n    Nova [0]\n[Q] If the library release is managed by the Openstack release process\n    does it use the cycle-with-intermediary release type?\n[A] N/A\n[Q] Do I need to update anything else?\n[A] No\n\nCo-authored-by: Josephine Seifert \u003cjosephine.seifert@secustack.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\n[0] https://etherpad.opendev.org/p/image-encryption-weekly-meeting\n[1] https://docs.red-dove.com/python-gnupg/\n[2] https://bitbucket.org/vinay.sajip/python-gnupg/src/master/gnupg.py\n[3] https://bitbucket.org/vinay.sajip/python-gnupg/src/6979164ad4a90037077ec5d4c1d62f0ef7c36a7e/setup.py#lines-41:44\n[4] https://bitbucket.org/vinay.sajip/python-gnupg/src/master/LICENSE.txt\n[5] https://packages.ubuntu.com/search?suite\u003dall\u0026arch\u003dany\u0026searchon\u003dnames\u0026keywords\u003dpython3-gnupg\n[6] https://packages.fedoraproject.org/pkgs/python-gnupg/python3-gnupg/\n[7] https://packages.debian.org/search?keywords\u003dpython3-gnupg\u0026searchon\u003dnames\u0026suite\u003dstable\u0026section\u003dall\n[8] https://centos.pkgs.org/8/epel-x86_64/python3-gnupg-0.4.6-1.el8.noarch.rpm.html\n[9] https://pypi.org/project/python-gnupg/\n[10] https://review.opendev.org/c/openstack/os-brick/+/709432\n\nChange-Id: I5d102e14bcf42470770c54297b9af320045413a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88ec401440dc48cd675ad5ddd2d04147eaff2335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88ec401440dc48cd675ad5ddd2d04147eaff2335"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
