)]}'
{"id":"openstack%2Frequirements~760160","triplet_id":"openstack%2Frequirements~master~I2ad858ed085f99b1ce546fa0d9a9ded30e006f01","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2ad858ed085f99b1ce546fa0d9a9ded30e006f01","subject":"update constraint for oslo.utils to new release 4.7.0","status":"MERGED","created":"2020-10-28 15:01:18.000000000","updated":"2020-11-01 21:27:31.000000000","submitted":"2020-11-01 21:26:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760160-1604265983988-b61dc1be","meta_rev_id":"ddf35bacf5eef6903a6dd0bab6a4d02f29ddbcb5","_number":760160,"virtual_id_number":760160,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-01 21:26:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-11-01 18:52:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-10-28 19:49:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-11-01 18:52:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 19:49:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-01 18:52:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-11-01 21:26:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56a5a45a9c895e8d9e2fe56bdb7093be5e0a3439","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-28 15:01:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ee4bbc317be1edd56962f6eca767e6382937812","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 15:21:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5fca0b0c46a143fd820bf7738b1448f2 : SUCCESS in 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c4ce2a97ff35e6e9e32b60530eb416fdcd0a8b2b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-28 19:49:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"58e14b234518c58062bd0895c75ac00ff9d5da47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 06:20:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70721666f35e4d358a43d037989e650d : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d065e5ddf88144c0aafb1ed155929349 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8371861ed2ef44d3ba481c9fbc347cb8 : POST_FAILURE in 11m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a43b72f039f4a56a7c14d11b971b307 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f82cd7984114097bb6aa48c98eb1909 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d6e1cb672ce4f08b08b5d541c0e25a2 : SUCCESS in 3m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/76c7f5fc43264781ab6e6cc18b230de0 : SUCCESS in 6m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cd9d7e3cc0d14a89a9a98ab93f3661f8 : SUCCESS in 6m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f3edd65f325449ab9fbdede4da8eca5e : SUCCESS in 17m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fbac4e32461d4e919f95ffb1265ae7ca : SUCCESS in 4m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ab0bdc719e8f422480081e5fc097642c : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8dc965045b5a4a8c8300ab2094445c93 : SUCCESS in 7m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2fc336adcbc84cff996a4bb3f2eea386 : SUCCESS in 8m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9c8ce2aa6b124620bc0c58d2703343e4 : SUCCESS in 15m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c7e3b0010a024334a4d7088f2ee83593 : SUCCESS in 16m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5e027baeba43461ea20ab32150c10672 : SUCCESS in 14m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c7effa42107547599d2762ae8bb337f0 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/878203eca6c3455d84fd4e07d4e2eca7 : SUCCESS in 18m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dbff54870e994e7ba7b8a4ab0512646f : SUCCESS in 19m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d3f0a744a3648a9bbe3a9db25ff4d5d : SUCCESS in 6m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/931b1604d08944c89da293b0af8b1293 : SUCCESS in 10m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bacc690a20e245b2917d6ed364265fed : SUCCESS in 41m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa81901705cf4b90bc96d0e7f614c3de : SUCCESS in 1h 25m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1b3fae66f03624c32c4249ea191309b8aff809b4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-01 18:52:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c31974000ecaec27815392fe9e54e4db83f8b03d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 19:08:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e79cf861f7d44d5789591b5940c526ce : SUCCESS in 13m 13s","accounts_in_message":[],"_revision_number":1},{"id":"11b535e3dc9e02fa9b5181178f10cb4f2e8341a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 20:14:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76d7625345644f21bbda109169d62fee : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f50143e500894403b3aee64d44979857 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/688732af206243bfaa30104ccd1349a8 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b55a741c3aa34f729f536b623677130f : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46c3ff92852f45f3b1bdc4f551c5ed3c : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93be8ec1b663490b92758c32ea7a5cb9 : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/52215013ba144ff3849999e40bc63697 : SUCCESS in 7m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3527ff14f23046dbb976f927d2b05176 : SUCCESS in 7m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0175b8ff1ea54a0da8c12cd38574f58d : SUCCESS in 8m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/36b6b40cd39743068bf0f574196a64e0 : SUCCESS in 4m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/29e4caeb2394421ab223a7dfb0da6ced : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1ce8bd948254d4494052906893170c8 : SUCCESS in 7m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4dbc835e50b641edb6b5fb0c979e9e5c : SUCCESS in 6m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6f916b5236ec4e4ea4c8be5f1ecf6e22 : SUCCESS in 12m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/35c5a469c47b4d5d8a215e27c06e0b96 : SUCCESS in 22m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/46ddd3204be74e138563c6cb5dac605c : SUCCESS in 10m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/41c87e09e44d4eaca506d3193721c727 : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/750dcd8be03d44b39dd8e120dfc205fa : SUCCESS in 15m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e30923d2db454f6e9a5cb515d27f65cd : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/95391aa21c8e4a33a56a13552bf35755 : SUCCESS in 5m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/abca5b6e486c47b7a434a9a2f24dd1e8 : SUCCESS in 9m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86e27d32458b41de88e602524638cbf6 : SUCCESS in 47m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69de7c8f69dd423ea30beec56129b375 : SUCCESS in 1h 21m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3eea0e1fade904066ac717f1cd9e9ec08c67594b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 20:15:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45d76fc9e26ced6ab57c37f5e7bc92e7655e58f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 21:26:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d54735d2c082427f8537790481c82a6e : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fbeecf41f0341dbbcf7a1cb4206651f : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15ec046164ad4b9794dbd8a530388be8 : SUCCESS in 9m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df137bd2291745618f2a5f4608a1844d : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f29145c13574b5f984181458903a1c0 : SUCCESS in 7m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf37a2809ba4416b936b5e66445c06de : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2af1903c29f045cfafc070de9af8a852 : SUCCESS in 9m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/640ff8ddd5d9432aa0b3bc4ae4466c7c : SUCCESS in 9m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e76d962bebdb4c239b39f0b6341b3506 : SUCCESS in 16m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2dc72d11e54e4b58a9897cb14ca7ca6d : SUCCESS in 7m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dba9374576d24e18836a5bb151d41c99 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4f7c7ee441804084a8ccde25c039afbd : SUCCESS in 7m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3770b12c48804d548a1a8d01e5815634 : SUCCESS in 9m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b0ec16d141cd4f4a93e6545daa4f9c13 : SUCCESS in 12m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/688416810e704fc2be2efb18ae1f23c9 : SUCCESS in 24m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4223bece96b14044851190ace40916ef : SUCCESS in 17m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0160083ba50e4bca8b27820b02eaf7ce : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8bfe00f2fba44c6b2776e09a99665ff : SUCCESS in 14m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/596581e9d2bd4f648eec202ef8f7ba31 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67b6dff7a0e943608aaaebc09a4eeff4 : SUCCESS in 5m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7f9e41aef4604f6da99529f54aadede3 : SUCCESS in 9m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2229b1b9200e4b209b42d3a886743641 : SUCCESS in 57m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25d28b8f07494d97aed0e3e4a6874f31 : SUCCESS in 1h 09m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5e532ff4667d4428e06f5e67c82b5ff993cb3104","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 21:26:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f51706f098233d8ef884f69f5bf5dc0c16d713ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 21:27:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba5646167be54f0d8da75b090489ae02 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"031c4976db3234e30d45430513628e7019c325bd","revisions":{"031c4976db3234e30d45430513628e7019c325bd":{"kind":"REWORK","_number":1,"created":"2020-10-28 15:01:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/760160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/760160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/760160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/760160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/760160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/760160/1"}}},"commit":{"parents":[{"commit":"2cf34dcd1dafe0a5976b406e782944eb1bdcd036","subject":"Merge \"update tripleo-common to 13.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2cf34dcd1dafe0a5976b406e782944eb1bdcd036"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-28 15:00:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-28 15:00:01.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 4.7.0","message":"update constraint for oslo.utils to new release 4.7.0\n\nChange-Id: I2ad858ed085f99b1ce546fa0d9a9ded30e006f01\nmeta:version: 4.7.0\nmeta:diff-start: -\nmeta:series: wallaby\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Change-Id: I6642b8dbbb2b5c057ba70c8b53661b083a22f95e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/031c4976db3234e30d45430513628e7019c325bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/031c4976db3234e30d45430513628e7019c325bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
