)]}'
{"id":"openstack%2Frequirements~760944","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I993ec3548d6b10ac2cd4061f97a814a37bdcfe0d","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I993ec3548d6b10ac2cd4061f97a814a37bdcfe0d","subject":"update constraint for ovsdbapp to new release 1.2.2","status":"MERGED","created":"2020-11-02 15:21:37.000000000","updated":"2020-11-03 09:17:16.000000000","submitted":"2020-11-03 09:17:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760944-1604395036657-aa5a3b26","meta_rev_id":"c93629cf51eb38d92bdb1b5949d3c03b41f88963","_number":760944,"virtual_id_number":760944,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-03 09:17:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-02 16:20:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-11-03 01:13:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-11-02 15:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-11-02 16:20:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-02 15:35:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-02 16:20:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-03 01:13:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-11-03 09:17:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"452003613737596e2b5109c191285cd27efa35f0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-02 15:21:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6525bd6bfc7b82ee7de8563c2c8a738363558805","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-02 15:35:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb07bd16f9a09bcd5709e7481cd782ad2b9502b9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-02 16:20:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c440ee7ed651bac8afe1fa54388422deef2144f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 00:14:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa06851dc7924cc1981fa10c43f65610 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cbb077dfc364e76953d58e5ef224f65 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a420f37fb79d4913b7bd839f23c16f04 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd022d41d1f84878ad54b22ba0bd9ecc : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/151308b5a0324e18a65cf447980ea6a7 : SUCCESS in 3m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5c44e25b94474c55b56af6593f63b85c : SUCCESS in 22m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2f4a1ffe40e642d7a956bb680204985c : SUCCESS in 13m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0d351be1d2454c6c98bd3c6476ed6e0c : SUCCESS in 4m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/73ff0cba00554cb6a73fff6042879e84 : SUCCESS in 14m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6c29763cc4b4da0ba4a582dbe0e4acf : SUCCESS in 15m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a4142701fe17450cadda82d567d5e97d : SUCCESS in 11m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/db67b103d122470c86175f5dc4bebb09 : SUCCESS in 32m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/12abdcecb78a4c3389883cfc3a091520 : SUCCESS in 12m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f3a49ab19cec4bab9b8c1fd27a3b7b24 : SUCCESS in 12m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/48a41536fb0d4a3f95e2b5c5ba0bc748 : SUCCESS in 20m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6bba0841559745a59ce6a0e71760a300 : SUCCESS in 5m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a7a6385da5dc44c2a28deba6fafd238c : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7df6de3957c14b0295c3a2374c2110cc : SUCCESS in 11m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/78028c32ae9f4814914929e70d612e7f : SUCCESS in 9m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/71c758683cb649ac9e7dd70fafdd763a : SUCCESS in 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f583647165e430bbff3f8b01e26c2f5 : SUCCESS in 1h 00m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d576684d6a824e4097aaf9ec4ba6f545 : SUCCESS in 1h 30m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ef82dc1d7004a77b6aafdf4dffeb4ba : POST_FAILURE in 1h 26m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/cb51d1ed6b6e4f03a995d0c1ba338831 : SUCCESS in 1h 15m 24s","accounts_in_message":[],"_revision_number":1},{"id":"57446a4537965eb45ba1e2315835f9b8ad656797","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-03 01:13:05.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"00eea0742d81a1beef71d874ca203102cc809a81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 06:12:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5837f41ec047b197099c58e712c66f : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b278cb479f94e33b5b8173257c668e7 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1093fc76109c4e35ab4eb87f7ede4c9d : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6da6aa638984a109438a829f41b1935 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/121e4c7f1d3f4193ad2d454066121d79 : SUCCESS in 3m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c74915d5591249b19aac2fc6574be44d : SUCCESS in 6m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/87224a4056694aa7b44c6017418fdb99 : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/57e594936ac4499cadeb2ece9048ca71 : SUCCESS in 8m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4ec570522576454aad980ecff74c678e : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b2e1c0c6f05456aa9bfaa1c931531d7 : SUCCESS in 7m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/213668002d384406be4e9d79057344ee : SUCCESS in 13m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/55b4d21a548a4770af5306d1f483027c : SUCCESS in 28m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/665ee59a51e9489d99f37ea57ffc2e80 : SUCCESS in 12m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3edab2ef32024dcf8a690d16322764a6 : SUCCESS in 13m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ddd257624adb4b45a622850438707599 : SUCCESS in 15m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/74dae4fe7b2e404d90ac04d58baa3318 : SUCCESS in 5m 49s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c102377ec4c64dba9c86bb414d3b32fb : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/87c4d6597f51446995c43daa1be18a26 : SUCCESS in 8m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dd732e3949d44ce2814ccf8897e27f05 : SUCCESS in 11m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0db746cd7c364159a99ca7ed004a380b : SUCCESS in 9m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55c8c407433c413eb9744c0ac43bac14 : SUCCESS in 44m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/93f6bc4f060d4c37ab9b1f3ed9e2a95a : SUCCESS in 1h 15m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf44abc1b4d24724a9846e1e241fa328 : SUCCESS in 1h 14m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/c16a288f49c24923ac55e7f8c0a7487d : SUCCESS in 1h 09m 23s","accounts_in_message":[],"_revision_number":1},{"id":"40642c49b82f26c0a6d0d2125322315bf05a9634","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 06:13:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"53807bc820b72fd8a028ad3f1c7dd8724adb8ed6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:17:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"35d6e37c111f2a240d77c7c5e2fbf2c4f3c2531f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:17:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2f14ce59ac84a89adc423b9fde72bcb : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08f987c3e6e94fad89a13488de4f78c6 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7edcf37526c0410ebb83658457a9581e : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f07b343281a41709e21b010a0b29ba3 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b252bcaf1b94cbdbc55c9e2909b8b07 : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/794cb27456ca41fcaa64e2179cdf5694 : SUCCESS in 8m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/32abb010da4b45859b89706ffa697886 : SUCCESS in 9m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0c66ef0115e34b68bbfa0ae0569fad89 : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/665bab2e427b48c6a007fcedf5548bc5 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2bd0b4f2dfa945ab91ef002da7b89b23 : SUCCESS in 7m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/644a06b37d524f6dba14e09112e5b153 : SUCCESS in 52m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/060caf91ffba4a1bad4cdaa6ff4cf1a4 : SUCCESS in 17m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c68db4b32ede4cf1883be81f983b781c : SUCCESS in 11m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fb0812bd616d43339b0a650e522cf599 : SUCCESS in 10m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2bf7531742544c783a5f4af37b8b36a : SUCCESS in 20m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6038a2356f8a44e2a9ea3d0fe351a6ed : SUCCESS in 5m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0d9c62507cb349f69428bc786f90313d : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/092a7f48f330479e9f20e908163095d1 : SUCCESS in 7m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a6b766eda025497b86d9a8706613afec : SUCCESS in 12m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9597b2778ba04780a00c81f82041a717 : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab7c8f5073954af48623ae6ea18b9cfa : SUCCESS in 1h 00m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c1f823f523a04b9f973d419cacabf9e7 : SUCCESS in 1h 17m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f78bdc31e7245e583f17718dc86ca80 : SUCCESS in 1h 13m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/cde41b80a46b43168586bef798d00fcd : SUCCESS in 1h 06m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07d1f919f3117f315d399a50ec10cb5030ddda61","revisions":{"07d1f919f3117f315d399a50ec10cb5030ddda61":{"kind":"REWORK","_number":1,"created":"2020-11-02 15:21:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/760944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/760944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/760944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/760944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/760944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/760944/1"}}},"commit":{"parents":[{"commit":"8f60f32770155075514ac00f24c401f328b7057f","subject":"update constraint for oslo.messaging to new release 12.1.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f60f32770155075514ac00f24c401f328b7057f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-02 15:21:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-02 15:21:35.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.2.2","message":"update constraint for ovsdbapp to new release 1.2.2\n\nChange-Id: I993ec3548d6b10ac2cd4061f97a814a37bdcfe0d\nmeta:version: 1.2.2\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta:release:Commit: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta:release:Change-Id: I6edf2d0b8f936339f1b1df7e062b036e54937a5d\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07d1f919f3117f315d399a50ec10cb5030ddda61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07d1f919f3117f315d399a50ec10cb5030ddda61"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
