)]}'
{"id":"openstack%2Frequirements~760947","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Id632090d2d6795961a4e67b586d6db64bb5b0a0d","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Id632090d2d6795961a4e67b586d6db64bb5b0a0d","subject":"update constraint for ovsdbapp to new release 0.17.5","status":"MERGED","created":"2020-11-02 15:24:52.000000000","updated":"2020-11-03 08:00:47.000000000","submitted":"2020-11-03 08:00:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760947-1604390447314-692fdd93","meta_rev_id":"9b41632ef81b5c2c38c4623366828ecc119a5f22","_number":760947,"virtual_id_number":760947,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-03 08:00:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-02 16:20:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-11-02 15:36:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-11-02 16:20:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-02 15:36:20.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-02 16:20:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-03 08:00:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0efabb25e0719f404a41d4a4b732df70536e71d7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-02 15:24:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b539920e688df3eecac78db66a63a118415fe2d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-02 15:36:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6c957a80a9f05e2cb2322ff241d69cf4f11295ff","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-02 16:20:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"520fca6eee1a521f80f9a345602177986cd7da9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 00:15:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4ec65fb2af4b3482424e657d4a9965 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3cf8f8c44d94fc4a5ea7e6f2169a0fd : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd576ec9745b4ffb9a3a717d3926fb11 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6936e6573b6d467285ae1b4e83446e2e : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e7dda437b73400b92684c99eea9fd87 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6473f391de24e129f634371331c2f80 : SUCCESS in 4m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d343dd345d224b8daf50580f17ab594d : SUCCESS in 7m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7778d3bd0f564e7085e6db060a70b971 : SUCCESS in 7m 06s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/22fa2d69209d4c8193cfd894f375f611 : SUCCESS in 12m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b78ea385d6974a96b86643ca607c4099 : SUCCESS in 8m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/062779bcc5704f66bdc195aec8d0deb9 : SUCCESS in 5m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2537e3c3b27c4268977679f547ce3e95 : SUCCESS in 5m 05s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2a87908095934d67912a057363897142 : SUCCESS in 11m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/65973341205849c28d65fb6aab7d31d9 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/857b2636b46b471bb6e63e42f8b19d9a : SUCCESS in 6m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0c627ac1d7814ad49165614d3becbc61 : SUCCESS in 38m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/93c1984861d24348be0acbe09a9247b8 : SUCCESS in 13m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/af7bda93ec3445c9a1f064eb98e4111e : SUCCESS in 21m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/59fb70660626488da4ed3e2d03a174be : SUCCESS in 18m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d067634cae384521994d015769681a20 : SUCCESS in 13m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d63fea96f95249b4857dbd304589aa29 : SUCCESS in 12m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c6b0fe169f8d4c91a7be0f4359faec9b : SUCCESS in 13m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e62770b7df484c30ac863aa1ceade7ae : SUCCESS in 21m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d0583bd837c4ea6be172624538de215 : SUCCESS in 5m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3194619555684f17be84db293c0a881f : SUCCESS in 9m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/81e299898ddd454c8aa0a94bb262a747 : SUCCESS in 8m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/80eb942f17f24a75b2f6cbf901856d81 : SUCCESS in 1h 11m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6620dd0c1a0b4b528769d82ec1004d6b : SUCCESS in 1h 26m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/86e6296a27c94788b8388698007e6c55 : SUCCESS in 49m 12s","accounts_in_message":[],"_revision_number":1},{"id":"980ce1fd6f3e3f5f619152521c240147be8d6482","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 00:15:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ce1ec9c3f38af768752b71a93fcc3abd48de92d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 08:00:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27943e8765f840a68ea24d699f95baf3 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b0c5fe998c8f41af8f8a8a540fb7bdb3 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24784bf426b047eebbc27f557344a1f5 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66c430e20c714b7dbb96b33451a2856c : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf57ac92cdb0477f8242033e6eeeed37 : SUCCESS in 6m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdfd4a41d02b46c0ba3d314b83c22733 : SUCCESS in 6m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/89ba5ccf3f5b43138b19881af03dee7d : SUCCESS in 10m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dc06d87402864334b5f6fba6242f1d3b : SUCCESS in 17m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/03c2597613d94ffeb567a947c83758f1 : SUCCESS in 10m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6be8586b44be4991ae3db3db34ddd640 : SUCCESS in 8m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e4e3901eea054f46990730c683aeaeea : SUCCESS in 4m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/67f9cff20f4244e099e4412efd59b736 : SUCCESS in 4m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2f2730feb04b4f108ced351583bb13c1 : SUCCESS in 11m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0078585af4f24251b6c1ae4915be6e54 : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19734d8f78f743deab7ea18cac92d12e : SUCCESS in 6m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ef468143969248fa94849d94abf0031f : SUCCESS in 16m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/534470e48b4b4c9888874d9aa4099f5c : SUCCESS in 13m 56s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a7ba983051fb4fdc97903aa5a51eefa4 : SUCCESS in 22m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/276d105e46674e1497cdf2fa3f6f15e8 : SUCCESS in 29m 01s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9505aa1738bf4a119c69f13bd7678232 : SUCCESS in 16m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9edc41b9cf394abd9fef21aeb8346119 : SUCCESS in 13m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/69e471dc351544d3ba69164b61deeef4 : SUCCESS in 15m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23e2e423c3ef40b79d0357ec3d882450 : SUCCESS in 21m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fbb957c0562049c99fc6fce3fd25a612 : SUCCESS in 6m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a73bab4f7ca24e52a276045ebdf6a67a : SUCCESS in 9m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7911fa3b96a34da18e5aaaad9f3d4f07 : SUCCESS in 10m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/273b86dfaf6048ef8c2fbbe299e5243e : SUCCESS in 1h 04m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4ec0c5c85be540f194978295d4a660e8 : SUCCESS in 1h 25m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ac00ccbf96c144739e65e3811365949c : SUCCESS in 55m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ec0aaa6575a036eb79b69b8757b1ae92a69a9c94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 08:00:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ef1cbc5e8b8b412e42ac8c846207a7503dc8274","revisions":{"1ef1cbc5e8b8b412e42ac8c846207a7503dc8274":{"kind":"REWORK","_number":1,"created":"2020-11-02 15:24:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/760947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/760947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/760947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/760947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/760947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/760947/1"}}},"commit":{"parents":[{"commit":"cba060af6fca6c89e0df804c05949e0533d504f8","subject":"Move ansible-runner to 1.4.0 to fix unclosed fd in 1.3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cba060af6fca6c89e0df804c05949e0533d504f8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-02 15:24:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-02 15:24:50.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 0.17.5","message":"update constraint for ovsdbapp to new release 0.17.5\n\nChange-Id: Id632090d2d6795961a4e67b586d6db64bb5b0a0d\nmeta:version: 0.17.5\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta:release:Commit: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta:release:Change-Id: I0f7c47407ffb3899b92eef03be48242252f067e8\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ef1cbc5e8b8b412e42ac8c846207a7503dc8274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ef1cbc5e8b8b412e42ac8c846207a7503dc8274"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
