)]}'
{"id":"openstack%2Frequirements~761060","triplet_id":"openstack%2Frequirements~master~I217aff17580a76cfe0904861aa9fe56e295dd9c1","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I217aff17580a76cfe0904861aa9fe56e295dd9c1","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-11-03 06:57:28.000000000","updated":"2020-11-03 15:07:17.000000000","submitted":"2020-11-03 15:05:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761060-1604415911482-3f776333","meta_rev_id":"deb0514066f6d4d21e520a8ee2d5ee4f36fd36ad","_number":761060,"virtual_id_number":761060,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-03 15:05:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-11-03 07:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-11-03 07:10:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-03 07:10:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-11-03 15:05:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"866092a1740d6f030d1bae16642497c4e82a18a4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-03 06:57:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7810b8132f5caad10b7681753c0a3a9998a5e819","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-03 07:09:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"639a9128d9be4a5eeba6508a10e002fc292f1e98","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-03 07:10:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2ac300e01b7fbbedd1f3a851b7de5dc87f2ac293","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 07:27:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4cc6cfd5d22845939bc2fde80545dde6 : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":2},{"id":"a1cebe46614e2ba08e996f786f7c9c31db8f8a4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 13:48:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3d466680d5946ff96df0cfd1550685d : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e447e62d568a47a2a178efa730b9f8c9 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3298dd68ab104b87bd77fe2849b7d789 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5087675550d4e7eaad49c85b97dcb5e : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d64442b111f84dd19e632c1f952dedaf : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f310fa21de2e4c57b95be9d6518483ca : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56328896363d4b4f90d367b67c85ec50 : SUCCESS in 3m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/33782360e3354dc6bc93c0e04ac4860e : SUCCESS in 7m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/36ad91c6d993429881980dadd3a61a44 : SUCCESS in 15m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9ba6b00488234c4aa517af676239afb1 : SUCCESS in 8m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/16bf45d71cae4de78135cfabe81c2944 : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9811e3aa20ec4d149cba6933bfeeebd1 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b32e5ececd8f4ab285ad9a4777d43a26 : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f081c6fc94854ec0936b68da4ef8abf3 : SUCCESS in 6m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f85d18d5dd144673995cc0378cf05c37 : SUCCESS in 13m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e76209771ec84b99b43d3c27dfcc3bfc : SUCCESS in 25m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5ed3e1d3045340dfa43a09ee7652326c : SUCCESS in 13m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/59ed6a9e1d4747a8bb9eda98c6ad91f9 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/49da001d10c04e70b4f0c988c635530a : SUCCESS in 17m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a5033b31928a4a5694700120179e3f09 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67c1d5dd6ea24cfe9bd79000f8abb18c : SUCCESS in 7m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/73730d0e6d1d497f9313c84b4a90c374 : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8eeecc6c9ed24650a38ed46ddb2777cb : SUCCESS in 50m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cdef70472294a6f99f33a747d96e002 : SUCCESS in 1h 20m 19s","accounts_in_message":[],"_revision_number":2},{"id":"24f25cf91e7b91cddfff9210d68d1b459fb3d882","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 13:49:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc91b4a5eff0e40d916794d3f7c5362883fb5594","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 15:05:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"17537f1035b8c91b074f69a58acf8042432d9f9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 15:05:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c15d99d2d80c435f9f3006d8ccccc013 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afce6d58960643b589108daf62fba970 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02974449b9a64df9bdc3b80538ee8502 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/460ca279796845aca4c62d9a3455f595 : SUCCESS in 5m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/67ccfab7f4184155951e825c2e1e1f9f : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/40e3ae84fb9a41ee9506e9fc4ce71b70 : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a9c01509851b4fb986988b6394f23e03 : SUCCESS in 7m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a5bb68e5f8eb4fde92eeb8f72983c590 : SUCCESS in 8m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b9d9677f876b4d1596aed66851118110 : SUCCESS in 9m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/535ad703f64844f5ba02a40c38620fbc : SUCCESS in 4m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/977a28aa221048c29c8bd2ce8d244a3a : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34849c0bf0274e31b66dac81b749be30 : SUCCESS in 7m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e5a7e1add8334bd1812065d15328ecac : SUCCESS in 7m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7b79137aa0d249be9227f4e193d3d91f : SUCCESS in 15m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9f6beb32be1145789e6418eabb8c6945 : SUCCESS in 24m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ba84ec8a7cde4af58703344d23190437 : SUCCESS in 13m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/50fe3f9af077496eaaedc965e639eb3c : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/165d7858985b460e8c2977f41e334853 : SUCCESS in 17m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2c035fd198854f07b0beed72d4a578b6 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/01fb725185284cf5bbd3ea0c1526023a : SUCCESS in 8m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2392271278234c988a5131b3ee0a768a : SUCCESS in 10m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d9290ab20de4c5cb6d7a1509cf3586e : SUCCESS in 1h 04m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95ed0229a72a41f7ae9937a3dcb81afd : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":2},{"id":"71474c2b14a6736483d2a7df06bcd7f15dfc0e40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 15:07:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a60ac4006654f90bee403e079a8ec7b : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb0e01fe5d65c4439231d06588f38ae95fa2f4e1","revisions":{"d79822e4141bb09cd8abefc88e631accc17bd5db":{"kind":"REWORK","_number":1,"created":"2020-11-03 06:57:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/761060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/761060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/761060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/761060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/761060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/761060/1"}}},"commit":{"parents":[{"commit":"6f836ec417e1893723bf5fffce0be2a410254d80","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f836ec417e1893723bf5fffce0be2a410254d80"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-03 06:57:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-03 06:57:25.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I217aff17580a76cfe0904861aa9fe56e295dd9c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d79822e4141bb09cd8abefc88e631accc17bd5db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d79822e4141bb09cd8abefc88e631accc17bd5db"}]},"branch":"refs/heads/master"},"bb0e01fe5d65c4439231d06588f38ae95fa2f4e1":{"kind":"REWORK","_number":2,"created":"2020-11-03 07:09:25.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/60/761060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/761060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/761060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/761060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/761060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/761060/2"}}},"commit":{"parents":[{"commit":"6f836ec417e1893723bf5fffce0be2a410254d80","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f836ec417e1893723bf5fffce0be2a410254d80"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-03 06:57:25.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-11-03 07:09:08.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I217aff17580a76cfe0904861aa9fe56e295dd9c1\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb0e01fe5d65c4439231d06588f38ae95fa2f4e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb0e01fe5d65c4439231d06588f38ae95fa2f4e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
