)]}'
{"id":"openstack%2Frequirements~761194","triplet_id":"openstack%2Frequirements~master~Ic800d4f93cc324787fe22e27a25411e1565185d5","project":"openstack/requirements","branch":"master","topic":"bug/1902696","hashtags":[],"change_id":"Ic800d4f93cc324787fe22e27a25411e1565185d5","subject":"Pin kombu and amqp requirements due to incompatibility","status":"ABANDONED","created":"2020-11-03 16:43:09.000000000","updated":"2021-03-08 11:18:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"afc51ecc7a8b991c72d14e1bc9bcd757bb82a53e","_number":761194,"virtual_id_number":761194,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-02 02:25:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-12-17 10:53:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}],"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-03 17:19:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-04 13:38:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-04 16:29:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-11-05 18:53:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-11-18 15:18:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-07 01:15:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"}],"messages":[{"id":"2a47607ee3fc0c47178567c39c49f673a7629d62","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-03 16:43:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cd046a8a2a47d63459bd501d4a4cc4ddf8b4fa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 17:08:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b85d717af7b543ed8146066d9281160b : FAILURE in 14m 38s","accounts_in_message":[],"_revision_number":1},{"id":"6423d9a2c170cc237198278c01b792f37688cfef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 02:52:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7583a107abc54666a5c7e463985af6b0 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e380c3a7aa34eb2a85d2732b63b5655 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ce80dfeccd542d088c3789fb7e94169 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/255c2a8f0f104c778a4440579692b608 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/262dc043f4ed49068edd817c02c9a572 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/acd119b6521f4ccf86aa92eb14bf9267 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/065feb02a55e43519f779eb9e4c6c350 : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3c22e4918cd24e95b8ded6738d8bf2e6 : FAILURE in 6m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6cacc1e252514ba5bbf4715bea571a57 : FAILURE in 7m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b788f45501ee4f8990d1c420aa34dc53 : SUCCESS in 8m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4bcf077a545d49dd97e69b46b4269096 : SUCCESS in 4m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b53d595e204b40adb9d3eefee6b0c0ce : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ffbb7bd242df4fb3aa991c7d7bcd0417 : SUCCESS in 8m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/208e120a0ff74c08b8902de6268a5de6 : SUCCESS in 10m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fc3d81642ade49e3ba19de08ed21583e : SUCCESS in 12m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3246ebfddd0c40039dc8e169bfa9330f : SUCCESS in 25m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c2f6cd0f94ca4e5cb0a7405d85d17b46 : SUCCESS in 13m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/537afd3d44584e24868afeeb4ce5bdb6 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a41be18e621044fdb2be993f3ddd4914 : SUCCESS in 18m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8d8e61a48a49437786216a42ca7e6b5d : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6948d620558f4acebe355d76cad7ba3f : SUCCESS in 10m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/886bdc8fe56945139639a1e1e5353914 : SUCCESS in 9m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6dd9d37b80524cb5ae05082d38cd8dd9 : POST_FAILURE in 21m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e33968a8b404a51a1517ce72b2c4a0a : FAILURE in 26m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f823f13e24c7c0e66e65566fd333c7c28da3adb3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-04 08:25:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2912691122e762a3d31639b99ee369963df9bba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 08:43:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b19015ef6b2a477089fa4e5ce9bba36b : SUCCESS in 13m 43s","accounts_in_message":[],"_revision_number":2},{"id":"370cca4571d3d1ceefc1ae4b5eb54f48f3b411c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 13:38:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c72e4f2758e46719d6778870bf21a17 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e6a2b236c954806925b1acf8e12b6bb : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a1772ce70d7491fb006d15411e3dac3 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b8a33f84b6044109b7e9ad52f1b3f2c : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0357793a7ef45f9a3b488c7e7927fd9 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e728e242f774e2691c9bd6d3aaa820d : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08be0e351b2b45369b4312f1a28ac06c : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b986e2b2d6b74af4b20de345cc77cf1f : SUCCESS in 9m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/86ded7114cd24492a089c33b8de0a8b8 : SUCCESS in 8m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dc298884469849a9b8067f4d68a535cf : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/da35dd385809464eae34d2d597526b71 : SUCCESS in 4m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/de847781e77747efb255cb731a4c6ddf : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b871f789c4d3460d8b68d1b42307b4a0 : SUCCESS in 7m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/381e25c5d50144f08a48cfa451de060b : SUCCESS in 6m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8ae298128d474ed7b238e207142331b3 : SUCCESS in 15m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/82560c8865094ea2b6e48b1cafbc3f7a : SUCCESS in 29m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d5cb8970eed9472cbef8263d7575c8ca : SUCCESS in 12m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/05060b6eb21a4d3d93010d91bb2ba35d : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3801f54985c44ae4808b76b61dc88c4a : SUCCESS in 19m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/709a9279ecbf4afd8770a17ddc07be9e : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5bfbb39f2e5f4ef297ea32e92fdb921e : SUCCESS in 8m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b61a4fd6331445d3bf7c8f5dbacdab25 : SUCCESS in 9m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50e3a1284c12414a8593c4f43fbaa7e5 : SUCCESS in 56m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0042731bc7f419a9c45489b824a7515 : SUCCESS in 1h 25m 16s","accounts_in_message":[],"_revision_number":2},{"id":"f5afca7fbe1a1ba3bf59c7b5cff7187736832e5a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-04 16:29:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b7c1da80ed9f610cd5f73b3979f8e44537b38d95","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-05 18:53:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"706363e2147febec92ffaa14bc254870c650e545","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-11 12:18:11.000000000","message":"Patch Set 2:\n\nIt seems that pyamqp re-introduced the ca_certs parameter in [1] and released that fix in 5.0.2[2]. Still we need to wait for kombu to release a new version that requires pyamqp 5.0.2. Then I will updated this path to bump kombu and amqp package version upwards instead of pinning downwards.\n\n\n[1]https://github.com/celery/py-amqp/pull/344\n[2]https://github.com/celery/py-amqp/commits/v5.0.2","accounts_in_message":[],"_revision_number":2},{"id":"7bb01e7612ce4be91420eb13b883554156144de8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-18 15:18:40.000000000","message":"Patch Set 2:\n\n\u003e It seems that pyamqp re-introduced the ca_certs parameter in [1]\n \u003e and released that fix in 5.0.2[2]. Still we need to wait for kombu\n \u003e to release a new version that requires pyamqp 5.0.2. Then I will\n \u003e updated this path to bump kombu and amqp package version upwards\n \u003e instead of pinning downwards.\n \u003e \n \u003e \n \u003e [1]https://github.com/celery/py-amqp/pull/344\n \u003e [2]https://github.com/celery/py-amqp/commits/v5.0.2\n\nDo we need to wait for a new kombu release? They have 5\u003c\u003dpyamqp\u003c6, which should be fine.","accounts_in_message":[],"_revision_number":2},{"id":"421b2045ec62b5056570946c132cdc7c966f9a42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-24 12:06:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7e6582b8b1f00f79497d45e26c77681352d1850a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-24 12:07:19.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e It seems that pyamqp re-introduced the ca_certs parameter in [1]\n\u003e  \u003e and released that fix in 5.0.2[2]. Still we need to wait for kombu\n\u003e  \u003e to release a new version that requires pyamqp 5.0.2. Then I will\n\u003e  \u003e updated this path to bump kombu and amqp package version upwards\n\u003e  \u003e instead of pinning downwards.\n\u003e  \u003e \n\u003e  \u003e \n\u003e  \u003e [1]https://github.com/celery/py-amqp/pull/344\n\u003e  \u003e [2]https://github.com/celery/py-amqp/commits/v5.0.2\n\u003e \n\u003e Do we need to wait for a new kombu release? They have 5\u003c\u003dpyamqp\u003c6, which should be fine.\n\nOK, lets try this. See PS3","accounts_in_message":[],"_revision_number":3},{"id":"fd0a40c46dd04cf4596f148997121a8c06012f02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 12:28:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4eebee9d55af4d98931df4460310b0d4 : SUCCESS in 15m 57s","accounts_in_message":[],"_revision_number":3},{"id":"10cbe7dc4dd89c64eb09edb7ea963f3019bfd5de","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 12:42:14.000000000","message":"Patch Set 3:\n\nTesting here: https://review.opendev.org/c/openstack/kolla-ansible/+/761519","accounts_in_message":[],"_revision_number":3},{"id":"20f94724929ab242e79a079990a7e555ac49289e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 15:36:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e7fae7256814d4aaca7944b7f42a206 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/783edbb3e9df4cd79927fb79c66d2b47 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1839f677ac9c4799b2462ec2a1660434 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/793fdd7c38cc4c47b0fc838f41c50da0 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f2ce40d815d4889bfad32cbdc7c7336 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/902139d4a8f143028d902cbd36735683 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e792947d7bd48559f3660e6b3f1ee5f : SUCCESS in 7m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cc08b89c3b524c188a91130ca6ae06a3 : SUCCESS in 9m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f2ea8e0644c440749f42fe1d1f5486f2 : SUCCESS in 12m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6300cca69466465abcb102ca029fb50c : FAILURE in 6m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/64529a1b0a95485b81a2b8f9152ad48d : SUCCESS in 4m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/eaea46eb14fa411bb073fff7c40ab100 : SUCCESS in 8m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/839ef67d5f134774b92a88d034bff824 : SUCCESS in 7m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5d970638b20b453bbb94942b0a58f10d : SUCCESS in 7m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dbb997bc2ec04c3f8621c796525881e8 : SUCCESS in 16m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b3c61d95938c4b36a343170656169dc6 : SUCCESS in 30m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e247bf786d734a44b6cb4d99b84fc9d5 : SUCCESS in 12m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9c0ed7fc319444c9c35cfcabc4bfbfd : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a51d2e60f1a044e0a734af6dd5891e5b : SUCCESS in 18m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/632530b024734c039b37d1e9c9d7f6f2 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b87c20e601d47e8816bf0195c95bf05 : SUCCESS in 6m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/30fae003f36746d88769517d78770f2a : SUCCESS in 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/428a919ef76d4e5586ad8a0164d57b4f : SUCCESS in 55m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53133098f6864f8aabc333b4a58d624e : SUCCESS in 1h 23m 08s","accounts_in_message":[],"_revision_number":3},{"id":"165804097d011046105928950d550e29c23dca11","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-24 17:11:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e1ff57a605807041c629cf106575b402e0179501","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 17:29:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f029a9c8cbe64611a852be36b57f51b0 : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":3},{"id":"823bf2ac9a0b300ae332094d4cef8f357fafb1d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 22:52:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d4331ddfd1d4efdb6fb110bd99a3ec9 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0494ba9557e949adb390d0fe2de3a457 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/067d881fe0e141a4ae5ca0c1a195a54d : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/196f842ad8a449ea9f3a90de77918c43 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ece935d7f19448b0bbc8d0282f4da744 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a60196dc93a849629f06713a61812235 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bcb39251a4234ae68f56706dce5560c8 : SUCCESS in 3m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b139ea7d6dd7424f9ebff2734bdaa5ae : SUCCESS in 7m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/44952dd5707e4d51bd77cdb1275fe98a : SUCCESS in 8m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/055727a02a1e44139228c6f34b08976f : SUCCESS in 11m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cf43b1c5f05b4ab0b713395c0b13d535 : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cf598887b5bb4519a3a0aee74afa17ed : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/44d35c6f8fd14b6097067b0f5eb705f8 : SUCCESS in 6m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c08b709124684ab5b22bf539882ff940 : SUCCESS in 6m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5bc8bf59fa7e4c13b70d7a50c6a0f86b : SUCCESS in 12m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2dca8b3c34e04db19698cad0600b1c76 : SUCCESS in 23m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/280d187b72484fe1b6fcebdbf3dd98f7 : SUCCESS in 14m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/39ab95baae254fde820d53ee02c20b3f : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f272f8532e914edeb5d6d562c590bf5b : SUCCESS in 25m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1fa53e8bb8174ede81421894480cffb1 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8037da60838946e393acc5819ff0d563 : SUCCESS in 7m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/564fe59c325a4d1f916823ec46738eb4 : SUCCESS in 9m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52985ab9777f424381a2ef4e2f80839c : SUCCESS in 1h 23m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4397d1e202ad495a87f5a518283e90fb : SUCCESS in 1h 49m 38s","accounts_in_message":[],"_revision_number":3},{"id":"9ca358c2ad18f221e919a4d5728020cef7686eb5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-25 14:33:24.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e Testing here: https://review.opendev.org/c/openstack/kolla-ansible/+/761519\n\nFailed","accounts_in_message":[],"_revision_number":3},{"id":"d268eb8c2fb684a929620d25af46ed2a4814b9de","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-25 15:12:37.000000000","message":"Patch Set 3:\n\nINFO:kolla.common.utils.nova-base:Requirement already satisfied, skipping upgrade: amqp\u003d\u003d\u003d5.0.2 in /var/lib/kolla/venv/lib/python3.6/site-packages (from -c /requirements/upper-constraints.txt (line 218)) (5.0.2)\n\nbut 5.0.2 does not help tls","accounts_in_message":[],"_revision_number":3},{"id":"ab2a373dfff8c9fedbe63373ebf461c1a01b9154","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-25 15:13:40.000000000","message":"Patch Set 3:\n\n2020-11-24 13:17:52.772 6 CRITICAL nova [req-74089aa5-46e0-4759-a131-77154d4833d7 - - - - -] Unhandled error: ValueError: check_hostname needs a SSL context with either CERT_OPTIONAL or CERT_REQUIRED","accounts_in_message":[],"_revision_number":3},{"id":"0d3f6cdec52c640c8ed5387d3735a8a02d1e607a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-25 15:16:01.000000000","message":"Patch Set 3:\n\nMark discovered PS2 worked","accounts_in_message":[],"_revision_number":3},{"id":"b510d076d8e8f6ca71406a96121ed305b04eee80","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-26 13:34:41.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e 2020-11-24 13:17:52.772 6 CRITICAL nova [req-74089aa5-46e0-4759-a131-77154d4833d7 - - - - -] Unhandled error: ValueError: check_hostname needs a SSL context with either CERT_OPTIONAL or CERT_REQUIRED\n\nI guess then there are multiple bugs after amqp 5.0 bump. 5.0.2 fixed one, but not all. I will push back the downward pin patch (PS2).","accounts_in_message":[],"_revision_number":3},{"id":"a98fd2eac62bbcd93f718016a96e2de385643d0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-26 13:55:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b725518bd6224c0446d056bc26c8bd17544d94db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 13:56:07.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"efd5abc0470dcb202b387f6e8da7f8855cb2befa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 13:56:13.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"7a86987b387ba17f870900a0f8ae1b2c3fe26095","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-26 14:17:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8fbda9223cb600d7092db085e529dd69d896c161","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 14:41:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d263c6a0e1a045c0a984b7955b6cb01e : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":5},{"id":"b3a5578ebdea3a42b7b2a8d91c6415c0b1c4740c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 19:51:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8199d86002f41a1a36f5a9ac6672a47 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aa899567dab4220b26db257452766c0 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7501a2c0a864f96bceca48c271d5945 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d232e99b7e24f1f9c4074c02912ca99 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fb212da9d7e4627b9f4999de1d6b953 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1fda7c088a7a4a8fa39c20b48483ca2d : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bc23e04f76e443029ea5f4d85d693ab1 : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d327ed8098894cb6a62747368f65ffa5 : SUCCESS in 8m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/11499d9ee161488ab7a35a3a1dd3bff2 : SUCCESS in 7m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/108e99583de442c590d318e029e7c3c6 : SUCCESS in 7m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/45a2a69f4ea14789ab0f69b7c2c056df : SUCCESS in 5m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4b5e00d2996348319c7e73bb2b99680c : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34c99f62353245de9e2f183cc1266485 : SUCCESS in 6m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2af6620612b646a890cff2cf65fa8721 : SUCCESS in 6m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1f060827c0d44cb5a0647b9e98117ccf : SUCCESS in 12m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ac7e1d7a477d49a184b49f40fce6f123 : SUCCESS in 26m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fd69498251554b7d811b51e8c6aec5b9 : SUCCESS in 12m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a45af8ff26fc449d93e5caed3685f437 : SUCCESS in 5m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6706cb4c0f4941f690420f01a73bf4ee : SUCCESS in 19m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/45adbf0fcb67429798d0e87579776d22 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/65e0adb8356f4b6fae06faeb45b5dcd4 : SUCCESS in 7m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/00b88214eba54ede9d0df1bbada15412 : FAILURE in 11m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b530ad299e5b46c09b8b02bdc9412f73 : SUCCESS in 2h 32m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a98828e2137743f69bc93b89279dff70 : SUCCESS in 1h 30m 48s","accounts_in_message":[],"_revision_number":5},{"id":"51ba80b17d745913fcdcf75b277510eed67a36be","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-30 09:00:32.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"5644ae1ba980b73313cb1fb438bb4c276ed050d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 09:20:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/454f5bcf555e411ea1449d0035e76850 : SUCCESS in 16m 31s","accounts_in_message":[],"_revision_number":5},{"id":"9c4a383a61bbc5f92b0979d7ed3d145a3f3ef964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 11:53:11.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/142d852f58a84bf58cb75645e08d13fb : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed66a59d75ae44309072459984931bba : SUCCESS in 11m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c963a040b35d4dbe90319ab14af733c0 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1808614fdda4aa6bd37315cca4b8dde : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ea5b44d12784756ac9b283e78b52922 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a026cf6647049a39ab060e7e16bb8bf : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e31bb280d477416c9273d72d227eb5cd : SUCCESS in 6m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/29d35b87cc9740628393f6bc93723c4e : SUCCESS in 21m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/777865a5e3554973b69f8fdeadf6db8b : SUCCESS in 7m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f359082075bd460c86f5146ba2299f62 : SUCCESS in 13m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/13d4ee69913e4c4a9bcb260f335c5806 : SUCCESS in 4m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e4a2d7bc3c1141f7b338eb64f9057eb9 : SUCCESS in 7m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b739c590e9fc4af2b3026a305cd4aa89 : SUCCESS in 6m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4c6f4ee74a12430f8420df75a53aeadd : SUCCESS in 6m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b48ff195f03a425eaeb927a2047a3312 : SUCCESS in 12m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e555dea1ce794c609a2e555779c85e32 : SUCCESS in 20m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a705061fd92742cb9d3dd4f536a056fa : FAILURE in 25m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d66578175b6a4be7b81f8d37ed5fdac2 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7fb92b619a24a23b1286ba38fb8b200 : SUCCESS in 17m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a6b7a82193d544e59aa4a733730a66fe : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a0d4fcbcd5d0425cabebb97a2251e3b2 : SUCCESS in 5m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b97c938d1ceb4b89ba0def1aa6d025ab : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/982e674ac02a4de5b0453f1403cb290a : SUCCESS in 56m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4d16804e3af475190bb2a74e7f015d6 : SUCCESS in 1h 32m 54s","accounts_in_message":[],"_revision_number":5},{"id":"41349a3ba148b75bc52a8a7c135b410c7045d770","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-12-01 16:05:11.000000000","message":"Patch Set 5:\n\nrecheck bug 1823251","accounts_in_message":[],"_revision_number":5},{"id":"73a1eb2185b566365829cf0c640246f83c5f93eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 16:25:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8dfa912943b64f2b989a1e0bccb82d03 : SUCCESS in 17m 04s","accounts_in_message":[],"_revision_number":5},{"id":"ed3913bfbf94b77955864242355b500f066e78e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 02:25:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/150b711d4ef947928d83fe465ada11d6 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b791fbbd650408696f436a9858f5e9e : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/991c3c38a54a4d7bb734b47adbf3450a : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60abad0cf42d44b1a67d8d07ae857d9e : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fe49bf3773c49b4ba114089b8e6b460 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98f5b6dad314405ab4f8b58013bcce09 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f637eb2807d34403ad110d71d9f6486d : SUCCESS in 3m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c493055dca3549349324aa56f1c956bc : SUCCESS in 8m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f89a9844279e4c0fb90ef215476323c1 : SUCCESS in 8m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/27de9899c7e14b159e03f8d961c935ed : SUCCESS in 8m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/22285633ef624ba08b6230071ddd4cab : SUCCESS in 4m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aff9f1bf44a342a0a60a6d14094cf926 : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fcce965be0e64a8895c5ab0aa108803f : SUCCESS in 9m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b17864f4c64b4768b20b6e5db35b12cc : SUCCESS in 20m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/51da1ca1652140118d4da78cb8928960 : SUCCESS in 29m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f6b1ca107bfe491aa6d14d885f677df7 : SUCCESS in 24m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/099b2127f09c4698b27d3877a2d96f28 : SUCCESS in 30m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d6b09670d555406d8adb3585a17e2722 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b4573174ded04548b84db985fb77007e : SUCCESS in 19m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aab799d7ac0742a4ba52cd3564a36e24 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90b4bf498c9549cbbcd0ee8897dd8ffa : SUCCESS in 6m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5895e9835f0740d3a52983c0aeee7fb8 : SUCCESS in 9m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13a6359c75574f3fae84237469565335 : SUCCESS in 58m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/862313a794964a71b033aa65c2d5728d : SUCCESS in 1h 58m 14s","accounts_in_message":[],"_revision_number":5},{"id":"82b807e76f173fd1769d92291379483d7b6955d2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 10:53:50.000000000","message":"Patch Set 5: Code-Review+1\n\nThis does indeed fix the issue.","accounts_in_message":[],"_revision_number":5},{"id":"9dae49ac118e33cb2bf49d0923a2626ace868581","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-07 01:15:01.000000000","message":"Patch Set 5:\n\nis this still needed?","accounts_in_message":[],"_revision_number":5},{"id":"d1cdab54af68250256e128395536f1ab961fd06e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-07 08:45:03.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e is this still needed?\n\nI doubt. TLS seems working again.","accounts_in_message":[],"_revision_number":5},{"id":"ebe52d671191fc7cd989b514d74f5d562e68c2f8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-07 18:21:24.000000000","message":"Abandoned\n\nok, feel free to restore if needed","accounts_in_message":[],"_revision_number":5},{"id":"afc51ecc7a8b991c72d14e1bc9bcd757bb82a53e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-08 11:18:53.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e is this still needed?\n\nThe original issue is no longer present on master, presumably due to some other requirements change. However, the stable/victoria branch still sees the issue.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e6c457ce452586749e554e036f92ec06876be39f","revisions":{"81810db85e4984acae7601adb2367a52a715ed7b":{"kind":"REWORK","_number":1,"created":"2020-11-03 16:43:09.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/94/761194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/761194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/761194/1"}}},"commit":{"parents":[{"commit":"bb0e01fe5d65c4439231d06588f38ae95fa2f4e1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb0e01fe5d65c4439231d06588f38ae95fa2f4e1"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-03 16:40:24.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-03 16:40:24.000000000","tz":60},"subject":"Pin kombu and amqp requirements due to incompatibility","message":"Pin kombu and amqp requirements due to incompatibility\n\nkombu 5.0.2 bumped the amqp dependency from 2.6.1 to 5.0 but amqp 5.0\nremoved ca_certs ssl parameter from the connect call path. The\noslo.messaging still passes that parameter making nova fail if ssl is\nconfigured for RPC.\n\nRelated-Bug: #1902696\nChange-Id: Ic800d4f93cc324787fe22e27a25411e1565185d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81810db85e4984acae7601adb2367a52a715ed7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81810db85e4984acae7601adb2367a52a715ed7b"}]},"branch":"refs/heads/master"},"5c19f97c1a3997cee442f6ebdff7a54320a39790":{"kind":"REWORK","_number":2,"created":"2020-11-04 08:25:07.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/94/761194/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/761194/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/761194/2"}}},"commit":{"parents":[{"commit":"bb0e01fe5d65c4439231d06588f38ae95fa2f4e1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb0e01fe5d65c4439231d06588f38ae95fa2f4e1"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-03 16:40:24.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-04 08:23:29.000000000","tz":60},"subject":"Pin kombu and amqp requirements due to incompatibility","message":"Pin kombu and amqp requirements due to incompatibility\n\nkombu 5.0.2 bumped the amqp dependency from 2.6.1 to 5.0 but amqp 5.0\nremoved ca_certs ssl parameter from the connect call path. The\noslo.messaging still passes that parameter making nova fail if ssl is\nconfigured for RPC.\n\nTo keep the resulting constraints consistent the vine constraint moved\nback to 1.3.0 from 5.0.0\n\nThis patch partially reverts I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n\nRelated-Bug: #1902696\nChange-Id: Ic800d4f93cc324787fe22e27a25411e1565185d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c19f97c1a3997cee442f6ebdff7a54320a39790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c19f97c1a3997cee442f6ebdff7a54320a39790"}]},"branch":"refs/heads/master"},"f7ddb1139854f1bd9169c1d336f65b99235a042b":{"kind":"REWORK","_number":3,"created":"2020-11-24 12:06:15.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/94/761194/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/761194/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/761194/3"}}},"commit":{"parents":[{"commit":"bb0e01fe5d65c4439231d06588f38ae95fa2f4e1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb0e01fe5d65c4439231d06588f38ae95fa2f4e1"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-03 16:40:24.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-24 12:03:48.000000000","tz":60},"subject":"Avoid amqp 5.0.0 and 5.0.1 due to missing ca_certs","message":"Avoid amqp 5.0.0 and 5.0.1 due to missing ca_certs\n\nDue to a bug [1] The oslo.messaging fails to work with amqp if SSL is\nconfigured. The bug has been fixed in amqp version 5.0.2\n\n[1] https://github.com/celery/py-amqp/issues/342\n\nRelated-Bug: #1902696\nChange-Id: Ic800d4f93cc324787fe22e27a25411e1565185d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7ddb1139854f1bd9169c1d336f65b99235a042b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7ddb1139854f1bd9169c1d336f65b99235a042b"}]},"branch":"refs/heads/master"},"d24d0b75baff50e7190fca478ec2e0fbc65b963a":{"kind":"REWORK","_number":4,"created":"2020-11-26 13:55:51.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/94/761194/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/761194/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/761194/4"}}},"commit":{"parents":[{"commit":"bb0e01fe5d65c4439231d06588f38ae95fa2f4e1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb0e01fe5d65c4439231d06588f38ae95fa2f4e1"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-03 16:40:24.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-26 13:53:48.000000000","tz":60},"subject":"Pin kombu and amqp requirements due to incompatibility","message":"Pin kombu and amqp requirements due to incompatibility\n\nkombu 5.0.2 bumped the amqp dependency from 2.6.1 to 5.0 but amqp 5.0\nremoved ca_certs ssl parameter from the connect call path. The\noslo.messaging still passes that parameter making nova fail if ssl is\nconfigured for RPC.\n\nLater amqp 5.0.2 reintroduced the ca_certs parameter but kolla now fails\nwith\n\nValueError: check_hostname needs a SSL context with either CERT_OPTIONAL or CERT_REQUIRED\n\nTo keep the resulting constraints consistent the vine constraint moved\nback to 1.3.0 from 5.0.0\n\nThis patch partially reverts I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n\nRelated-Bug: #1902696\nChange-Id: Ic800d4f93cc324787fe22e27a25411e1565185d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d24d0b75baff50e7190fca478ec2e0fbc65b963a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d24d0b75baff50e7190fca478ec2e0fbc65b963a"}]},"branch":"refs/heads/master"},"e6c457ce452586749e554e036f92ec06876be39f":{"kind":"REWORK","_number":5,"created":"2020-11-26 14:17:56.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/94/761194/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/761194/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/761194/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/761194/5"}}},"commit":{"parents":[{"commit":"4c7d26b3e3b3bef05443d66661b6d7b53f755879","subject":"Merge \"update constraint for oslo.upgradecheck to new release 1.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c7d26b3e3b3bef05443d66661b6d7b53f755879"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-03 16:40:24.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-26 14:17:30.000000000","tz":60},"subject":"Pin kombu and amqp requirements due to incompatibility","message":"Pin kombu and amqp requirements due to incompatibility\n\nkombu 5.0.2 bumped the amqp dependency from 2.6.1 to 5.0 but amqp 5.0\nremoved ca_certs ssl parameter from the connect call path. The\noslo.messaging still passes that parameter making nova fail if ssl is\nconfigured for RPC.\n\nLater amqp 5.0.2 reintroduced the ca_certs parameter but kolla now fails\nwith\n\nValueError: check_hostname needs a SSL context with either CERT_OPTIONAL or CERT_REQUIRED\n\nTo keep the resulting constraints consistent the vine constraint moved\nback to 1.3.0 from 5.0.0\n\nThis patch partially reverts I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n\nRelated-Bug: #1902696\nChange-Id: Ic800d4f93cc324787fe22e27a25411e1565185d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6c457ce452586749e554e036f92ec06876be39f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6c457ce452586749e554e036f92ec06876be39f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
