)]}'
{"id":"openstack%2Frequirements~761256","triplet_id":"openstack%2Frequirements~master~Ib7cdc07f991e0900f0ef5d9406f90deb86a62443","project":"openstack/requirements","branch":"master","topic":"bug/pyghmi","hashtags":[],"change_id":"Ib7cdc07f991e0900f0ef5d9406f90deb86a62443","subject":"update pyghmi to 1.5.19","status":"MERGED","created":"2020-11-04 00:52:15.000000000","updated":"2020-11-16 19:50:02.000000000","submitted":"2020-11-16 19:45:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761256-1605555923633-28503777","meta_rev_id":"9c16d4faa70dc6c6fa8afede16e51ea0cfae7efc","_number":761256,"virtual_id_number":761256,"owner":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-16 19:45:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-11-16 17:58:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-11-04 08:28:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-11-16 17:58:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-04 08:28:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-16 17:58:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-11-16 19:45:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d12d69655b04c04936230bee9fb5136fb956379","author":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"date":"2020-11-04 00:52:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ad367f17160155d37c5e7bdbdc28457dda41cd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 01:11:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/04ca3d3ef4a14e20a16c34e3f521a0ab : SUCCESS in 14m 51s","accounts_in_message":[],"_revision_number":1},{"id":"4fd05900af5f5058c8f4947b7f8e604f3946e041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 06:57:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c96a87d423104b2699aac0ef98c12539 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d625f9e54c66453cac950d58f099c279 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51f811c1df9c4e94a1280f0a2be8443f : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69e938cf43a34b95b16c8b088f5b933d : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ba8729fedf9490eb4af3b2e1e4bb2e5 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03ac46ea95d244d3a98814e9e6afd371 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ef9da9ff3a1b4165b31ff2d10bfebd20 : SUCCESS in 3m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/335051953b274f62bc8c6c7a66eafd5a : SUCCESS in 8m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/86af24171ef441e386176d6d435f1375 : SUCCESS in 7m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5f29d7f31ff24a2d9057d78c2eb291fe : SUCCESS in 8m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fba75ea138fc454dbb0e2347bd12a083 : SUCCESS in 6m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e01e3100fbbe476eb25d6c2272d45d80 : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/36921f25170749078965d1e341db339c : SUCCESS in 7m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0fdc54afaf5c4c9980875dfae4a28e8a : SUCCESS in 6m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b323b2148c8242859120cf215bc77a06 : SUCCESS in 32m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b99d0bf1db6b4618ad68836d1bdead7a : SUCCESS in 26m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/39e3d3638e814013b6ea9bda2a3d60db : SUCCESS in 12m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/256c47d253354349aff847eadc42b833 : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/59493f24234746e88372e6b0c25ce80b : SUCCESS in 18m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a6b202198aef4fbd87d4d1734e432cfd : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c04bde6216944e519c628dcdd205ca9b : SUCCESS in 8m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8a65baaf14f3421085c702bded9d6ba0 : SUCCESS in 9m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34dfb1048d374ef892101e2f402b2638 : SUCCESS in 1h 12m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/799915246fd7426e8fe766e502a53cd8 : SUCCESS in 1h 18m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b8fc45f0884c79309e2de04a0d12d29c92e2472b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-04 08:28:56.000000000","message":"Patch Set 1: Code-Review+2\n\nDuplicated with https://review.opendev.org/#/c/761279 but isn\u0027t a big deal","accounts_in_message":[],"_revision_number":1},{"id":"d938ad1b19c2404462f7de04ee4dbb2de2bdfd3d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-16 17:58:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bdc6460a9fe181700d4a5dee22f475ad65d3e00d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 17:59:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"902d5b3d82bcf5a7b074f394bd7f56021befb5e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 19:45:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"98f3ef19c0249e94eb3144abf8f1ed8a906c6e74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 19:45:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ab48f1a03834274baf7477512d7e129 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/655e9bea77274b6fb325765314d46347 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07ac54eed0ce4b57abf356cf27510337 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94aca86be0ee473491936366184717bb : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a34b168a26044d199fc96d42333c950f : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/32646d3da8b64adb9fe010551ca499b5 : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9af28498b2de4c3c8d0c57791d2c643a : SUCCESS in 6m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3bcd49f2ce6d492785f82c4aa96b02a9 : SUCCESS in 7m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/93bd40b3080f4a1eb58b0a6a8e4355ce : SUCCESS in 9m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/406b75bad73548f2ae0cbe802980e4d2 : SUCCESS in 5m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2d37047caaf14392ace387f4d62c17dc : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6aa72d59b9534339bccc4251daef3fe3 : SUCCESS in 6m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/debccb7507814336a3316abd2207712a : SUCCESS in 6m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e7b0bf9eaffc40da8980b0933607083d : SUCCESS in 13m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1f2c1a5b3efa444db220592f37c67d56 : SUCCESS in 28m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aa146b1cb9fa4d358af250892edcf80c : SUCCESS in 15m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a020db38c676418581ca2e031cdd6341 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7865a333a3814b5086cbc2734335af7f : SUCCESS in 17m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/634ca4e22c16472db9d4a14af02e564a : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d0ac59b837204328aa15ed995e546df4 : SUCCESS in 7m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f924813f110042ea8c184297491eae84 : SUCCESS in 9m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e44d5863a9a54f968b8a711ea8a86c6a : SUCCESS in 1h 01m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9294d06bf7e34c989fb0f9b7b35ec78c : SUCCESS in 1h 21m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1c5de924a4fbeeeb94bd99b714dc4bf419b786c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 19:50:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05f9362f63ab4f5b8ddbbd09d4c34828 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34f5ec4825ee4bf8d8103d728b1aebc50713623d","revisions":{"34f5ec4825ee4bf8d8103d728b1aebc50713623d":{"kind":"REWORK","_number":1,"created":"2020-11-04 00:52:15.000000000","uploader":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"ref":"refs/changes/56/761256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/761256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/761256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/761256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/761256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/761256/1"}}},"commit":{"parents":[{"commit":"bb0e01fe5d65c4439231d06588f38ae95fa2f4e1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb0e01fe5d65c4439231d06588f38ae95fa2f4e1"}]}],"author":{"name":"chenming","email":"chenm30@chinaunicom.cn","date":"2020-11-04 00:53:01.000000000","tz":480},"committer":{"name":"chenming","email":"chenm30@chinaunicom.cn","date":"2020-11-04 00:53:01.000000000","tz":480},"subject":"update pyghmi to 1.5.19","message":"update pyghmi to 1.5.19\n\nChange-Id: Ib7cdc07f991e0900f0ef5d9406f90deb86a62443\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34f5ec4825ee4bf8d8103d728b1aebc50713623d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34f5ec4825ee4bf8d8103d728b1aebc50713623d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
