)]}'
{"id":"openstack%2Frequirements~761279","triplet_id":"openstack%2Frequirements~master~I0b1fa63092f6fd39b2cb606e974f0575faabaabb","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I0b1fa63092f6fd39b2cb606e974f0575faabaabb","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-11-04 06:30:33.000000000","updated":"2020-11-05 23:21:56.000000000","submitted":"2020-11-05 23:20:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761279-1604618450253-294064b8","meta_rev_id":"0ab435f1674fea4370524f2970c89cff4b35fe9d","_number":761279,"virtual_id_number":761279,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-05 23:20:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-11-05 21:59:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-11-05 21:59:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-05 21:59:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-11-05 23:20:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23bb5c9b40ee1ad48e1538a08805a375a0aa02c2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-04 06:30:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40c576935a80dc201823531e321dfe647bdb4814","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 06:47:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/17092650f0b54435852fd0491ba7533b : FAILURE in 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e7c9b73d4e3666f00061ff01b4c92e66a1def9f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 09:38:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f034762e243a43feb433dac893ede925 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27e86fe22f664fe298a06610dbd529d5 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/912a68747b2d4874a5a7ddc32bf437fa : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fbfeb92a2074a44be05638dd373947f : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/421a04a2a7c5404cbfbd5803c04844c1 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f62e8512e9d4d26bcb33532181b77e4 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b4ec94a9a1c4c8081023d14ee59c45a : SUCCESS in 3m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b7b64cb8de7644b292a73ec1d1827f40 : FAILURE in 4m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8c9e8741f0ca42f29ddc505c771b8a15 : FAILURE in 4m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1c5a8f777f404c0ba54158a999123a25 : SUCCESS in 8m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f897dd7c699e4e2eab24a0e00e924015 : SUCCESS in 5m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6a18b5f49b6c4fbab0cad79403db5784 : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d23f9860c48b4e5894644814affdf081 : SUCCESS in 7m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e0748ffc1d084850ab6b5a169ad19dc1 : SUCCESS in 7m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9e8870e68dcc44f29af93fafa6dede29 : SUCCESS in 24m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8975d18e6d904912b7327fae6beb61a0 : SUCCESS in 26m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/321800c85b4f447d9a0bd75a67e93a91 : FAILURE in 13m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a28b971689434fdaaf45225d9631896c : SUCCESS in 6m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6817f52cbdae447eb86d55e512350a82 : SUCCESS in 16m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/514b701f52ba44a0af7e23e6a0108fd2 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5d873283c184091855009d9a283e406 : SUCCESS in 7m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6b792e91f6bf4f989bbf68478a75c3cd : SUCCESS in 11m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3977cb4802b546d2a6c3fdbc1c769dd6 : SUCCESS in 1h 05m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b2125326d244954bfdbd4f10e681aed : SUCCESS in 1h 25m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1e0d2c2697b86b131df7ba808989e3969bf74ccb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-05 19:40:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"118738eed62c64a56b685871a97992de2888d8ae","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-05 19:40:09.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"db5c162fae928dd6d3037fe02ff93d8f23bbdf06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 19:58:28.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b5bbb932d7334d76aeba555282e5a514 : RETRY_LIMIT in 2m 02s","accounts_in_message":[],"_revision_number":2},{"id":"ed7da40cf8da2cab11b5289bcee1291867d58d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 21:55:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d90505ff53764b89b46d95999b034c43 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8979b136cad44fe18b1af36dd60af7ef : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a48d48292d20459cb3f8b5aebe6b96e9 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b929e33ab71e4fc785c8f1feeceafdde : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66fc392cb9e84c33a2f1fedf245e22c2 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bee7771954b744bc8a9e31760575b851 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4f296d2a51e4197a15dc68b3e6c86a0 : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e39b21b7fe2640af9e70d0a177502f3b : SUCCESS in 7m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8343032a34174701b63a32d0c60bcdf2 : SUCCESS in 7m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/658a230863fb4d639e4ba593f1dd59fc : SUCCESS in 8m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d59a05ef8efe44c093c763a1c15904e4 : SUCCESS in 4m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/67801ec1e629413ab3e3c52a4564a232 : SUCCESS in 8m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f9f5172e9234d9982cfec31be331d2d : SUCCESS in 7m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c159765b52f04f8e8a0515b94800a7c6 : SUCCESS in 6m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3a677eafe17d493d844c2a8972349bdf : SUCCESS in 14m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8ec6c2a6300c4b999f5475c912833f79 : SUCCESS in 28m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7f3ab3a501274c77ad033f40bd0b0494 : SUCCESS in 13m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/78222f492c634a87ad0180200a1dbe86 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/782c7d55094643b8bf2c796623de0f73 : SUCCESS in 18m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8578e89173e54d54bef8aade4d0b62ec : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/88c22d58a17f4134a3e2e503b61f53a9 : SUCCESS in 8m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9f0e5478bfa74c8fa5d2608db0e35543 : SUCCESS in 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/550a19f145ae46d3ac42bf2453412203 : SUCCESS in 55m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fe0ed8d61d54fa6b3eda0aefbe60abb : SUCCESS in 1h 15m 11s","accounts_in_message":[],"_revision_number":2},{"id":"19e700f9a45bc691bf03ebcc9f758b4c8b974e7d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-05 21:59:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"596fc9790a63b01772bac98d7559c3fc2fc27c3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 22:00:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"085fc901e6d4082f7f1394c94d095ed0dc731176","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 23:20:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/683fa28a5ae140839150171a0fe23405 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ead3b15bd274566bd066855c959cf75 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58a74ccccb954a258efb201a3b8b229f : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8298308613dc4adfa0a19104bdc81339 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3877d524e5524dc0a7ac7bd8a57290b9 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d28d83751d5548e8ba35f3af3aeb370d : SUCCESS in 3m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/441fd32cf22549dca3c0b028c7557e52 : SUCCESS in 7m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d773162cc4cf4a1b87a0633be65af56e : SUCCESS in 7m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/70e04e19bc5543698ac70705d9e07f41 : SUCCESS in 8m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5f72e4e3580b40669e270c4c8c0b0adc : SUCCESS in 4m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/980e8e85169c4532b87963d302630171 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b43e9ee21dd14ea3922a3b675610680b : SUCCESS in 6m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5befd063834f44708c1c9874aca56a1e : SUCCESS in 7m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e1f90dfaa5184211a1c8908607496077 : SUCCESS in 32m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4b1ded5a30b2470fa46cc1446a115efb : SUCCESS in 16m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c4283e0f1e2f4c3d98c9a9910dff3ff1 : SUCCESS in 10m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5aa6a1e039da446783f5aa2bd6a2aad2 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6bd9fac6c31a4abab2fd50c2daf27d13 : SUCCESS in 16m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cc420a3d843a42649be7e5ea968213c0 : SUCCESS in 4m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b74f1bc2ed8d4beea3203b49a88b4b31 : SUCCESS in 7m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/44d1a6f485dd4295bdfee2ef9e7058e5 : SUCCESS in 10m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5dd6fa86873a4cabb0e513195f4873ab : SUCCESS in 1h 00m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7f0ce7146644e3b80c6fcb387e427e6 : SUCCESS in 1h 14m 37s","accounts_in_message":[],"_revision_number":2},{"id":"9c508d30e1e39e8ba951718c7e361784351a482b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 23:20:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0ab435f1674fea4370524f2970c89cff4b35fe9d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 23:21:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce0177f671704aa7b289541b17b4262d : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"65894e7c3203586a3b1ed63a552a17a25aab19a5","revisions":{"cbe8c19d8e672df15141d136c8fd757d3f75a4f6":{"kind":"REWORK","_number":1,"created":"2020-11-04 06:30:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/761279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/761279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/761279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/761279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/761279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/761279/1"}}},"commit":{"parents":[{"commit":"bb0e01fe5d65c4439231d06588f38ae95fa2f4e1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb0e01fe5d65c4439231d06588f38ae95fa2f4e1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-04 06:30:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-04 06:30:31.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b1fa63092f6fd39b2cb606e974f0575faabaabb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cbe8c19d8e672df15141d136c8fd757d3f75a4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cbe8c19d8e672df15141d136c8fd757d3f75a4f6"}]},"branch":"refs/heads/master"},"65894e7c3203586a3b1ed63a552a17a25aab19a5":{"kind":"REWORK","_number":2,"created":"2020-11-05 19:40:08.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/79/761279/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/761279/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/761279/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/761279/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/761279/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/761279/2"}}},"commit":{"parents":[{"commit":"bb0e01fe5d65c4439231d06588f38ae95fa2f4e1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb0e01fe5d65c4439231d06588f38ae95fa2f4e1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-04 06:30:31.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-11-05 19:39:56.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b1fa63092f6fd39b2cb606e974f0575faabaabb\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65894e7c3203586a3b1ed63a552a17a25aab19a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65894e7c3203586a3b1ed63a552a17a25aab19a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
