)]}'
{"id":"openstack%2Frequirements~761432","triplet_id":"openstack%2Frequirements~master~I2acf0c3c6e7f919570fc35fb2bba668b7e3245a8","project":"openstack/requirements","branch":"master","topic":"bug/1888237","hashtags":[],"change_id":"I2acf0c3c6e7f919570fc35fb2bba668b7e3245a8","subject":"Pin dnspython \u003c 2.0.0","status":"ABANDONED","created":"2020-11-04 15:39:15.000000000","updated":"2021-02-23 17:30:59.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"585333333cfede9f5d124b62e460cc28f52c33d7","_number":761432,"virtual_id_number":761432,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-11-06 08:20:57.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2020-11-06 08:14:08.000000000","_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-04 21:27:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-04 21:27:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-06 08:20:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"15d6c1887d3e0d9f22ecc75e2e5118cd3d31534d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-04 15:39:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28a7c880c6da3878f2cd51c7e6ea915181d61b85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 21:27:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b65b8e61066f4b4bb56aa6d1ebe9a58a : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a78ce7231c7c4a8d879e4c2353e38bf7 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e3c1b5730ae485c8b54ff03f612b00a : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7db67e19fa6346a59c7ffecca04fa171 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6e1a68a050c425595cbde8cb4422de9 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf354843f0cb4727bba21b0856b0793f : SUCCESS in 3m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2013d4fda8d941029ea36305c3025fa8 : SUCCESS in 59m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db225eb9df4140c2b96fc3fda92fd98d : SUCCESS in 1h 17m 49s","accounts_in_message":[],"_revision_number":1},{"id":"924698e13a2753a465912e62d5442b638e8ae2a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-05 00:32:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8cfc7f05c2a920312e21555caa4dd4bf0506be55","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-05 08:22:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3857b8a7cfd94435cbe48d5f985615d0932cf4a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-05 17:50:44.000000000","message":"Patch Set 1:\n\nthe RC won\u0027t be included here unless someone tries to propose it manually, 2.1.0 would have to be excluded as well.  I\u0027m more inclined to push eventlet to update their library support than track what they exclude (on top of what projects like boto excludes).","accounts_in_message":[],"_revision_number":1},{"id":"8547cbdc42223b249fe36e3870725e5b23252678","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-06 06:46:56.000000000","message":"Patch Set 1:\n\nI guess what I\u0027m saying is that eventlet is not special about imposing a hard cap openstack wide.  Honestly the new pip resolver \u0027should\u0027 solve this.","accounts_in_message":[],"_revision_number":1},{"id":"315781ec92c4966e2211a4292cbee08ac29aaeea","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-06 08:14:08.000000000","message":"Patch Set 1:\n\n\u003e the RC won\u0027t be included here unless someone tries to propose it\n \u003e manually, 2.1.0 would have to be excluded as well.  I\u0027m more\n \u003e inclined to push eventlet to update their library support than\n \u003e track what they exclude (on top of what projects like boto\n \u003e excludes).\n\nAt the moment I don\u0027t see how can I encourage eventlet devs to do something for us other than go and contribute the patches myself, which is not happening due to time constraints on my side. \n\n\n \u003e I guess what I\u0027m saying is that eventlet is not special about\n \u003e imposing a hard cap openstack wide.  Honestly the new pip resolver\n \u003e \u0027should\u0027 solve this.\n\nOK. Let\u0027s wait and see what happens when dnspython 2.1.0 is released.","accounts_in_message":[],"_revision_number":1},{"id":"72a981658865a0e78c9c75a1a474ab640923979e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-06 08:20:16.000000000","message":"Patch Set 1:\n\nya, I never said the situation was nice (for either side), but putting caps in place just means we are another place that has to be check periodically for if the cap has been raised or removed. :|","accounts_in_message":[],"_revision_number":1},{"id":"bfdda3b1117760f82657eb4f5dd4cb4751f12cc2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-06 08:20:57.000000000","message":"Patch Set 1:\n\nshit situation all around, it gets everywhere, all I can do is try to prevent the spread :D","accounts_in_message":[],"_revision_number":1},{"id":"585333333cfede9f5d124b62e460cc28f52c33d7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-23 17:30:59.000000000","message":"Abandoned\n\nno update in 3 months, caps are bad :P","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c8712306d16ff7b69cad3c2ee1473f6a9c4821e","revisions":{"4c8712306d16ff7b69cad3c2ee1473f6a9c4821e":{"kind":"REWORK","_number":1,"created":"2020-11-04 15:39:15.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/32/761432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/761432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/761432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/761432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/761432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/761432/1"}}},"commit":{"parents":[{"commit":"bb0e01fe5d65c4439231d06588f38ae95fa2f4e1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb0e01fe5d65c4439231d06588f38ae95fa2f4e1"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-04 15:36:13.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-04 15:36:13.000000000","tz":60},"subject":"Pin dnspython \u003c 2.0.0","message":"Pin dnspython \u003c 2.0.0\n\nAs dsnpython 2.1.0 is around the corner and eventlet restricted the\ndnspython version to \u003c 2.0.0 we should change our dnspython!\u003d2.0.0\nrule to \u003c2.0.0 to avoid a break when dnspython 2.1.0 is released.\n\nRelated-Bug: #1888237\nChange-Id: I2acf0c3c6e7f919570fc35fb2bba668b7e3245a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c8712306d16ff7b69cad3c2ee1473f6a9c4821e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c8712306d16ff7b69cad3c2ee1473f6a9c4821e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
