)]}'
{"id":"openstack%2Frequirements~761737","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ib7bbdc0891640772008cdf087c6bb271cca4290c","project":"openstack/requirements","branch":"stable/train","hashtags":[],"change_id":"Ib7bbdc0891640772008cdf087c6bb271cca4290c","subject":"Bump eventlet u-c version","status":"MERGED","created":"2020-11-06 12:37:47.000000000","updated":"2020-11-16 19:48:32.000000000","submitted":"2020-11-16 19:48:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761737-1605556112559-33b2152c","meta_rev_id":"5e23a59f79f5096af4e2f977033e6ba16a8e665c","_number":761737,"virtual_id_number":761737,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-16 19:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-11-06 15:36:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-11-16 17:58:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-16 17:58:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-11-16 19:48:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af3d970f1ba5100ca88400a915858965d3590db7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-06 12:37:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9682891d297b7271e9ff6da601a977841fbabd46","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-06 15:36:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c3b16f5f68d622c334ba0964fbb6567c759842e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 16:24:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aad8efa227094d19b46872bb41bf7011 : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38dc86ee204342fd86cf25c0326435f2 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6696f4a6f4fd4ad080876630c0e44ea8 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf26917df676405bb14493f155d8c82c : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba87650ad9764a20b6d312b848cf24c5 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/de44c0c9032f4ea79ae864a259b2f239 : SUCCESS in 4m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/04db159b2a8741b6abb0706535c4ff34 : SUCCESS in 7m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/011a57d8bd184247aa8827e67d3153df : SUCCESS in 9m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1ce12cf5b9504f3a821ef6459f6402ae : SUCCESS in 10m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fe210fda494045f48ea092d181eca21c : SUCCESS in 15m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c97783780e9044548fc848243226b49d : SUCCESS in 4m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/38a2d22029204e69add1309642927188 : SUCCESS in 5m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/27cbce3b112349cf85f60c0670570e83 : SUCCESS in 8m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e19ba3d73ba34597b546d79a4022132e : SUCCESS in 9m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/881dc7c77d864ddd8d59802f317e6b10 : SUCCESS in 6m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/90e83685a1c8418aa59eda8ffc28285a : SUCCESS in 19m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d0b6585e03224182bfcac0a31a108ed7 : SUCCESS in 11m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/45472d9266de46a68eb3f757ba93b007 : SUCCESS in 30m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/338038f12afc47199c97595bf4e1f0e1 : SUCCESS in 27m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8594c8aa237346dd833a1ff151f9ee0b : SUCCESS in 11m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bf9064a8aab242db83b38de674e3d5b8 : SUCCESS in 13m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8548083789704360b13182ad3d1bb0c5 : SUCCESS in 15m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f5697bdd6bb24bf78af7fcc206ed9cbe : SUCCESS in 20m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c346f35a62a840e590a8fa147934938c : SUCCESS in 6m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/171c52a2813149fcbb85d8da6e8e736a : SUCCESS in 9m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0713495b99be43b2b0b576305671b8be : SUCCESS in 10m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/27a495e805a749918b1806a825d22da4 : SUCCESS in 1h 01m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bbc9b0bb12cb4f3aa21c4393299d6037 : SUCCESS in 1h 24m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1e266e0110b14c4caf29c83339040b5a : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":1},{"id":"9aec1fd3b7303542ecbd2bfeb7993f17c63ed1d1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-16 17:58:44.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d55310fed334d9cc32f1d95fd5e0aa69703adda6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 17:59:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"833e783394a7a14f03aaa679746a5854112381a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 19:48:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9687d71f98544cafb03af7895162ff28 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c19dba07ab045b5a744217b44c6e76a : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f704a911d035480bac03cab6267634e0 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e40148147de04010b2f492633a58b735 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd49f77fb44145c1af20e528339f378f : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a757814941c4234b4d67ba88dd6847e : SUCCESS in 3m 26s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/12af4d2f794c427d81e871a2edd10e87 : SUCCESS in 8m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b8efebe1b62e4f468507624bed194df1 : SUCCESS in 9m 06s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5e4f9ddbaccb438c848bb227b97db254 : SUCCESS in 10m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d61540a183224b7ba2d62ab2b526453f : SUCCESS in 8m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/829ad4e4b7a94790aef0784d2cf7c6fa : SUCCESS in 6m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7d6947e27067460ea5eb3e0ca112ac25 : SUCCESS in 4m 40s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e8b2e984595c48a6ad42350e1f54fbd4 : SUCCESS in 10m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3a6967ca8de4482e8a2ed77a795029de : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62d6a58bfbde4079a0fa5542873076e6 : SUCCESS in 7m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/747b921e569a4b8da0e76a5bc2d493ed : SUCCESS in 12m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/48df1962767f4d3ba9fb035a280b1601 : SUCCESS in 21m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1a4de05242884ba4bd7439d0e0c8f553 : SUCCESS in 33m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fd7d0dfd283f4f6a99254dcc8b01842f : SUCCESS in 24m 00s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/52f6c2a95a8542b6bf116338a3a7ff19 : SUCCESS in 14m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/84ef89dca5014674b9df26813e5f3f10 : SUCCESS in 12m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/026b775f721e4f12bdb990ba399eee31 : SUCCESS in 12m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de0f6a5ccf9f4969a499ead3763c0a1b : SUCCESS in 21m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e585b7bf15114efe8dfe89b7589cad2e : SUCCESS in 5m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0886dcb5584945259cfd080af14a1f7e : SUCCESS in 8m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e099cb608f874a8783138b9a286ef629 : SUCCESS in 8m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd08b5221d4842c3a2ca6082ecd5a79f : SUCCESS in 1h 14m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7eee5bc063e74aa098e5439ef3c6ba10 : SUCCESS in 1h 24m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3f72f0f6eb944c63931ede799d651091 : SUCCESS in 1h 09m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5e23a59f79f5096af4e2f977033e6ba16a8e665c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 19:48:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4a4e25974a1a435f59446ea715ea05ef8b1102d","revisions":{"c4a4e25974a1a435f59446ea715ea05ef8b1102d":{"kind":"REWORK","_number":1,"created":"2020-11-06 12:37:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/761737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/761737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/761737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/761737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/761737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/761737/1"}}},"commit":{"parents":[{"commit":"18ce60534c7d70c97e317e5404716d791a616aa7","subject":"update constraint for oslo.messaging to new release 10.2.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18ce60534c7d70c97e317e5404716d791a616aa7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-11-06 12:37:33.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-11-06 12:37:33.000000000","tz":120},"subject":"Bump eventlet u-c version","message":"Bump eventlet u-c version\n\nEventlet package prior to 0.25.2 has broken SSL [1]\n\n[1] https://github.com/eventlet/eventlet/issues/308\n\nChange-Id: Ib7bbdc0891640772008cdf087c6bb271cca4290c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4a4e25974a1a435f59446ea715ea05ef8b1102d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4a4e25974a1a435f59446ea715ea05ef8b1102d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
