)]}'
{"id":"openstack%2Frequirements~761793","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I94e4cc4a17a9a85ff53cbc5030d6f2c472265bc2","project":"openstack/requirements","branch":"stable/queens","topic":"bump_uc_versions","hashtags":[],"change_id":"I94e4cc4a17a9a85ff53cbc5030d6f2c472265bc2","subject":"WIP Bump outdated upper constraints versions","status":"ABANDONED","created":"2020-11-06 21:13:09.000000000","updated":"2022-08-02 09:25:09.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"af9ae68864246751c7cdc39c5fc78c991b18f3fe","_number":761793,"virtual_id_number":761793,"owner":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-14 20:05:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2021-03-07 01:23:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2020-11-06 22:35:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-07 22:37:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-11-24 17:06:40.000000000","updated_by":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"reviewer":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"state":"REVIEWER"}],"messages":[{"id":"743902fd8bc9b89ae05dbcd7d0f5fd7cb538ff83","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-11-06 21:13:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d428f32548ca1ebc695fc6f026021c1aa0b5934d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 22:35:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d1374764604950a61217151b7e01f7 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab103906e1b840a48761946af3bf82cd : SUCCESS in 8m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c21c8760ea99419a9d240dd34b175229 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27fbc37cfd804d89aeeefd92e43d41fa : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2c24025fd05542a28b68fd5fc66946f8 : SUCCESS in 4m 48s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2649cbd639a14c5d8fa077a9daf17f55 : SUCCESS in 5m 41s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/940ddf199c1a426ea2b17d5b9f110896 : FAILURE in 3m 21s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/560bd0ac54e744ff9d1b4f65d5338fec : SUCCESS in 12m 05s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/29686fc0166c4f88996b05b15e2ed9ba : SUCCESS in 8m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/af44ea22512c4327ba329d297e058d88 : SUCCESS in 4m 24s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/545b0b76b7fd4ffcb50ea412d5460f07 : SUCCESS in 6m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1e64d9d495d545fba8ba08d927daa65d : SUCCESS in 8m 37s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/71d9407d14194012aae0e3b0c0b92e9a : SUCCESS in 6m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5e3302ebe15042cf92c8cce0c886dca1 : SUCCESS in 34m 56s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/806800dff9304b7e8f1d2be1b2c195fc : SUCCESS in 15m 24s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3f6a93eb5c4e411f94e73f2612c4b122 : SUCCESS in 14m 04s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/bfda972d63f944edb9ee0be51750fce3 : SUCCESS in 23m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/10c481aac5a64df7bd59fb0f150bcce0 : SUCCESS in 17m 55s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/71e04896197f4ca1b063f5d032b15491 : SUCCESS in 12m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a04950e4e9844f55ab48097aad8611d7 : SUCCESS in 22m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/82f227025c1246de99195e3fe1cec240 : SUCCESS in 6m 02s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/6e0ba651e13f46fa98526483036e0bf6 : SUCCESS in 3m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/811714f6f1894ecea10fd44b53fc4226 : SUCCESS in 1h 02m 48s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/539a56e4a55f48c7ba0c743a3b941e9a : SUCCESS in 1h 20m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/eb07e0743126424ab6d820cf73bd6fcd : SUCCESS in 49m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a0170ed64ff2801b6bc9c58abb7851ff56a8aff0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-07 22:37:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e66b0b93ae8a4ae8783685843f0c18a7e907472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-11-24 14:40:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"07324320e960932d82857eb1800722ef0f77d324","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-11-24 17:06:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"347c2297c5d1364a51ee1338f097d1917310d170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 21:38:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afdedf9514c1422dbd2477da667cc984 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9aeae8125ca460b9c07f2dc6e39f836 : SUCCESS in 5m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/655bbc3dd5a34707b4580cb6f5e323d7 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6fa6344cc3724391a646aa7f84a232df : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d727a31870f8424da04a5168be0514eb : SUCCESS in 3m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a667fa508f9146a5843d987598751018 : FAILURE in 4m 34s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/1cb808e27a084ab1a377d65e3b41e46f : FAILURE in 5m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/79137ec566a346148fe4eacd807a8ed5 : SUCCESS in 9m 36s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/48d50f2e816842cf8268679da0bbb8b9 : SUCCESS in 11m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f7c48556907d499e8d432b72c2cf4af2 : SUCCESS in 5m 21s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/ff476574e03b4e35864c3601545430af : SUCCESS in 5m 47s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2a5c7f2c8f3c41e288326b6007837bd6 : SUCCESS in 10m 43s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/e7b7a2ff96a141b68bc39039e7144aa6 : SUCCESS in 9m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cca8f4cdbe71434f84d2397e62594b2b : FAILURE in 10m 56s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/4cabaf0d625644f791932ff99d701d99 : FAILURE in 13m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a181f4c631d84163a613e2cbe18585e0 : FAILURE in 24m 35s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/cea2519ba90f4926a30a8622ace49d0c : FAILURE in 28m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f0ca935ab36e4c5099740db830d33bbe : SUCCESS in 13m 04s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/3cd459d4e11b4baabc5a68283c6fca88 : SUCCESS in 15m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4154a190874f41ee9d9d6f886ec492f6 : SUCCESS in 20m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4ff7931939c5429498bd9d9b0887962e : SUCCESS in 6m 32s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/c6ad53304f6b471fb737d5992a6c4f55 : SUCCESS in 4m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/48f08183e29441bf9997218aad734f72 : FAILURE in 54m 37s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/03bb77eeb9cc4fc4bd94401497ed67c5 : FAILURE in 15m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fa17d65b38b64a4dba9e980fab19ce86 : FAILURE in 50m 24s","accounts_in_message":[],"_revision_number":3},{"id":"a984bf95a0361dfe9e7fa73d563b7736b11b6606","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-11-25 14:08:07.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"d17fd9890268ad747e330b916d88c9f90b169778","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-11-26 21:59:29.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a9fce40e35fdc8c235382493f7e37ed80a0d7322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 23:26:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fac192249f94bfc8ba2498a26b583d7 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ba1604cb6c749f090187f136f916c99 : SUCCESS in 4m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/650e9a4857bc4526b3f6f37712b22bf1 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9289f353fc2740c0880bc25bb5d40e89 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c49ff2e755da4765afd6f9637e8044b4 : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/52453b4af348475eb22943f961ce0c78 : FAILURE in 3m 58s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/371928bf5ee64ceb9115fd60501294c5 : FAILURE in 3m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7dce12c9c8bc49ac9ad0e7607f4f13c9 : SUCCESS in 9m 04s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/05516542e6eb4daeaf86d3d5ebddab9c : SUCCESS in 8m 20s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2237b9a4212743daaba10102fef6e5b3 : SUCCESS in 4m 36s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/85992c923ee640ba96cd52f638afb8d2 : SUCCESS in 5m 36s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2cd08cc3fa93442e91fca1503c9fb258 : SUCCESS in 9m 04s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1bf30d3a60f142748ff296d8ec0036d5 : SUCCESS in 9m 11s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0f4fde461c484783b3e9fbce74d5f63f : SUCCESS in 12m 47s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/574da79874514e8da08217264f15eb61 : SUCCESS in 12m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fe6691d97ee347f184acb7852a3d52c1 : SUCCESS in 22m 00s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/f11dfe77e57b4b72baad2e5750b6def6 : SUCCESS in 25m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/55994c5800724db4abcf1c2bc708621c : SUCCESS in 13m 25s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/75def15bdf07427ea425b7445ece7b5e : SUCCESS in 11m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7e4c978ff618434cba6e343cb85b5056 : SUCCESS in 20m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/739eb69c071945768c99cb9e1553a06c : SUCCESS in 5m 58s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/560305f3820d479ca8dc5c992040d8f3 : SUCCESS in 3m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/efb6e6a933734667865fd69f8b8806fc : SUCCESS in 1h 03m 13s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/039bff3d0a6342208bfd909427f1eba4 : SUCCESS in 1h 20m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ae6f30e4bfc44568aa48f43db1577e63 : SUCCESS in 53m 01s","accounts_in_message":[],"_revision_number":4},{"id":"835291d00a42c46a1f0153a74c71c96e3cd05474","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-11-27 13:51:38.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"e59aea1d5015ec36aa889a1cdcf8f8682ea71fa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 15:10:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca1e6c1b506745a291163e08ab566c3e : SUCCESS in 9m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/606ce1afce594f3bb22e830853ca9a70 : SUCCESS in 8m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/67740efed300469fa842eb9f56039ab5 : SUCCESS in 9m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5bdd3075904243188590bc0047762729 : SUCCESS in 9m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a0830f212804f8693edd7cccda263f8 : SUCCESS in 8m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f3819a7dbd3c47d1a7dde60a5ffaeb4f : FAILURE in 7m 19s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/23771c9fb4a74e0b8911a6b392bcb554 : FAILURE in 8m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d02ca5ca0f2a4f0792f0205a8ced5ad9 : SUCCESS in 11m 20s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/da7ac76c3a0f43b3990c90df2dc6b099 : FAILURE in 5m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0cbe5f12b2404832aff7d243cd421e17 : SUCCESS in 4m 35s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b36f63979e4c47a1959b100f9abadd48 : FAILURE in 3m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/04e5836933814ea5abfa065aa048c0f9 : SUCCESS in 9m 57s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/8dbc494ef93e45859a22a88d4cc7617a : SUCCESS in 9m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3243e5e44c5d4b178d3535a5bc19b6fe : FAILURE in 12m 24s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/26ee8b2a6a9949e3be645d0b546b8db6 : FAILURE in 11m 57s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a6bcb8a11b0b419986fafa81f3d5cc33 : SUCCESS in 21m 44s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/7f340f457fef4abeaca96d8642781fa5 : FAILURE in 4m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9257f3a271ef4be59fa454d8279f04c6 : FAILURE in 15m 02s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/7203d2348c7d432da595d5bdae4583c3 : FAILURE in 12m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ae108a61923f430694d2b41f9ddfe180 : SUCCESS in 20m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/328759bc21674de8b11032113138e5db : SUCCESS in 6m 12s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/0f6e85b5ceb54e6196f1f62e35587035 : SUCCESS in 3m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/602983707c834ba882885f211d4a04ef : FAILURE in 1h 00m 23s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/e8d78a06b21247b488e130647d9cb462 : FAILURE in 16m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/95e459c3fb13467f8a9f0d5b69ae2bfc : FAILURE in 40m 23s","accounts_in_message":[],"_revision_number":5},{"id":"301e1aaad846eac68808605bc403e498c52312d7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-11-27 19:54:26.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"7ae18e0b713d1d0c316a199ab2d029090b97e985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 21:18:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e6e9c46e24048c284f25f1eb1055297 : SUCCESS in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7c00006a4514547955beff11aa88d52 : SUCCESS in 5m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9e5059f57e034920b9b33ede17a6ab1a : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9cf1ea481f164e3ca9f04c6fadd7ea8a : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ccc04eb8cb0e48128c016c5af68a89bb : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/07c0d8335e3c49fcaf0777089fefbc3b : FAILURE in 3m 56s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/607b8d6fe9c54ee1a78fd9f989b044d8 : FAILURE in 4m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a2b474d771064141b4be762f0598194b : SUCCESS in 8m 08s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b15f8d7e4f31440c911c830a0e340b98 : SUCCESS in 10m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/101d0bb770014b63b12cd866c0894254 : SUCCESS in 6m 02s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/021a924ce101463492bbd256aeec6fa5 : SUCCESS in 5m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fbf0805eff544f3d887ca8920948be61 : SUCCESS in 8m 19s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a9038899e5eb4617b7ab6d3e87bf46a2 : SUCCESS in 7m 53s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/92bfabe88b7545018432f9c27e5faa95 : SUCCESS in 12m 43s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/48bfebb93f7946969456237c6e6938b9 : SUCCESS in 12m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8d430b20fcf94f3292f1e8e3ea51bbd5 : SUCCESS in 20m 34s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/20f10fc7da164d5c84173cd2b392fffa : SUCCESS in 24m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/15a1535de89a42618d9c833cd48b8c1d : SUCCESS in 15m 01s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/76bc7ec323934d348046532e860089ca : SUCCESS in 14m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88b0308ce4ef453aba8953e9a429de89 : SUCCESS in 20m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bc77ce517cc649abb44b71e33e1c1733 : SUCCESS in 5m 59s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/213b65df760f4750893420d4822a70d6 : SUCCESS in 3m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3bfe546fd5264bc1b77695b4e0f00ce3 : SUCCESS in 54m 28s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/8446df42157f41b88e4d724a862047f7 : SUCCESS in 1h 17m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e0f5a9448b2d45f6a8b77b30f8b852c3 : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":6},{"id":"58967f412e5f9c677fda7768d21b15a03e8d8ebc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-12-14 17:39:37.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"673a7d64dfce34a613414554497f2baf0328e7b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 20:05:43.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/488ace86dc8f44cdb8c69062a37b6de3 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5bedac22515e44fdab55ddedc19c8127 : SUCCESS in 4m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d70d4b1b317348d3a1e80acb450a63a4 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffb138ba06b34d109c4f509f855d9503 : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f13c9114af1b432db38c78f1adeaa997 : SUCCESS in 4m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bf3301c1caa14f4aa4fd04dd42e2cb8a : FAILURE in 8m 39s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e7bd5b91a9864f149feb6755b7707e6d : FAILURE in 7m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/81031b965b0c4873934608fd50190277 : FAILURE in 5m 24s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/585f42efa78c4f25b31daf31b8100018 : SUCCESS in 10m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5f77cc23596948ee997279eb0f83c64c : SUCCESS in 6m 16s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/18b2fd233b4c4feea93524b9a304bb9e : SUCCESS in 4m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/70427709e7b94a6f859149ad00a3e5bf : SUCCESS in 10m 36s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/892986a98cb64db8aca5a07182e32537 : SUCCESS in 8m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/44ede2fcb248457293d5e9deee8b9a36 : SUCCESS in 13m 33s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/8f849e039c1e4ab0a38280b9adcfc940 : SUCCESS in 13m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/efed4464223546ae8311b9652b34b1e3 : SUCCESS in 23m 24s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/037e1569c783472c9e4c84b7b0d6ba96 : SUCCESS in 28m 00s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b686dd786015447ba6ed1437e7134f29 : FAILURE in 4m 15s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/6f59c3b2bae4434997de30f720a7b528 : SUCCESS in 13m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7730d30df114479dbff0be1401d6f732 : FAILURE in 4m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/956268cff0cc4ce59bfc0cb728cf6d97 : FAILURE in 3m 37s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b019cea7b2ee43ef8af3283810a93f42 : SUCCESS in 4m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aca7391f554948d3a894dca7e57b3532 : SUCCESS in 1h 10m 37s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/ef3f89a20c64443ea09f9a1885042799 : SUCCESS in 1h 24m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/178dbb876b914f108ff8e118c69f876c : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":7},{"id":"1eb0db22b919ff110b7e7ab0ea38258e34adc211","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-07 01:23:24.000000000","message":"Patch Set 7: Code-Review-1\n\nis this still needed?","accounts_in_message":[],"_revision_number":7},{"id":"af9ae68864246751c7cdc39c5fc78c991b18f3fe","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-02 09:25:09.000000000","message":"Abandoned\n\nThis is a very old WIP change.  Please restore and update if still needed","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"aca9dd9dc2e45364129d79d512283303c4b108d4","revisions":{"15c1371e4a5230f3c5bb835da9e74eb14e8e0f60":{"kind":"REWORK","_number":1,"created":"2020-11-06 21:13:09.000000000","uploader":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"ref":"refs/changes/93/761793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/761793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/761793/1"}}},"commit":{"parents":[{"commit":"5cdb94c26bf26b085302517f409eac1d448f709f","subject":"Revert \"[stable-only] Upper constrain wheel to 0.34.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cdb94c26bf26b085302517f409eac1d448f709f"}]}],"author":{"name":"Lucio Seki","email":"luciomitsuru.seki@fit-tecnologia.org.br","date":"2020-11-06 21:06:25.000000000","tz":-180},"committer":{"name":"Lucio Seki","email":"luciomitsuru.seki@fit-tecnologia.org.br","date":"2020-11-06 21:13:02.000000000","tz":-180},"subject":"WIP Bump outdated upper constraints versions","message":"WIP Bump outdated upper constraints versions\n\nsetuptools 46 broke old MarkupSafe and pyScss installation.\nBumping their versions to address this.\nAlso, trollius is deprecated, so had to explicitly append \u0027post2\u0027 to\nits name.\n\nChange-Id: I94e4cc4a17a9a85ff53cbc5030d6f2c472265bc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15c1371e4a5230f3c5bb835da9e74eb14e8e0f60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15c1371e4a5230f3c5bb835da9e74eb14e8e0f60"}]},"branch":"refs/heads/stable/queens"},"b0157eaf1f3dcc5368551178a3a447fc2ff66736":{"kind":"REWORK","_number":2,"created":"2020-11-24 14:40:05.000000000","uploader":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"ref":"refs/changes/93/761793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/761793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/761793/2"}}},"commit":{"parents":[{"commit":"5cdb94c26bf26b085302517f409eac1d448f709f","subject":"Revert \"[stable-only] Upper constrain wheel to 0.34.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cdb94c26bf26b085302517f409eac1d448f709f"}]}],"author":{"name":"Lucio Seki","email":"luciomitsuru.seki@fit-tecnologia.org.br","date":"2020-11-06 21:06:25.000000000","tz":-180},"committer":{"name":"Lucio Seki","email":"luciomitsuru.seki@fit-tecnologia.org.br","date":"2020-11-24 14:36:02.000000000","tz":-180},"subject":"WIP Bump outdated upper constraints versions","message":"WIP Bump outdated upper constraints versions\n\nsetuptools 46 broke old MarkupSafe and pyScss installation.\nBumping their versions to address this.\nAlso, trollius is deprecated, so had to explicitly append \u0027post2\u0027 to\nits name.\n\nChange-Id: I94e4cc4a17a9a85ff53cbc5030d6f2c472265bc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0157eaf1f3dcc5368551178a3a447fc2ff66736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0157eaf1f3dcc5368551178a3a447fc2ff66736"}]},"branch":"refs/heads/stable/queens"},"cbf9bf373b37f19fcd525da8c5e1689318f07a69":{"kind":"REWORK","_number":3,"created":"2020-11-24 17:06:40.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/93/761793/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/761793/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/761793/3"}}},"commit":{"parents":[{"commit":"5cdb94c26bf26b085302517f409eac1d448f709f","subject":"Revert \"[stable-only] Upper constrain wheel to 0.34.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cdb94c26bf26b085302517f409eac1d448f709f"}]}],"author":{"name":"Lucio Seki","email":"luciomitsuru.seki@fit-tecnologia.org.br","date":"2020-11-06 21:06:25.000000000","tz":-180},"committer":{"name":"debeltrami","email":"debeltrami@gmail.com","date":"2020-11-24 17:06:13.000000000","tz":0},"subject":"WIP Bump outdated upper constraints versions","message":"WIP Bump outdated upper constraints versions\n\nsetuptools 46 broke old MarkupSafe and pyScss installation.\nBumping their versions to address this.\nAlso, trollius is deprecated, so had to explicitly append \u0027post2\u0027 to\nits name.\n\nChange-Id: I94e4cc4a17a9a85ff53cbc5030d6f2c472265bc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cbf9bf373b37f19fcd525da8c5e1689318f07a69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cbf9bf373b37f19fcd525da8c5e1689318f07a69"}]},"branch":"refs/heads/stable/queens"},"bc5105a6bd74a6fc11e0d51505350f3125f2e947":{"kind":"REWORK","_number":4,"created":"2020-11-26 21:59:29.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/93/761793/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/761793/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/761793/4"}}},"commit":{"parents":[{"commit":"5cdb94c26bf26b085302517f409eac1d448f709f","subject":"Revert \"[stable-only] Upper constrain wheel to 0.34.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cdb94c26bf26b085302517f409eac1d448f709f"}]}],"author":{"name":"Lucio Seki","email":"luciomitsuru.seki@fit-tecnologia.org.br","date":"2020-11-06 21:06:25.000000000","tz":-180},"committer":{"name":"Andre Luiz Beltrami Rocha","email":"debeltrami@gmail.com","date":"2020-11-26 21:58:47.000000000","tz":0},"subject":"WIP Bump outdated upper constraints versions","message":"WIP Bump outdated upper constraints versions\n\nsetuptools 46 broke old MarkupSafe and pyScss installation.\nBumping their versions to address this.\nAlso, trollius is deprecated, so had to explicitly append \u0027post2\u0027 to\nits name.\n\nChange-Id: I94e4cc4a17a9a85ff53cbc5030d6f2c472265bc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc5105a6bd74a6fc11e0d51505350f3125f2e947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc5105a6bd74a6fc11e0d51505350f3125f2e947"}]},"branch":"refs/heads/stable/queens"},"e1ec563b0ca1cb01c5d22ef1ae12850c78392daf":{"kind":"REWORK","_number":5,"created":"2020-11-27 13:51:38.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/93/761793/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/761793/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/761793/5"}}},"commit":{"parents":[{"commit":"5cdb94c26bf26b085302517f409eac1d448f709f","subject":"Revert \"[stable-only] Upper constrain wheel to 0.34.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cdb94c26bf26b085302517f409eac1d448f709f"}]}],"author":{"name":"Lucio Seki","email":"luciomitsuru.seki@fit-tecnologia.org.br","date":"2020-11-06 21:06:25.000000000","tz":-180},"committer":{"name":"Andre Luiz Beltrami Rocha","email":"debeltrami@gmail.com","date":"2020-11-27 13:51:26.000000000","tz":0},"subject":"WIP Bump outdated upper constraints versions","message":"WIP Bump outdated upper constraints versions\n\nsetuptools 46 broke old MarkupSafe and pyScss installation.\nBumping their versions to address this.\nAlso, trollius is deprecated, so had to explicitly append \u0027post2\u0027 to\nits name.\n\nChange-Id: I94e4cc4a17a9a85ff53cbc5030d6f2c472265bc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1ec563b0ca1cb01c5d22ef1ae12850c78392daf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1ec563b0ca1cb01c5d22ef1ae12850c78392daf"}]},"branch":"refs/heads/stable/queens"},"0160fa1e23971379b2c826e62315e50547b9b494":{"kind":"REWORK","_number":6,"created":"2020-11-27 19:54:26.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/93/761793/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/761793/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/761793/6"}}},"commit":{"parents":[{"commit":"5cdb94c26bf26b085302517f409eac1d448f709f","subject":"Revert \"[stable-only] Upper constrain wheel to 0.34.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cdb94c26bf26b085302517f409eac1d448f709f"}]}],"author":{"name":"Lucio Seki","email":"luciomitsuru.seki@fit-tecnologia.org.br","date":"2020-11-06 21:06:25.000000000","tz":-180},"committer":{"name":"Andre Luiz Beltrami Rocha","email":"debeltrami@gmail.com","date":"2020-11-27 19:54:15.000000000","tz":0},"subject":"WIP Bump outdated upper constraints versions","message":"WIP Bump outdated upper constraints versions\n\nsetuptools 46 broke old MarkupSafe and pyScss installation.\nBumping their versions to address this.\nAlso, trollius is deprecated, so had to explicitly append \u0027post2\u0027 to\nits name.\n\nChange-Id: I94e4cc4a17a9a85ff53cbc5030d6f2c472265bc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0160fa1e23971379b2c826e62315e50547b9b494"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0160fa1e23971379b2c826e62315e50547b9b494"}]},"branch":"refs/heads/stable/queens"},"aca9dd9dc2e45364129d79d512283303c4b108d4":{"kind":"REWORK","_number":7,"created":"2020-12-14 17:39:37.000000000","uploader":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"ref":"refs/changes/93/761793/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/761793/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/761793/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/761793/7"}}},"commit":{"parents":[{"commit":"5cdb94c26bf26b085302517f409eac1d448f709f","subject":"Revert \"[stable-only] Upper constrain wheel to 0.34.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cdb94c26bf26b085302517f409eac1d448f709f"}]}],"author":{"name":"Lucio Seki","email":"luciomitsuru.seki@fit-tecnologia.org.br","date":"2020-11-06 21:06:25.000000000","tz":-180},"committer":{"name":"Andre Luiz Beltrami Rocha","email":"debeltrami@gmail.com","date":"2020-12-14 17:39:26.000000000","tz":0},"subject":"WIP Bump outdated upper constraints versions","message":"WIP Bump outdated upper constraints versions\n\nsetuptools 46 broke old MarkupSafe and pyScss installation.\nBumping their versions to address this.\nAlso, trollius is deprecated, so had to explicitly append \u0027post2\u0027 to\nits name.\n\nChange-Id: I94e4cc4a17a9a85ff53cbc5030d6f2c472265bc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aca9dd9dc2e45364129d79d512283303c4b108d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aca9dd9dc2e45364129d79d512283303c4b108d4"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
