)]}'
{"id":"openstack%2Frequirements~761860","triplet_id":"openstack%2Frequirements~master~I5f665c4fc1c63b01046c57528adc3c64a333e0c1","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I5f665c4fc1c63b01046c57528adc3c64a333e0c1","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-11-09 06:19:40.000000000","updated":"2020-11-09 10:33:20.000000000","submitted":"2020-11-09 10:31:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761860-1604917911074-96dc3111","meta_rev_id":"a8935b7751952cd03f2e83910c43d9137882519e","_number":761860,"virtual_id_number":761860,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-09 10:31:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-11-09 06:35:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-11-09 08:52:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-11-09 06:35:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-09 06:35:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-11-09 08:52:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-09 10:31:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb6f1266f73e39ae4a6b237eae2ff9e6298a35ae","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-09 06:19:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12b0ab1ea61e2f84ea47f1aa7a634f7ff0681f9a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-09 06:35:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eb721b891595cbc70e3369c3c9a36f0ab47da600","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-09 06:35:27.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"e9420f6daf00676f2a117ad3913d6149d9cc1b17","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-09 06:35:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4ea2fba117a9529dccc56da1ec87f3c3ace91e83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 06:59:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/45811799e07149eca107c5cfaeae52f0 : SUCCESS in 18m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a6f2830a06f97f1ba2a694862a2745291715a396","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-09 08:52:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"754352440b8f2e917b5f891e5765d2ed09105486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 08:59:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c65588422bd4ebdb30f9f8f575e0442 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2f99cd87de24d478d0ff4ce8269e51e : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d84bef08e7d4f17b12c84b7aa3d11d1 : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4267c2dd6094f3e8e383a1ce5c49653 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c2e200f3b3c47b08c2575886d6f621e : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eaf72b67f2f647e188ed2a2e6b892bba : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af35d4de2242480b89a2747c9558b039 : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6493812a09964fcc83b53fc008f0e67b : SUCCESS in 7m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fcb931b0ead84bc38eb02f0f1e444a29 : SUCCESS in 18m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/19176f8c8c2b4b57b8b3ba6ca059a95e : SUCCESS in 12m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/99c10f4d2e5c41409778cc51f3b930ca : SUCCESS in 8m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1741b066277344f6a174679b2e7b4f1c : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf371c11c07746b080522aa048f94146 : SUCCESS in 5m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b45dfb1014634d7ab4a6e98b3fc99b1c : SUCCESS in 6m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c593313bb10644b3971f8dbb4a8b8776 : SUCCESS in 12m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e92b4b1077a54824a945bc261dbdfd40 : SUCCESS in 33m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/766b225612f144959b5e905c4681a1d6 : SUCCESS in 15m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8c2cd688e5e546ad9861251df5c8fe5c : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23f25ba83f164c56a55f7578cd861413 : SUCCESS in 18m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c90f3ae406794d39864537852e5f9cad : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3fd4f38d39164b79be6ab31d65e5c2aa : SUCCESS in 8m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/866c7b222a7d4f56a4222e03b6870b95 : SUCCESS in 11m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd58d5e8a9b6418f946191a2e651f434 : SUCCESS in 1h 05m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9243ebfedde14ed79d85732709cfc646 : SUCCESS in 1h 26m 36s","accounts_in_message":[],"_revision_number":2},{"id":"e3609541d3a0ddd6f90be2241d87df8994c02016","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 09:00:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"38b7edaec1d479b8b72964cf54575266d2998ed8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 10:31:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569382e25e8f4b3da3166cf63bedcb17 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db58ba5317e842a18bfdbfb30c69d81c : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f864045f6fa4f3e90ca5cde5d7a345a : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/076a6ebc557846f88ff0d48347dd71ec : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/179b58eee21745d2a47cfdea6b046ec8 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dbefdee91320428caceab425e486ed8c : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/144f0c65776d4405b64c5f5b97011c19 : SUCCESS in 9m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/10a54045ab4d42538abb362c401faefe : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aae9c9953e1e47b282561e556eec5ebe : SUCCESS in 9m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3548219ab12f4ef1a8aaaa3039fd1829 : SUCCESS in 5m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aee17c7d3296452f92ddf3c19e4b6666 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9e6c23b9a12a4de3912824d435398cdc : SUCCESS in 6m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/04d0be7c6e35409db311314433f841ca : SUCCESS in 6m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/06c670e8ad8a4d6190bfd8925eda9a5c : SUCCESS in 12m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c2511e9494494b3196bc89640c70a7ae : SUCCESS in 27m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/536638441469475f9f57d63d727e8596 : SUCCESS in 11m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4f46cfa07cf943fc847a03c85bbba999 : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/725e36f05bcc42a4bbcb628e3c8f87b3 : SUCCESS in 15m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5407f33fe8574c3db0c34837d2d5c87a : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/032aaba157ad4b0e9800b042f051b127 : SUCCESS in 6m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/56e0f404614e489e895b295a7476e1eb : SUCCESS in 9m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c54edc6cbb04ecbbcd2015a5fd28e49 : SUCCESS in 51m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7a0a89a7f82406aa5cc679793677f93 : SUCCESS in 1h 24m 40s","accounts_in_message":[],"_revision_number":2},{"id":"26185476a387c26dfc1adc31b3ae6a3be23ded2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 10:31:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a8935b7751952cd03f2e83910c43d9137882519e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 10:33:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2caf3b192ab7405a95454019735e4bf2 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"96a331ef9f8de55ca49ce21ed096afacf1315035","revisions":{"3b51311a7581defac61453a7ea9ab7546a4b3b98":{"kind":"REWORK","_number":1,"created":"2020-11-09 06:19:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/761860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/761860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/761860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/761860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/761860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/761860/1"}}},"commit":{"parents":[{"commit":"a59755ec7583b71e2704124d432a25ce7d8ba278","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a59755ec7583b71e2704124d432a25ce7d8ba278"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-09 06:19:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-09 06:19:38.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I5f665c4fc1c63b01046c57528adc3c64a333e0c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b51311a7581defac61453a7ea9ab7546a4b3b98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b51311a7581defac61453a7ea9ab7546a4b3b98"}]},"branch":"refs/heads/master"},"96a331ef9f8de55ca49ce21ed096afacf1315035":{"kind":"REWORK","_number":2,"created":"2020-11-09 06:35:23.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/60/761860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/761860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/761860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/761860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/761860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/761860/2"}}},"commit":{"parents":[{"commit":"a59755ec7583b71e2704124d432a25ce7d8ba278","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a59755ec7583b71e2704124d432a25ce7d8ba278"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-09 06:19:38.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-11-09 06:34:28.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I5f665c4fc1c63b01046c57528adc3c64a333e0c1\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96a331ef9f8de55ca49ce21ed096afacf1315035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96a331ef9f8de55ca49ce21ed096afacf1315035"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
