)]}'
{"id":"openstack%2Frequirements~762296","triplet_id":"openstack%2Frequirements~master~I27fda78e9f16794e7c89e4c59c6f0667f01ed7ef","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I27fda78e9f16794e7c89e4c59c6f0667f01ed7ef","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-11-11 06:52:17.000000000","updated":"2020-11-20 09:09:21.000000000","submitted":"2020-11-20 09:07:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762296-1605863277902-11ac1cc1","meta_rev_id":"707530f237bd26d09ccefea4451c59157f80f33e","_number":762296,"virtual_id_number":762296,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-20 09:07:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-19 22:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-11-20 07:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-11-19 22:05:46.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-11-20 07:32:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 22:05:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-19 22:05:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-11-20 07:32:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-20 09:07:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed6e94ee329cd0b9545f0bd9dd940707a200a695","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-11 06:52:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2012eb792aaf5ba0dbe1605690e1bc2747de63f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 07:06:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2eae5e4acf264e9193261e1a6f722c13 : FAILURE in 8m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ea8ec6ceb3877bc2af0a0329da5d7d4d17d6dcb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 13:44:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c1686c7a4b245709080762a775e6d95 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c736eff237034e4e8fbbda87bc052b89 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ece5a759c6421b8369ed6080d9f7ea : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03a55cf2b48b454caf0f7ee9783f50a3 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dfa67b0a8c2426c858d068df2385783 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe6d36392b514f08858ad961aebb2372 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/58b95af61a9145f6b1a5dc1bb09fa0cf : SUCCESS in 3m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1ca58a514b9144818730c55b5c9fd252 : FAILURE in 4m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f3007ef4d6854917a2f5e05f815d8b85 : FAILURE in 6m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0622db31ef7d4e4493e1caeb0e10e14b : SUCCESS in 8m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/85ce41fec59543058e55e67dd440f4b8 : SUCCESS in 4m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5a5c1142eace4330b10dd51d96de7f22 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a446647c0d824ee3927b43fbc45057a7 : SUCCESS in 6m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ae734564c1fb4aa58a1071f82e0e9512 : SUCCESS in 6m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/796e9a019b35423e838f8a7830541fa4 : SUCCESS in 13m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c8dcc6682ac24933ac6077f77f03d6c4 : SUCCESS in 32m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1d662a203f284d859dc5b894d15faf37 : FAILURE in 11m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cf2b4ac105864ff5907cbd3972268ff6 : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b848dca35ae5409ab5e8e34166d10140 : SUCCESS in 16m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/162557e046bd4fd7839656b2f4854048 : SUCCESS in 4m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b49a43a2d53c4615b19a99fddebf5843 : SUCCESS in 7m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/636b48fdbba94f679cd12c06a44b2432 : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3406f0819933423cbe4ab4b31340f8b4 : SUCCESS in 53m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f72a4e4e6ee47898bde596d65b8e336 : FAILURE in 1h 03m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0bcc293a443f4392fdebb3b7bd0dc25c86d51f01","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-12 06:22:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"86a2cb131f02b07016b7c933de14fa5ff5d10767","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 06:44:25.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/051833693f0c4cbdb6749927b771cdbf : FAILURE in 13m 43s","accounts_in_message":[],"_revision_number":2},{"id":"a995b1728be81857a1f0880a4d54e5f79ed9ee2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 08:16:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b08198315417462889c1ac1e6cdb73fd : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42ebbda242c54b1b9e982d2c85d3190f : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c60c3ec60d0451695e4be7205a8f9a7 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/becca48983e3401e9d2b218f5e984197 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a9e27e8c88e471087a7c02ad5a18b88 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36482479df4c47a5b1f335d9536171b5 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fa44543ade444a5fb78699d50553b7ef : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/70a27dc3dbbd4e91b5ebe467d1241d6a : FAILURE in 6m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0dd568eb352b4dcbb10a2760086b5abb : FAILURE in 4m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/67807cc0c3a54b52a63e5e83b9b83182 : SUCCESS in 8m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/341289bcc6d74967bf4fc42370894341 : SUCCESS in 5m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f21af5e277be49f1a54eda1de5008aa6 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2c9423c0c44f44789248e10182ffe876 : SUCCESS in 7m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ca743dec4dc24ba986a2099df88dfa58 : SUCCESS in 8m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b38b7068b01d463782f89438e1c30318 : SUCCESS in 15m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/268ec0883f004a5a8c8ef18463ec6c7d : SUCCESS in 16m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9c938500cf3d460db66d542aefd62428 : FAILURE in 12m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3018038490b54d5589dc9c94ac5bffc6 : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a3c0a0b1a1444a095b55035bbf4edb3 : SUCCESS in 17m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d3ba12de83804a829ecba0d4d67c9805 : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/14d660dfe8864bc88fc905c363afe62e : SUCCESS in 7m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4440d6dbb8744430963de9256d990865 : SUCCESS in 11m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1a2b9daa0364b71a2111849b188126b : SUCCESS in 1h 03m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1d7af06f336043598664a18ac7665e20 : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b5e5f3f12152549cb673860f1a3ea4f3121e635d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-13 06:25:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7e0d2188f4067b70516c4599fbe91ab27a62d82d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 06:49:14.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5f811fa2927743a198be1fbb3cde0a0b : FAILURE in 15m 15s","accounts_in_message":[],"_revision_number":3},{"id":"be3ea176a3af7268881ad641714abf8100d47d46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 08:49:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb538bcc1c24a60b71f916b387f378e : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed35ea205c3a439cb4742df158d927f4 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86fc4a185f0c413eb80bb1dd95a329c2 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/444d0bba9cbd450483cd590f036fe49a : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3030ddfacff34831a5acfcde3132daa2 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac112777b3584b6d909b1155da285a00 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7eed094d072a4a89aca2c7a9719a08da : SUCCESS in 3m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aac24d3362a645a390ffb768d046f9ea : FAILURE in 5m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5e822a6ce7ba4bc09db873530bb29dd7 : FAILURE in 4m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b86eebda1493439aa9af26d4fec50b82 : SUCCESS in 8m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6893a787405a4297bb02d497c7e2de97 : SUCCESS in 5m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/794176a42411422b80263e168aa68bbd : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71d1e0b627704940945d168b82b2941c : SUCCESS in 6m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b4b9ad0c081943e898a65c603a54eef0 : SUCCESS in 6m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f97d54d6af76421f80e0e4d39817653c : SUCCESS in 13m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d6318745d5f84f64a405a5f422301f7b : SUCCESS in 24m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9c2981405b8a44e4940abb1880df5860 : FAILURE in 11m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/55b8d79d1c7d4657a6848309db2091c5 : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2065b2f0d3204b8b8d53e6050684c9bd : SUCCESS in 19m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6a6064dfbe724ec2b2826c9562518ca5 : SUCCESS in 8m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c3a30dac8d645bf920282fd85d84afa : SUCCESS in 12m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9570e5941d774ad095aa73d037e00b74 : SUCCESS in 9m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f3769083b4a4d52a9e260696005baa1 : SUCCESS in 56m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8c5ab13b1764682bc3a1b6551c6650a : SUCCESS in 1h 32m 16s","accounts_in_message":[],"_revision_number":3},{"id":"8d3100653e9ec005bf0470577c252697a4101ecf","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-14 06:21:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8d897c691ce7f28a6ee5968af467f2189518dd6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-14 06:41:16.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2b0bc5768f2c45269011e76ff0bdfadb : FAILURE in 11m 57s","accounts_in_message":[],"_revision_number":4},{"id":"76b7432eade3b3b3a88a4285ac68bcf60ca9df37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-14 07:58:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db5f1ea727a74f1984c9c483557d910a : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a90d0b88e5a94c4a84c4c93ae7b9f278 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50b00cc62b42431c95a85420e82e5c36 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22a85d938d324edfa45765b811747924 : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/870db80bf7d7413ba1037c8e26095221 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ddd4125be414b55a5bd05812ed81271 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3b698bd05df4e14a4903c60c428baa0 : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f6a359126c574b3a93ca3a7d649b0274 : FAILURE in 6m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e4d56db9bab34556ae6972a5ae892a00 : FAILURE in 6m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ac4d307f760e4a638d33ad3162fdfbbc : SUCCESS in 17m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9838112feb2046fb8c79a6780730a8e0 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8afa318f58964716a521f7a0a4f8fab8 : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27838f49d5744dd4afb1ea834b01c9fe : SUCCESS in 5m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/16cba711933a446ab97386ee56bde7b2 : SUCCESS in 6m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eab65dda6f5e44a08c995c78a926df65 : SUCCESS in 16m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/460dc522186c44a0b7e944ae90ec22d8 : SUCCESS in 29m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1bb581924ac64a52826599e34f8551de : FAILURE in 16m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0dc1da845a9b48efaf983bc15fc635c3 : SUCCESS in 4m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67464e7619764a99a00051b493f66354 : SUCCESS in 19m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/23e69a6532b547c887b448905033d196 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b788b0125928484cb864a1d8ce774a02 : SUCCESS in 7m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ebfd22d3808a4d178f42563075447c23 : SUCCESS in 9m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dddbf69c0d964f00aa48ebfc8bf35fdd : SUCCESS in 39m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc8dfba76c124666a65ec0aed4f06b65 : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":4},{"id":"ea388e6dce59845fedd11f8664cefb2f23f51f63","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-15 06:39:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ee2352b8ad3cf017141c377c0fdc8d7e283e07ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-15 06:54:36.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/74112ab91648404a89e61352daa9e5ff : FAILURE in 8m 28s","accounts_in_message":[],"_revision_number":5},{"id":"4beb7161981b03484ee42cb8b48a862c031a0331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-15 07:57:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d49ec5cb6040069758e545e95bbaf2 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/467c5322fb574decb331b5187de12b3a : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/815572ba68574a75b04f5d80d8a97a5c : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e06055fe16a74bcaa07ff1717bdf5420 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96198674d6ce4531a6f47e9eaab67a53 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d9174ae5d2840ec886d9583e16d3b7e : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/975b4723e1ce43e080fd625cf078a380 : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c176e19f02b545a4b076522dad7cdb64 : FAILURE in 4m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4b4c60f2a2a64beb9ee38bc3e73f3c50 : FAILURE in 4m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8a236dd838b84fc09b6da69aedbb3baa : SUCCESS in 10m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9012e2be61e742fd8dcb25d58ee8573e : SUCCESS in 4m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f36afb3fc1ea4846bf9558e1adfdf72a : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e38b5e4cf8f845a99476a105d3f3e23a : SUCCESS in 5m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7ee96736bf99447484813a8082e73a65 : SUCCESS in 6m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2508dfe9957e435aaffda979ffd85752 : SUCCESS in 12m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8ca2ec42a3a34fb49877aff44b947c39 : SUCCESS in 26m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/77d2ee59fa5a4df0a845ee488524bdcc : FAILURE in 13m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a665b0cb32204583a3bacda6528d7b3e : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9a2f1ed583c4a898cf5273fcf77e7d1 : FAILURE in 24m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/96c15cb0e2ed45abae572338036d4246 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6fbfeb0621944e178d3a1863ae3bc70d : SUCCESS in 6m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d02d4d142b1744c28e00c46e4efc3d38 : SUCCESS in 10m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30277a09a8904daca3e1412903bdc1fd : SUCCESS in 53m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e5c49aa866c40afbd09187c5f560f19 : SUCCESS in 1h 14m 56s","accounts_in_message":[],"_revision_number":5},{"id":"1a6b8200945c6b94e1415c3f21fe2395089012f4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-17 06:23:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e0828bd52c6bd974b4040666e8e92d65892e433d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 06:39:20.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0ff1910798fc4fa9884df861fcfe368c : FAILURE in 11m 23s","accounts_in_message":[],"_revision_number":6},{"id":"8eee0b35b470a84b5087fd826ee5870f4babe473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 08:31:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1594ba24a9264f0f9bce41e2787c7e05 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad96e0e8ac9f471b95566380a3b71824 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e98e395e714e4504858870286ffccfc7 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6003012484e4ba3979e3952e8b25ad1 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2eb846835e024db7a5f57c2c35a41413 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/940c3d6500b445e6910b25e989c4deb4 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d05a646097714575ab730aa3a17b081a : SUCCESS in 4m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9921818a831d42fb93008e468e5edd02 : FAILURE in 11m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/48f2ef1b82374798a481cfbf9dcbc4ae : FAILURE in 11m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e50c275070ac4b11963d859bcb7409f1 : SUCCESS in 9m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1157d22c620a445ea41be416b267146e : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/928f4de9c96240d8926320a5b3f628e3 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0677c0e0af674663829331e26e65cf01 : SUCCESS in 9m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d4add158704f47b1b553d80098311865 : SUCCESS in 6m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4c9abeaef90d4a318c5fa50f670adce8 : SUCCESS in 15m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fce68e437ee34ac2b3577ac6ef2e9540 : SUCCESS in 25m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9f289b9530274ba7b085e8fac37ee67d : FAILURE in 11m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8d86e3ceee304bb7b619e78506273281 : SUCCESS in 5m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2029a9684b304dd28da75890cf90be75 : SUCCESS in 15m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f14478d07c6541a48c6bb08428cefca8 : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cb994298d0214edc9728d4c2e04ded81 : SUCCESS in 7m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4e37db6f3ea742f3b9dc672fb96e5972 : SUCCESS in 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/384ac4cabac84374b6c78f8068e3d5f6 : SUCCESS in 1h 00m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7005cec51e7f49a78e95e256c429474c : SUCCESS in 1h 33m 47s","accounts_in_message":[],"_revision_number":6},{"id":"b7073bae0e5eab620c03c6d125cce4302a02002c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-18 06:17:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"59400b880e04d5c399d4a7e78cb0191b6cb2db57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 06:35:07.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f9b44a86a33149eab2a00b3c84a0f46e : FAILURE in 11m 43s","accounts_in_message":[],"_revision_number":7},{"id":"2b8021b83fd9d263725d42d28f0eeaf26177b1a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 08:42:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a9e2d0124744d2bf456996466feb19 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f02c8116cc046afa095e2b4fcd947b4 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/536a481ca7e3467a892d08561e8d33fa : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fc90d7b71a444d8a7d742db5731bb74 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5529ca22c5824932ba0b98d1c424b4b2 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7495c5aa6d764f53ad9b59ea100a5d50 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d4c481c63cc46e09715d666c00f8de6 : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4e9c723fdd9a4898bd22733403af01d3 : FAILURE in 5m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3655cbc7e161414a91cd8d90e4734d23 : FAILURE in 12m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/02a26c8ed90344e4a3815c2c8305c30e : SUCCESS in 9m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/015581a569fe47d3974411321666f873 : SUCCESS in 5m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/df54e9e632b54c57bb002ebb461debbc : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d8b66411e38e4713905cfa4799c2d2c3 : SUCCESS in 7m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/be2be8c143ed452eb38ca0e1cfdfd554 : SUCCESS in 7m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e1b45219a032492499fc1ca20856cae8 : SUCCESS in 17m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b936e58b86eb4cd68f69bb763a100fe0 : SUCCESS in 30m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8343b0c2db554bf3898c6eebfc043bc4 : FAILURE in 12m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8be12210dfd941af8591bb7923527cd2 : SUCCESS in 5m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac4e7e2087f1440c8131ead3a152deea : SUCCESS in 18m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/68b4ed1f14064a91bb3b5caf0153a641 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2d82f9cebeb24bf49a506647a32f8025 : SUCCESS in 6m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/db97bd03c91b43e0b583d733e419308f : SUCCESS in 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14568731a5f84fad9396305e2fbbc9b7 : SUCCESS in 1h 08m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f363032107f492295a9ca6e567eeea7 : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":7},{"id":"b179c89e7ac869bf014db12d285096d6257fbc19","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-19 06:25:08.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0d769ec7cf77178c018845fa9f76a49f591dedd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 06:42:03.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b65f8d0f5f8a4a929d3c78e537cbc74a : FAILURE in 12m 21s","accounts_in_message":[],"_revision_number":8},{"id":"3bf9838d15319347494e40ad485b324a18d00a3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 09:13:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f833407d3a64be88013563c8ac70fcb : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f6fd30c903d44b89817159e8e4bf50f : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a2861a8962c48e289ade5e01cc0b707 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/652c50c30eda40258e01497df3cd6f83 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a556ccd543df4799a84d2a05fee8198a : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/391065ca44414f36a68668efc23a7c13 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ca6af6eee4d41fd80b7b4826413e994 : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/713a387073f145ea879dabcab70a3272 : FAILURE in 5m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/246c8523b84f4c1f8d618e7e5acba92a : FAILURE in 5m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dde113b507844bd686977396e128f6b8 : SUCCESS in 9m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8ad99afead5f4cd8b072c0887b1f75c9 : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d2e370a7b8d946c58ce21cd2280e1dd4 : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d6316b6c2b442a5827b43a943761564 : SUCCESS in 7m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/feedb089fdf344119be9f47399dce3db : SUCCESS in 6m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eed6bfd603f44891a2ef2e8f190f0f1b : SUCCESS in 13m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9887ced097224672b39fe1d220023003 : SUCCESS in 28m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f1461e86db274e6c841488b8427782c9 : FAILURE in 13m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f1417bfee100440397c08c346fe23bc2 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2c75112773e485fbc9820c695dfdd90 : SUCCESS in 16m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dc3ad584e1ec47eaadb8f6923495d76b : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89e44e3094c44efb9ec4eb3b8f3f2fe8 : SUCCESS in 6m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/59fc6934334c41b78b0a3a392f50926c : SUCCESS in 10m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/106191777bec42e79d1936f9edee9855 : SUCCESS in 1h 06m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a192fee172fb435687eecf4281953343 : SUCCESS in 1h 30m 53s","accounts_in_message":[],"_revision_number":8},{"id":"36868c820e1cf189d08fbd7aa2fcaa5a5a73925a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-19 16:38:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9c81baff4bcb973b32f2d2cc9a30afa07549c3e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:01:57.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/04bd53ec863e4fa589da433ad696459e : SUCCESS in 17m 32s","accounts_in_message":[],"_revision_number":9},{"id":"3ae00919d853265578b00b4cb5315c2e168ced13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 22:01:37.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7300c0567104e4db6278cf0eb6120fa : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e295ae3d86247cfb0683f1517ad37e1 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7f4d0ce157846fc88cfafeae12c5853 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/664d4552e20141ce8e0f1c5cd2c07ec8 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0996d67047394a3ebf40e8cd3e648ade : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76ce523ce8fb4551bf8c000533465953 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d118646e8a3e43d58bc61ccfad21e5b7 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cffd6958c5bf473388b9791da33a9250 : SUCCESS in 9m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0080583e617a4fb9811c003d87bcf4a6 : SUCCESS in 8m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0a321310eab34d55a892c464edf13389 : SUCCESS in 8m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fae795dd9d864b7b863e5b78bc534760 : SUCCESS in 5m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/467f3a809c7a485bad74352517e507c0 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a74b385fa2b41cb905dc3b7c17c71e0 : SUCCESS in 6m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/590e233e5f5e4d5099efae165e3fb95a : SUCCESS in 6m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bac923a3f80b4c85b48bcdc1505a1445 : SUCCESS in 12m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/90d47c915f9340488dcb20787c40d72e : SUCCESS in 22m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ea4e2c22544f4d22a11b37bbe948a272 : SUCCESS in 11m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/52281cdd18f84130abe9fa06235b13bd : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d085304d5b1d41e49d436fd5b28b02c1 : SUCCESS in 17m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/50d128f58ead4a8c9afc8beecae8e3a3 : SUCCESS in 4m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/833bc4d6ae974c69975375c13b83468f : SUCCESS in 6m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/92758b5cd8a14ae4ab051e3d543e3921 : SUCCESS in 12m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/553be8ee77be42e2b4985042c50c6edb : SUCCESS in 54m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a36717cf50ad4ef1a7748c2e5ce0d936 : SUCCESS in 1h 22m 12s","accounts_in_message":[],"_revision_number":9},{"id":"d8dc39fcee902d630c0a17ddbf85cf36885dd444","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-19 22:05:32.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"3979f02589c006d4b2afcfa40d1d363608012ece","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-20 07:32:52.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"77f3aa3ccb05d9f5a74706cf909f1c9aaaf243cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 07:33:08.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"df0de2af3588208be6c65ef524c3cc55437dcb3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 09:07:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"21b9d4b55039472f84d89440f62890cc8dc7b108","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 09:07:57.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec54205492fc474fbb56e8c08fb36560 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/543b50eb316a4341af0cdbb335075e4f : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd78c8ec85474e1d911497e482f9bd38 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23587de7ed414363819db9dcec8d3329 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e296916b767b46c59c41b3b8b80c0b72 : SUCCESS in 5m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0acf6c8cead4cca9b9a62d92b46e1b0 : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/72797840c8404b3cad6a5316687bb94e : SUCCESS in 12m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/68f7b586640b4eb786442dd8b2ee4df2 : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0a64a09fbede4454ac744f921521fe98 : SUCCESS in 14m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/469334ad68e640eea0437a0cffac5505 : SUCCESS in 6m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/942cc68df26c4fdaaa428a4cd7faf8dd : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87e44843b3d8424d9c866a5c2f7090ad : SUCCESS in 6m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/34c9b82eb5244d24b066ea8e496361ff : SUCCESS in 8m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/96816869633842d38df9ee908dbd1101 : SUCCESS in 15m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2ac58fd731154d4ebb111a743000c338 : SUCCESS in 26m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b3de9517d05649a58542d3f3bdf5fe3e : SUCCESS in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/498bdf7b56744db2abdccd04930d2399 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dfbb866bd40a4456adfae51a1f819902 : SUCCESS in 18m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/beb43b2918d14e0b95ed6468204a26a8 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a44766d50604264820638237291ee27 : SUCCESS in 6m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c58b175bdf414f42b38c5fa6cb9b3b52 : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b20385de7f6443f5b33f9151fa9b42e3 : SUCCESS in 1h 02m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9d4dc14b9c64044aedffa0253e7b3c5 : SUCCESS in 1h 26m 33s","accounts_in_message":[],"_revision_number":9},{"id":"eefb5bd7794e30a4c18b092ea49ce0a9cf5b3b3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 09:09:21.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b91aa12164db454099c0df15e7821e7b : SUCCESS in 57s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"937ccadec9b94ebb111bbde8d4a9b5eb6690365e","revisions":{"581c1cc6ef27b5d9557bf999ae2d26dd1c961148":{"kind":"REWORK","_number":1,"created":"2020-11-11 06:52:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/762296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/762296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/762296/1"}}},"commit":{"parents":[{"commit":"96a331ef9f8de55ca49ce21ed096afacf1315035","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96a331ef9f8de55ca49ce21ed096afacf1315035"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-11 06:52:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-11 06:52:15.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I27fda78e9f16794e7c89e4c59c6f0667f01ed7ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/581c1cc6ef27b5d9557bf999ae2d26dd1c961148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/581c1cc6ef27b5d9557bf999ae2d26dd1c961148"}]},"branch":"refs/heads/master"},"fddd9ebed0f48493f56c1768861ebe46224884fe":{"kind":"REWORK","_number":2,"created":"2020-11-12 06:22:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/762296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/762296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/762296/2"}}},"commit":{"parents":[{"commit":"96a331ef9f8de55ca49ce21ed096afacf1315035","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96a331ef9f8de55ca49ce21ed096afacf1315035"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-12 06:22:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-12 06:22:25.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I27fda78e9f16794e7c89e4c59c6f0667f01ed7ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fddd9ebed0f48493f56c1768861ebe46224884fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fddd9ebed0f48493f56c1768861ebe46224884fe"}]},"branch":"refs/heads/master"},"c6052ee2f1291444bcf07b61f08736048887b21f":{"kind":"REWORK","_number":3,"created":"2020-11-13 06:25:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/762296/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/762296/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/762296/3"}}},"commit":{"parents":[{"commit":"ca76284604e0d6afdb28ba17965604ad8de1ff3c","subject":"update constraint for osc-lib to new release 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca76284604e0d6afdb28ba17965604ad8de1ff3c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-13 06:25:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-13 06:25:11.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I27fda78e9f16794e7c89e4c59c6f0667f01ed7ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6052ee2f1291444bcf07b61f08736048887b21f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6052ee2f1291444bcf07b61f08736048887b21f"}]},"branch":"refs/heads/master"},"04024bdf41ba46c0e830af64982f4efc3a8d59aa":{"kind":"REWORK","_number":4,"created":"2020-11-14 06:21:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/762296/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/762296/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/762296/4"}}},"commit":{"parents":[{"commit":"ca76284604e0d6afdb28ba17965604ad8de1ff3c","subject":"update constraint for osc-lib to new release 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca76284604e0d6afdb28ba17965604ad8de1ff3c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-14 06:21:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-14 06:21:54.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I27fda78e9f16794e7c89e4c59c6f0667f01ed7ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04024bdf41ba46c0e830af64982f4efc3a8d59aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04024bdf41ba46c0e830af64982f4efc3a8d59aa"}]},"branch":"refs/heads/master"},"59f2e177f76d1b9012be1e3cf0ca3860907a778d":{"kind":"REWORK","_number":5,"created":"2020-11-15 06:39:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/762296/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/762296/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/762296/5"}}},"commit":{"parents":[{"commit":"ca76284604e0d6afdb28ba17965604ad8de1ff3c","subject":"update constraint for osc-lib to new release 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca76284604e0d6afdb28ba17965604ad8de1ff3c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-15 06:39:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-15 06:39:33.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I27fda78e9f16794e7c89e4c59c6f0667f01ed7ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59f2e177f76d1b9012be1e3cf0ca3860907a778d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59f2e177f76d1b9012be1e3cf0ca3860907a778d"}]},"branch":"refs/heads/master"},"8f3658a09d21bf5f5464da3d04e8b842fc001a8c":{"kind":"REWORK","_number":6,"created":"2020-11-17 06:23:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/762296/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/762296/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/762296/6"}}},"commit":{"parents":[{"commit":"ecd19592c15f53a7f9ad8bc78620488d1aab3c1d","subject":"Merge \"update pyghmi to 1.5.19\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecd19592c15f53a7f9ad8bc78620488d1aab3c1d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-17 06:23:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-17 06:23:09.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I27fda78e9f16794e7c89e4c59c6f0667f01ed7ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f3658a09d21bf5f5464da3d04e8b842fc001a8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f3658a09d21bf5f5464da3d04e8b842fc001a8c"}]},"branch":"refs/heads/master"},"c25a3e857d8c65f4923ba42b8ec563586ca938c9":{"kind":"REWORK","_number":7,"created":"2020-11-18 06:17:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/762296/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/762296/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/762296/7"}}},"commit":{"parents":[{"commit":"ecd19592c15f53a7f9ad8bc78620488d1aab3c1d","subject":"Merge \"update pyghmi to 1.5.19\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecd19592c15f53a7f9ad8bc78620488d1aab3c1d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-18 06:17:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-18 06:17:53.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I27fda78e9f16794e7c89e4c59c6f0667f01ed7ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c25a3e857d8c65f4923ba42b8ec563586ca938c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c25a3e857d8c65f4923ba42b8ec563586ca938c9"}]},"branch":"refs/heads/master"},"63266cf94dd4210921f5d0b3bf3828cfd39e07b5":{"kind":"REWORK","_number":8,"created":"2020-11-19 06:25:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/762296/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/762296/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/762296/8"}}},"commit":{"parents":[{"commit":"ecd19592c15f53a7f9ad8bc78620488d1aab3c1d","subject":"Merge \"update pyghmi to 1.5.19\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecd19592c15f53a7f9ad8bc78620488d1aab3c1d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-19 06:25:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-19 06:25:06.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I27fda78e9f16794e7c89e4c59c6f0667f01ed7ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63266cf94dd4210921f5d0b3bf3828cfd39e07b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63266cf94dd4210921f5d0b3bf3828cfd39e07b5"}]},"branch":"refs/heads/master"},"937ccadec9b94ebb111bbde8d4a9b5eb6690365e":{"kind":"REWORK","_number":9,"created":"2020-11-19 16:38:35.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/96/762296/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/762296/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/762296/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/762296/9"}}},"commit":{"parents":[{"commit":"c15df92629dff5640c8ced0abe2f66e1134f2f5f","subject":"Merge \"update constraint for openstacksdk to new release 0.52.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c15df92629dff5640c8ced0abe2f66e1134f2f5f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-19 06:25:06.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-11-19 16:37:09.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I27fda78e9f16794e7c89e4c59c6f0667f01ed7ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/937ccadec9b94ebb111bbde8d4a9b5eb6690365e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/937ccadec9b94ebb111bbde8d4a9b5eb6690365e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
