)]}'
{"id":"openstack%2Frequirements~762541","triplet_id":"openstack%2Frequirements~stable%2Fstein~I872c2bec74e28bd42265dc909f945a7873eccabc","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I872c2bec74e28bd42265dc909f945a7873eccabc","subject":"update constraint for horizon to new release 15.3.2","status":"MERGED","created":"2020-11-12 16:04:47.000000000","updated":"2020-11-13 22:20:43.000000000","submitted":"2020-11-13 22:20:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762541-1605306043069-4b8f775d","meta_rev_id":"b2785107e0f6e81a32c397bef9f3c94eb67c7519","_number":762541,"virtual_id_number":762541,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-13 22:20:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-12 16:05:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-11-13 16:58:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-11-12 16:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-11-13 16:58:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-11-12 16:10:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-12 16:05:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-12 16:10:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-13 16:58:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-11-13 22:20:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1fa7c2dfe057242c509ff716305eab2a56fda11","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-12 16:04:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6863c8c8376269203c41696772ac5344834c821","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-12 16:05:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e5e0199a0ceb94991d0cf461fa8c3045f49b21a9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-12 16:10:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"289f57afb962f76e90714ce1a39d380e39e69602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 21:27:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deacdd530fa7421c80b40766a1bbec48 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cde47f244356492484e54d71646e9157 : SUCCESS in 3m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4984f5a5f92f49d3b05d453570093f0f : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31cbd12bf35948d2adae2f243d2453ac : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ef2242f634e4a8caeee864836c41363 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b32ed655cf9240c391320579788dfa7d : SUCCESS in 3m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/50949679b9e841468b71287a19314c41 : SUCCESS in 8m 12s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/45cf39f02ed04dff8b7d0292b3f0a40d : SUCCESS in 7m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a2cb194703c24d5d9e77cd83addf4731 : SUCCESS in 7m 12s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/61cb3c85c08045b593c10f2f9a96eb04 : SUCCESS in 9m 04s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b57eae28ed59415a844eaed2e5c6a889 : SUCCESS in 9m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/df3bd12f47da4a6fb503900f09a146a9 : SUCCESS in 7m 08s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/781c38761850412f990b095265f7c1fd : SUCCESS in 7m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2cbf95cd0ea34b2fa0c90b4e7aee8568 : SUCCESS in 20m 12s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a4594e7f60c948a7a740b46c8499b494 : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f3eea19f16444f2a5e89a7926646302 : SUCCESS in 7m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d20ef9cfa5d54192bd00cbf83ee1f283 : SUCCESS in 22m 03s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/65f14d6b472446c384d359d7239ab038 : SUCCESS in 28m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a4a00e49ee954a45a3cbfc9edcf10188 : SUCCESS in 22m 08s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/43fab00032144877894b331639714924 : SUCCESS in 22m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f90e8cbbaa564da49bbee0ebd0d68887 : SUCCESS in 13m 07s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/6a2ff7f9d9264e70b506de05243fcefe : FAILURE in 31m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ff9be6db6884c54a570be232ba1fc10 : SUCCESS in 20m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4996fb13214c4e6db966f8a0868f19e9 : SUCCESS in 11m 00s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/7937a9bdbc30482099267e40d375c51a : SUCCESS in 10m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/01df47d54d124ee6813ce048e56eec39 : SUCCESS in 1h 05m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/81be5cfb03b049d392668d60f352994f : SUCCESS in 1h 08m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/02906c0349674864be12403ed5a8fb4d : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4d6fbbe4e5484c243ca58df5c3326eac73948e02","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-13 16:58:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"664993d069ab9d0cc33ca2b5654df426e72c44e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 20:01:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e96119e0653b488e9b249b86e2965716 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b59aab083cf64e3cb042cf4d051eb2e4 : SUCCESS in 5m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d4077c91b58540a1b6ad0bb948f43393 : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9bffd24dbfb48d48f28245639a8ee26 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/39fccb0550744b228064644d08d9ed79 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4aa7407d60b240b58744805738b44e95 : SUCCESS in 4m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6b00b3652dbf45bd991ce8f771e48466 : SUCCESS in 13m 55s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e6078345abe94d69b2889b363ddf6556 : SUCCESS in 7m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/24ce225a072c46a6be147db3499f4677 : SUCCESS in 8m 05s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9780bfc325fd4067875a5e8a752ce020 : SUCCESS in 9m 08s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/f53b837ddf2749f98c815a05b228dd4c : SUCCESS in 10m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/79d4cdd23b6f405b89ccc26e4cd2b838 : SUCCESS in 5m 45s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/db3faecbf6d84e2d87c149fee46f3d74 : SUCCESS in 6m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5519973e9f504fc684edb2f301f01f77 : SUCCESS in 11m 00s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/7f43bd4929614bbf84a6ea7d2c8246ae : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0fd4ca9d88794509b3d39c8f1e7a86c8 : SUCCESS in 6m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c1cf75aa64d3435dacc319f8dfbeb5a2 : SUCCESS in 32m 12s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/fe46eb13877c4150ac406c0b7b41cc3f : SUCCESS in 31m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0f476e60d41849fd8b5697df2fef4dad : SUCCESS in 25m 31s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/977f5d142bbf4dd493adaccc402101ac : SUCCESS in 26m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/03e77ecf0a1c402a99ae7ebac313f0f8 : SUCCESS in 11m 25s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/1f8517dff5b449c589e1b87313e69944 : SUCCESS in 12m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7efe8e6b7c14785a1f66087c7616168 : SUCCESS in 21m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/acfbb8440ac345cb94bf5a38579d977e : SUCCESS in 8m 59s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/d5f77c6055884fe282f88a7efbc3bc81 : SUCCESS in 9m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf3f3e976ff549f191090274b6d1b442 : SUCCESS in 1h 28m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f325f18e495641eeaa2bea8c9df446da : SUCCESS in 1h 23m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/172c2908d6ef4476b69995de6906d845 : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4281e7354e67f2d18381b9070d0c0aa97b23d657","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 20:02:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de86baa53ef13c4a32cf5274a108edc489015f52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 22:20:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b45a5d63d9243b1ba4f2733f7e5e34f : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1eddd98c243e44ed8b853fabd17d719b : SUCCESS in 5m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/731ed3ddfe58437fa61563467aa30e2d : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aeb634c458c44f60bba06ee53cebfdaf : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dae79b5254ad4e3d9869d621eaa0d80f : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b8a15d6644e240aea26de69c687dfa05 : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/031e00776b82410781607a6abc176ca8 : SUCCESS in 7m 37s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/1ea528c53a6649d197a5fc6bf2a2606a : SUCCESS in 8m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e8db78902b7b49248917854353ef5095 : SUCCESS in 9m 30s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/888ba1d60a104c06a12cbbf3a6279bcb : SUCCESS in 7m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d190ba37ca7b4983b56996d4b53bb281 : SUCCESS in 4m 24s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d2e02aba4eba4caf8ed5999147389f53 : SUCCESS in 4m 40s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/71e14e759c9249c893c01fe9b4e8e8f5 : SUCCESS in 8m 26s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/0a055c9de2c6461ca95d4548d56bf221 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/341ddedc9ab4436d8602179ad1526091 : SUCCESS in 7m 35s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/853e0ca6e64e45ab9e021d66549bd232 : SUCCESS in 20m 10s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/f178f456a76d4034913063c91eda5d13 : SUCCESS in 40m 20s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c1cb9a5f1b884746970fcc1a94cf97ca : SUCCESS in 16m 44s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/c59b7e64420c418a948ce9c140056636 : SUCCESS in 17m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5cf3646d8a9746a8b695fd9f3e7009a9 : SUCCESS in 19m 35s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/afd6953653c345c5819cea21a0bd3c55 : SUCCESS in 12m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bfe4544121d047a4a7e2c003f50f509b : SUCCESS in 22m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7887e397c91d42588a9a1736ab5d6ceb : SUCCESS in 8m 22s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/8373c7c7540749d680c465c5cdfd5196 : SUCCESS in 8m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1635e2a26a6840079fa55e7564903827 : SUCCESS in 1h 25m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/55f70ed8c582461d8fc24775d09aea2d : SUCCESS in 1h 16m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/58b34bd621df441581618778d6eb22a7 : SUCCESS in 47m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c5c2384597ad24c88cb4b128c60cf8e312db23dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 22:20:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e80e3ac77db4d6ecbe958299da302fc02f0b563e","revisions":{"e80e3ac77db4d6ecbe958299da302fc02f0b563e":{"kind":"REWORK","_number":1,"created":"2020-11-12 16:04:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/762541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/762541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/762541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/762541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/762541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/762541/1"}}},"commit":{"parents":[{"commit":"4ae866a67b3297dd82912c64c758dfc62186d62a","subject":"update constraint for neutron to new release 14.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ae866a67b3297dd82912c64c758dfc62186d62a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-12 16:04:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-12 16:04:46.000000000","tz":0},"subject":"update constraint for horizon to new release 15.3.2","message":"update constraint for horizon to new release 15.3.2\n\nChange-Id: I872c2bec74e28bd42265dc909f945a7873eccabc\nmeta:version: 15.3.2\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta:release:Commit: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta:release:Change-Id: Iabf8a84af52bb372a544ff8885e80369953aa653\nmeta:release:Code-Review+1: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e80e3ac77db4d6ecbe958299da302fc02f0b563e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e80e3ac77db4d6ecbe958299da302fc02f0b563e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
