)]}'
{"id":"openstack%2Frequirements~762545","triplet_id":"openstack%2Frequirements~master~Ic5503c27b0c4fa9268a35f787ca1759ecbfa806d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic5503c27b0c4fa9268a35f787ca1759ecbfa806d","subject":"update constraint for osc-lib to new release 2.3.0","status":"MERGED","created":"2020-11-12 16:06:20.000000000","updated":"2020-11-12 23:17:08.000000000","submitted":"2020-11-12 23:15:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762545-1605222946888-b3c3e99b","meta_rev_id":"a68fcc43b9b2e4b5896d9fac5ec7485a049f4e84","_number":762545,"virtual_id_number":762545,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-12 23:15:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-12 17:41:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-11-12 16:10:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-11-12 17:41:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-12 16:10:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-12 17:41:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-12 23:15:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97820b024764aaefee84adec027f99536633406d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-12 16:06:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40ecb4b44fd06fa626365ac453270c3361d2d011","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-12 16:10:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"92da1b7786e62ccde7aebdd6ac76acf7c8153bdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 16:29:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f9eab2d5f41a435d9748064d8561f9b6 : SUCCESS in 18m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a92fab3225821348fdeb5aa3d56f241324492a2d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-12 17:41:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d55d5513f0fe4c68a708e855ef18ae0bec5b2d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 21:43:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/767c6d7dd1954bcf9dcea958bef1a9a6 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40a87d985328435699ac57a3ce3d884c : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ba4aabe24f840c6a2ca616781994abf : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d50c0af345094deaa2293c0e6ff9db51 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f54fe04030a4551a1c6c9b02d6f9013 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d0193c07b1d47d6a1ad6f7da9b8b7ab : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e5c850c16a9746f5a81270171905c6ed : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cca627347e524f6d9d886f5cd05ec3b8 : SUCCESS in 10m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/723ce68c900449f0b976eaf70ed6fd9a : SUCCESS in 9m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0733b2712be84cc7a6968931843f80df : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/04bce90d455c4e3fb47d69865bba2921 : SUCCESS in 5m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c32f79c637d64fe190d1b5f8e1697175 : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b3a387d4abe456e9b5f884b8db9d53b : SUCCESS in 6m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a2c910179e584d06b25cdcbbed341dd8 : SUCCESS in 6m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f0831e64ead247df8bee59808e0f879c : SUCCESS in 12m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/296c4b1a91ba4ecb8dc28e479d858e94 : SUCCESS in 26m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3cb4bcd4403d4ca2ba3980ba408ca362 : SUCCESS in 11m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea757d32e1d6458da7ebbb0bfcab2046 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b4e1d6a6d2e9460daf6b080890b752dd : SUCCESS in 17m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a364de17938d48e897aa9e5409f4c078 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dac33b3c24104fe0ba0fd78522510b79 : SUCCESS in 9m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e29a9dc6ee5e4aa6af053562652bb29b : SUCCESS in 9m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91bc6c3697394f00945adf3723c3e3d9 : SUCCESS in 53m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/154f0034b3814401a22f0aa27fce995f : SUCCESS in 1h 26m 30s","accounts_in_message":[],"_revision_number":1},{"id":"06e0acc40b119dbb2f93ab6b900b2036f067c5ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 21:44:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbff319e8ec8aadf9a74dfd7c9b9127f0a5c59bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 23:15:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5bdf92ecb364a232dc7766c6980cd7681bee6ab7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 23:15:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63cd1d185b914971a2cb08d280bc651a : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66fad64c4dc9406182e3abcb6a8666e2 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a237ac614da4b6d87b4ba14dedc7640 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b80bac903c84be4971e7febaa462569 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8b861c4900c4acd95bc9b04dab40c2e : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cb144e1acb2a45c097073641ad92775d : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dc86f858fe684255a8eb81cb3a3fb811 : SUCCESS in 8m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f2db1b688824a4a97988824c7cfbccf : SUCCESS in 7m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e336aed46c5e4511a7869452cd5bdde3 : SUCCESS in 8m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c15b6c0de3d74d11af848ea20d7873a2 : SUCCESS in 5m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ae9c5f4c3c994878bb336d4f3fd28ad2 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c703deb9bb54b488b0d34ff50fce2ed : SUCCESS in 8m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0a961090598043cf8deddf65004d81b3 : SUCCESS in 7m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/309bb591ef4d49d1819485d48a486eb5 : SUCCESS in 13m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/379ef0570e0f4b49a1e8f5a061757cac : SUCCESS in 28m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f509abb1b6bb4a9f9a4a79fe193ba35d : SUCCESS in 11m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cfc8fb07b29f4593a4c56c413ef96124 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/011122b304d743c4bcf6019d61f35510 : SUCCESS in 18m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6a0eeae65f0b49a995f8eff92fde2d3c : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c68edbc15a1a4d10b665eb9cf3a5f4ce : SUCCESS in 9m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fb5fbc48698547fbb9f82f65208b1216 : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05511c590d8440bc99160557b628d2d7 : SUCCESS in 59m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3110290723f2411faca46dad48521898 : SUCCESS in 1h 24m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ad4124cd0a4f07769bdc934e014ef5062ae650e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 23:17:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b12a230c90c644b08e7e7d68fc52c80d : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca76284604e0d6afdb28ba17965604ad8de1ff3c","revisions":{"ca76284604e0d6afdb28ba17965604ad8de1ff3c":{"kind":"REWORK","_number":1,"created":"2020-11-12 16:06:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/762545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/762545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/762545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/762545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/762545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/762545/1"}}},"commit":{"parents":[{"commit":"c843db74b4558dcc506ac9220b013cc7a0130f38","subject":"update constraint for cliff to new release 3.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c843db74b4558dcc506ac9220b013cc7a0130f38"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-12 16:06:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-12 16:06:18.000000000","tz":0},"subject":"update constraint for osc-lib to new release 2.3.0","message":"update constraint for osc-lib to new release 2.3.0\n\nChange-Id: Ic5503c27b0c4fa9268a35f787ca1759ecbfa806d\nmeta:version: 2.3.0\nmeta:diff-start: -\nmeta:series: wallaby\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Artem Goncharov \u003cArtem.goncharov@gmail.com\u003e\nmeta:release:Commit: Artem Goncharov \u003cArtem.goncharov@gmail.com\u003e\nmeta:release:Change-Id: I08d5af57d9bb3f4209efa4424ef6d6427dec6b2e\nmeta:release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca76284604e0d6afdb28ba17965604ad8de1ff3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca76284604e0d6afdb28ba17965604ad8de1ff3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
