)]}'
{"id":"openstack%2Frequirements~763239","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Ifccdc38160811d2d1097607fa4ed031dc8a59136","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Ifccdc38160811d2d1097607fa4ed031dc8a59136","subject":"update constraint for oslo.config to new release 8.3.3","status":"MERGED","created":"2020-11-18 17:57:12.000000000","updated":"2020-11-19 09:07:20.000000000","submitted":"2020-11-19 09:07:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763239-1605776840460-b16e9fe7","meta_rev_id":"2bd9adf9226d112fb7cc67301675930b91b8a741","_number":763239,"virtual_id_number":763239,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 09:07:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-18 18:15:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-11-19 07:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-11-19 07:09:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 18:15:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-19 07:09:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-11-19 09:07:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1a22f2a78c33ca54d88adccc73b3fda304c8509","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-18 17:57:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e260de0187916cb3cefea8feb37fa9e23eb8081","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 18:11:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/812272b280fb407d8aa1e70a155f8ef5 : SUCCESS in 9m 26s","accounts_in_message":[],"_revision_number":1},{"id":"88bfd13bb8c4a934c3c010e5164aa0b085524e56","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-18 18:15:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3399fb48578444fc937ef27f2ee714419124cb16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 03:45:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f976164503640afb02db1a15f7ece6b : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cca2e9e4a8b49ffbe9ac25ae0adbae0 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e364f1e6729447068b4f9720ea91d110 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a326737095dc40c994f269a987b982a8 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e042716ded644524b675d044503aed85 : SUCCESS in 3m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/291d5496a3e742bc8b8f7faf256c0712 : SUCCESS in 12m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f349e297dbe941269b2d0b48eb03df7c : SUCCESS in 8m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8da14b474a5e4fd88379c46fe95f72f9 : SUCCESS in 14m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fe16ea9662524fb3a4062b70b4c65b3a : SUCCESS in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2f2cb07c136449f589888df14e16a1e7 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/afcf2a9953294db9898733b7085437b3 : SUCCESS in 11m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/90cce422d0644fd9a87b2dd3074110fa : SUCCESS in 10m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/348a1ea5c1b94775b0d742d7c175e002 : SUCCESS in 20m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/454796a1132e46a3b1e5f4267e506254 : SUCCESS in 29m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fd5bdf745b6645a3b9865d8440a3ae03 : SUCCESS in 12m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4301f954930548ff8eafab7780c387c6 : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a52a8779ac2b4675864a0643afb5275c : SUCCESS in 18m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/df4a5d6ab5a14c77a6b4841fa184a08b : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3806306498a04c0ea92753577001ce44 : SUCCESS in 6m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d30120990902430f923cb3ecc6c154a1 : SUCCESS in 10m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9a2e41e32df4b8dba409247e7f5e120 : SUCCESS in 56m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97ac7d262a214605b17d0c9ba127748a : SUCCESS in 1h 21m 19s","accounts_in_message":[],"_revision_number":1},{"id":"39b74e65f86edb4e86279028592e9882a429f1e2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-19 07:09:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff3de222fae2a011e05bb3d73eea567f54cacec7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 07:10:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e796ff4a1c397c2da29285f92ccc53d774d8efef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 09:07:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd4b1ed461548fcb3fefb5399ab86b9 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff9fd53c4d7e47889b8c857b35bac34b : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6555197001d4adca6119e4ecf460995 : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d141699d869849bcb68ee0c476fdc2a9 : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2365cfd2e8da497bbc270bc542b5fb16 : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3cabed145b274c1e833a524ecf261d09 : SUCCESS in 7m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ad7d4333a04d418880566f574953269c : SUCCESS in 8m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f02cc996fbd04cef95920fe70ea46a54 : SUCCESS in 10m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ef423fcfad104b4db1e4eb2c5127b9b7 : SUCCESS in 4m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c4a7c6df71974ed4803ea2f0c1e40301 : SUCCESS in 9m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e3e717da2e084482b629def2d2daa667 : SUCCESS in 8m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/561b85b8c5594f32b2e440d8ffecbd2d : SUCCESS in 8m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5bea42fab21f400eb2a0a5493a3184c6 : SUCCESS in 14m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5d3f2c2dafdc45fabea3bb46e3a5bda6 : SUCCESS in 19m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/344fc50d8d3249b1befb3a1734912f92 : SUCCESS in 12m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c6605f967a6f4599af7faf60a4eb5906 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8296d1f66838436098ae686ef0bc62cf : SUCCESS in 17m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/68062d461e1341ff9333ac0db35dae55 : SUCCESS in 7m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09227ff814e14f55816876c3cc6581fe : SUCCESS in 7m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b31d967a03054519901cd2f63d9fb756 : SUCCESS in 9m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/daa0c390e36a47018b151ca26f61d197 : SUCCESS in 53m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c655e2ddbf214d258cf047942d4757a0 : SUCCESS in 1h 26m 36s","accounts_in_message":[],"_revision_number":1},{"id":"fdef23c8abfd878a29665d4c743fd471eeb699bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 09:07:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c6ba9b1ee76e5670d201bed97fa44611b93ba8b","revisions":{"0c6ba9b1ee76e5670d201bed97fa44611b93ba8b":{"kind":"REWORK","_number":1,"created":"2020-11-18 17:57:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/763239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/763239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/763239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/763239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/763239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/763239/1"}}},"commit":{"parents":[{"commit":"e0317a9c681cf00f6418c35ed6e4f8335e6c44c6","subject":"Merge \"Update u-c for kombu, amqp and vine\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e0317a9c681cf00f6418c35ed6e4f8335e6c44c6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-18 17:57:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-18 17:57:10.000000000","tz":0},"subject":"update constraint for oslo.config to new release 8.3.3","message":"update constraint for oslo.config to new release 8.3.3\n\nChange-Id: Ifccdc38160811d2d1097607fa4ed031dc8a59136\nmeta:version: 8.3.3\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Change-Id: I6c54381078985a6af2788b9a876cc7f2351cdf21\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c6ba9b1ee76e5670d201bed97fa44611b93ba8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c6ba9b1ee76e5670d201bed97fa44611b93ba8b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
