)]}'
{"id":"openstack%2Frequirements~763357","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I3c820c5b49ccebc0841555c99ca08df4771fb438","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I3c820c5b49ccebc0841555c99ca08df4771fb438","subject":"update constraint for oslo.concurrency to new release 4.0.3","status":"MERGED","created":"2020-11-19 11:40:03.000000000","updated":"2020-11-20 03:32:04.000000000","submitted":"2020-11-20 03:32:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763357-1605843124879-b16e9fe7","meta_rev_id":"0b131cc14b567b008128ca12c206f3c679480172","_number":763357,"virtual_id_number":763357,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-20 03:32:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-19 15:46:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-11-19 21:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-11-19 11:43:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-11-19 15:46:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 11:43:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-19 15:46:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-19 21:19:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-11-20 03:32:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba7846eea19316ecc0fd4369c6998d842784092a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-19 11:40:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8476422bb79f1167e37701efd8d9bdf8332934f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-19 11:43:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"17ea54487532a9b51defd3e59fe303d51605f494","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-19 15:46:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c92c9978159bcc4987d8a80550deb201e247df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:23:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd4244e56dcf4df1ad4928888502e81a : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/283db9440fb540f6b8ac5040c4372f74 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50eefb53c7d242a38d4a0286fa9eda86 : SUCCESS in 5m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b572c9bdce24cdbac2e7a7db8cfc0c5 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1fd8f644b0624722a64fd79bb2e02d54 : SUCCESS in 3m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4f84f1af2df345a8aa9b74c59ca8e28b : SUCCESS in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d27975750e874470b686739f67e332db : SUCCESS in 10m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7d72a639c86d4c96839c503ae5c5db9a : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2dd4c441e6fd4631a077da61b5b70d32 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49174ab4e87f47c6ae41a5607b08a248 : SUCCESS in 6m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/044dff90180f4527addb7893c1d8c7bf : SUCCESS in 14m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/214243d6fe15473ca096d9396e2f373f : SUCCESS in 22m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/839dd26bfcbf4ad4b6d62109313e3271 : SUCCESS in 10m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/67eb0176f6aa4fba9a172c0a961d1b6d : SUCCESS in 11m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c362be0f568547b19ae187c1b9b15c90 : SUCCESS in 17m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/50bd496911d640cd9d714cb2bcd7bc2f : SUCCESS in 6m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a1bccc291eec42b8872b680c68684e1b : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9e262de114d413c8791e62387ddbf8c : SUCCESS in 7m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a1deff8129564ce7a1e62f214443a226 : SUCCESS in 15m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/26cf1b6ee7154afba31383184ce7f729 : SUCCESS in 15m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/775b10f0ae7a4e37a88f7ab4b0ab6fbb : SUCCESS in 48m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b43e4d985de4e209264f41e02619037 : SUCCESS in 1h 44m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16c3dbde89bc437aae8794b615203373 : SUCCESS in 1h 14m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/c8635e9e5c894e3ab958880c9d841634 : SUCCESS in 1h 20m 05s","accounts_in_message":[],"_revision_number":1},{"id":"23f3e34c0e32f64a171ae8c04bf02835626a791e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:24:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c3f00b94535c0c2a1b2b33e3b8f8e6d61185e24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 19:01:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f943305886f5458d93169f77e519742f : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb873f1843974f82b7440c591714d737 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcf92b627dc3475db405e3ecf936c08f : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93994752bc184edf9626d367a2ba4be3 : SUCCESS in 12m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/48dca76e8f5f4586bf1150b48d98b9f7 : SUCCESS in 11m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ca278cb52e504f90abcbc0a6fe5016b4 : SUCCESS in 7m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/622de7d4a3e04eaf99f7b81383b5a039 : SUCCESS in 8m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5130a8fe1a09450583749f7f3c03f7b6 : SUCCESS in 4m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a7b61546fb664d63b33540a270482750 : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5e850319a2f4d3c9be1505e805f5c34 : SUCCESS in 6m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/653d9237b092440aadf561e486ee683d : SUCCESS in 12m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8760be7405b1465fb7f02cee44b4b967 : SUCCESS in 23m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6999fcac6c7e41c59815cad655d15116 : SUCCESS in 12m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7ab129f28f574d52b9358a6c137dbc69 : SUCCESS in 16m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/096ced68851a4cc4b19e73a10de9c49d : SUCCESS in 19m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2efd7b5016134eb1826dda480d87fb55 : SUCCESS in 6m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c23da936d676475cb2297fda4e0dd99b : SUCCESS in 7m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/499ae0b817ad4da7902fcb913cf404ba : SUCCESS in 7m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e3b27e8210a94d4a936f1e4d93fc0dca : FAILURE in 12m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/61777914f002470b9b57f91a41fc07a3 : SUCCESS in 10m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/194b660c2e454014b6fde3870459379b : SUCCESS in 52m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/543ae35d827947f8bf299385eb050211 : SUCCESS in 1h 26m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae42d04e3a4b465da0fe852f7aeb5508 : SUCCESS in 1h 29m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/cb5521f9cd9a4cf2b554f897e5ee73d9 : SUCCESS in 1h 28m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a0566b5708b118b33673ac16cb2e9eb2d1362e46","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-19 21:19:09.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"82b2395eb167de3ea1a8352b71c526e2c82680d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 01:58:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8719a82a3bf94399becc89cf718f7c8f : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50f35d770bed4832b98d3679047af488 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36414cb0bd2749eca24b0258dd73bbf1 : SUCCESS in 5m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/72552bf4484040e6b84611beae6757ca : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c14b6d09a3454b06aa7613ea3613229a : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/250e2d88d7c04e9599a5e5bbe30340b3 : SUCCESS in 7m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/442554ba628e4c0d8c0b846bc7cd3206 : SUCCESS in 8m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/02a547d5e3514ce7b972dd78e0eba5ff : SUCCESS in 4m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/362c8ddc9cb24733a6837890581779ac : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0bc9887d82da45aa8071350678311b31 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eb4fa2993c4b4bfab61428d978c4087b : SUCCESS in 14m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1d9d33a00f694388b0883c6abfb801aa : SUCCESS in 28m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/29c450c8dd624380994b163c2971aab9 : SUCCESS in 14m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0c26cdc2ce4942f89c225ae424f21275 : SUCCESS in 14m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40f1e615592c41abb8c0448c6bb03cf3 : SUCCESS in 28m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/80b38ce16a064ea1adf1cab8183e7a50 : SUCCESS in 4m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ceaeceaa253c461ba84d21ad387f5e61 : SUCCESS in 11m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f5e9e2cc68d46d89d31bfc46c9e5cdb : SUCCESS in 6m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dc7f307a95234245a7ff3232a199a081 : SUCCESS in 9m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/83bfc676f31543ca9b336014b02e4893 : SUCCESS in 10m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ad603690c674a5592a3d76e25e8e920 : SUCCESS in 53m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c37d6dc7901a4792b7f9035874afda7e : SUCCESS in 1h 19m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f9bf8e2717e4144b6e458939cb03406 : SUCCESS in 1h 26m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/730fa159b02e44499e6e9c5fc2848e5a : SUCCESS in 1h 22m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8a5293e8655bbca1ae85d35b0467a99a661b7d75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 01:58:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60069543d434b1d1af6c73741675cda887398c23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 03:32:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/088a7b09032a45768b7aaf71e4771c4b : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a5476e12ee548ef849c8aa47d77d5fd : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce5e56309c8341d681da9181b09e8b95 : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a6cf18a72134601b7d92d71f3cb720c : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81d0ea7024bb406689e198be1691d4e8 : SUCCESS in 3m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/36c7acd801a24a89acd42cd14cc90428 : SUCCESS in 9m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ae846a17e2a7439b8c994c76ff94e50d : SUCCESS in 9m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/855265d0b1ef4c6ea0284bab876112a1 : SUCCESS in 4m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1143f755f5ff474694afac27cca25f97 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8d84119efad649059ffed533e9646346 : SUCCESS in 7m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f9dd32b04140485cb8cfa03d18d713ae : SUCCESS in 12m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bc5f5235b8c2436582ba22d717de5a8b : SUCCESS in 25m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5c305ba498db465dbe950ed0ee7791a5 : SUCCESS in 13m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9c3dc7d55aab4c088f512291f724b1fa : SUCCESS in 14m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2d87c06f6f4f41cca6924c17cd76be14 : SUCCESS in 17m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/afefe62370574a93b66f99ef145d140f : SUCCESS in 6m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a98e9d1c3677435c83c07203b250eab0 : SUCCESS in 8m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c0d6a7aeccc940e18a1a9007fb8112b3 : SUCCESS in 6m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e8410948f2554335ad1b7494e44a15ef : SUCCESS in 9m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aafb71fdc2c24ad8b5a7bb398a48995a : SUCCESS in 9m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd51763925754a8391639371a75ed696 : SUCCESS in 49m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/028e9a08abb64de0af753e49b538eb56 : SUCCESS in 1h 06m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fb52adda1ad4f7b97e694761c39dfe7 : SUCCESS in 1h 23m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/ff8802433be44e41bae646698b6c96e3 : SUCCESS in 1h 22m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e027a6ac291afbb1a696c97cc86ffeb5fe57a20e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 03:32:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bccd484efd26f493c00e5a41bcca29ea5a4865bb","revisions":{"bccd484efd26f493c00e5a41bcca29ea5a4865bb":{"kind":"REWORK","_number":1,"created":"2020-11-19 11:40:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/763357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/763357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/763357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/763357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/763357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/763357/1"}}},"commit":{"parents":[{"commit":"9bc9f1be6bc40caacbdc7dc258396a2b49bb7568","subject":"update constraint for oslo.config to new release 8.0.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9bc9f1be6bc40caacbdc7dc258396a2b49bb7568"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-19 11:40:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-19 11:40:01.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 4.0.3","message":"update constraint for oslo.concurrency to new release 4.0.3\n\nChange-Id: I3c820c5b49ccebc0841555c99ca08df4771fb438\nmeta:version: 4.0.3\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Change-Id: Ic3baff81d8dd4acd1fe2cf2c86f1780d6735a04d\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bccd484efd26f493c00e5a41bcca29ea5a4865bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bccd484efd26f493c00e5a41bcca29ea5a4865bb"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
