)]}'
{"id":"openstack%2Frequirements~763931","triplet_id":"openstack%2Frequirements~master~I57e2e63899e579e190d32c2e1b9c5ede623b26b7","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I57e2e63899e579e190d32c2e1b9c5ede623b26b7","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-11-24 06:21:12.000000000","updated":"2020-12-04 06:33:48.000000000","submitted":"2020-12-04 06:32:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763931-openstack/requirements/constraints/noclob","meta_rev_id":"08574adc6b47ae6a471be5a20fff693a23943d21","_number":763931,"virtual_id_number":763931,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-04 06:32:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-04 04:18:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-04 04:18:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 06:35:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-24 09:30:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 16:54:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"114f335c3ff05597c2a59d9c3d5ce4b3393aa0e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-24 06:21:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed456c09b1a90e97ac3769cdef770bb823f1fb9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 06:35:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e6e38911294b491281272ac1c4f1357d : FAILURE in 9m 32s","accounts_in_message":[],"_revision_number":1},{"id":"44a5cdd81e2c512cf18069362b7b1b61aa8a59ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 09:30:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51c640f6a5674aaf9af21ea2c9ead475 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/231906ada4844875ab75fd9e05bb2bc2 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a79931d032a45d8a6d055ea551363ce : FAILURE in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6207df4156ab4be0a1d55c16a588a720 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50022ed2f2974457ae3ca9d86c9fbc7d : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2ca5e5b4ff545e0bf4bf6fa9797298d : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a42875d949f44b60a40d4bbc1b4a4a87 : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bd136f3d44c340cd8a8d039ff0f118c0 : FAILURE in 6m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ffa5bebaf39c4b6cbd077fa48af04d97 : FAILURE in 4m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5f7bf55681ee4bc395aebe7eb16240c1 : SUCCESS in 9m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9ed1242e9b63459eb331941166b06e70 : SUCCESS in 9m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2421a922ac4b413991b127a2c84942ec : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d8ea2e376ef1437abad1f382482cc1b5 : SUCCESS in 7m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/26a2832fd66942c5bf35198bd15fb79e : SUCCESS in 6m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c9de54a81eb04b13942b6fba844e3795 : SUCCESS in 12m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e8641c1b99ef46e496283e8744a6fe78 : SUCCESS in 31m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/326121286c0f4727b96ac5d05c49d225 : FAILURE in 11m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de74c2b0a396491d84b64ac74bfdf95c : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/acb66b57e65940c6af6f61161ba1b9c8 : SUCCESS in 17m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/097ffe971b1b4c97b6fef1429a83c7c2 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a1e6c9959a264669bc1f0361059c7f7b : SUCCESS in 6m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9b9a0aafabdc424b9220a4547c3eb4c1 : FAILURE in 5m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f03a52e0e0354001a604aff8798c9cee : SUCCESS in 43m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14d52552e2b246aabd4a5dc4b662b637 : TIMED_OUT in 2h 10m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ec351ff0fea9098b1141edd2b1ae8464a9f30d90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-25 06:32:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"de5643a29789fa13ffbc354dff435387129dd4e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 06:48:51.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/65e6a570b6e04275a00c99c2a90fc84e : FAILURE in 9m 05s","accounts_in_message":[],"_revision_number":2},{"id":"e154d36b9595c4e1e9fc3f847156e58fb079b8fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 09:33:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a6beb4c1e143188187d81b16a02d1d : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54a41a8c24584336a26df7a20a9ee84f : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17a6d662530741f4a0cc2722abc7469e : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36337ff9b7514b09af5886f2a4bd710f : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0cc7cad9a72415dbb74ae13fd4b558e : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61dd11403a014f259806aed5140b7a40 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c661ece789bb4a6fa7121cfd0304400c : SUCCESS in 5m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6a8836b1786a428483695d4ee8a654bc : FAILURE in 15m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eaa787190d1c49699bd7e73833a05963 : FAILURE in 15m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/05cd0187666c41299d01a8d1a5015932 : SUCCESS in 11m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c05d8ee653df400c93474895a505bc7c : SUCCESS in 5m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9108c1251a994c97a2e00fccafd36e20 : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5a99255418b448038d1ca571d237e1da : SUCCESS in 8m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/92d024a312ea434295935139f0c62696 : SUCCESS in 7m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fce533ba7b934964947b11085d1a2c15 : SUCCESS in 26m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6fa47801b8c04d4aaeb9725a460d82bd : SUCCESS in 27m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/62466c29601e4ebc8af06413ba6f34d8 : FAILURE in 22m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/69f1af1c1ef346a4afbb6742e9ae946f : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d490ba606fd74c31acf9b130144995c8 : SUCCESS in 25m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a0c2cbdf6b9941bfbadf15b0e4e1b6a7 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/daa4a04887e541a092034d3cc9e511de : SUCCESS in 6m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/17c8b358788a4c8e8055cf7dfa61bfdf : SUCCESS in 15m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6e8dc1eb3c5472fb18422f019dba1e1 : SUCCESS in 1h 02m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10ef7287f72d4a108b91949baf65e7fa : SUCCESS in 1h 30m 08s","accounts_in_message":[],"_revision_number":2},{"id":"f7c06b9e2099a52d3d5c6ca05f4d71327d6ba7b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-26 06:27:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7e53e586ef969a7f69fcbb3939570b9475ffbcfa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 06:38:48.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ba946533e4194917a887d1ae0132c71e : FAILURE in 10m 31s","accounts_in_message":[],"_revision_number":3},{"id":"0b9a8771d252f53015b4c8d2403669abfaaecb50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 09:13:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8d4ad90fe6f4499bfe133047a108f5a : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d888c7213dde4ff582b329216f2291af : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64cd595dfe1041c18c215781e02877b0 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/111285afdd38450689f825c681de1a4d : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a3afaa8879e46e686a0c43c7f9611f0 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38fe5c0c257a4517afd7e96aea4db170 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/23440e58034c451a91aeca4267c4938c : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d43c6c839a6d4b2ca69e69665662c4c6 : FAILURE in 5m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a688ee9aba1b47e2b59c5ab8f7e80ab6 : FAILURE in 6m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0b582cbb0a6345d58d7a5b4a64b2cbb9 : SUCCESS in 8m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c6ec723e7de14c2db7d9c61afc929098 : SUCCESS in 5m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a507c37094d64e11afa7a37302345e3e : SUCCESS in 12m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c4ac0b3b62054f708184c5770367a2dd : SUCCESS in 7m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ee1024fbddfe4f9aa4bd22908c4c3bff : SUCCESS in 7m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/509b21d61c184669a38aeb68a0bf699a : SUCCESS in 13m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/695a61d2475d406b820d4440577a7e0d : SUCCESS in 24m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ab0ca61a7c5346c69133887ed887bf50 : FAILURE in 12m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e3750cd175d744d78ea64087cba53051 : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce80642cb4e745059556be5d86cd06a0 : SUCCESS in 19m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/044d5e194300410fb4b59ad0ee417a9b : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ea8789d84f84b38884f33d011c6ee7d : SUCCESS in 6m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/203ac38500554190bc80673e8c259b49 : SUCCESS in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af5d59c5f30945b1b58bfe8a19c0a80b : SUCCESS in 1h 02m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0af9be41babf4f1baae34fd29d3cebf7 : SUCCESS in 1h 32m 30s","accounts_in_message":[],"_revision_number":3},{"id":"b4774502c61ee4a534a98fb291301009897a6b3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-27 06:16:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c96492108a647b63203d03116edb80128621f6f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 06:39:42.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2ad70f892dd844d59378aba3b95a2921 : FAILURE in 16m 34s","accounts_in_message":[],"_revision_number":4},{"id":"b3692f65153c5abd82d310631aca5abef0964d47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 08:55:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9fae8e49f4f4d7e9969f49199041a0c : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80dd3b73420f457ba19cf662176a13f9 : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ead62df702ea48d98fb6c053e90df770 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eeedd6eae20c4b17a7d8e966dec408ef : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36c97f281a184c34b43d262c8a4573ea : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6157b9f22ea04448b1211dcc82500732 : SUCCESS in 8m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3a16231344ed4799921b2103cec8e219 : SUCCESS in 8m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f098b7ff9b2c4861ae18153e88d9db93 : FAILURE in 7m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d6cc5f33308f454b83e398275461b5e0 : FAILURE in 6m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/41f67716d33a4e948cb68e6e9374feb3 : SUCCESS in 8m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/93f10bde85784579b3fb188bc4fd6c54 : SUCCESS in 4m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a29da62615424da7a2bed3de1a6282f2 : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83d77fb435b34d69aae465c0cf08a598 : SUCCESS in 6m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a229f7567f6244d5b0fe2fa4eb881e29 : SUCCESS in 6m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/41791bfd0d414069859b6eca60d6aa79 : SUCCESS in 13m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f883079d0f7c4db3a72772f3b7e19292 : SUCCESS in 31m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/de85b515f5d149d9aecdb3325384a8e2 : FAILURE in 19m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/48a0c83d8f3a4c1c91f2e76bdc15a208 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc7995cbba3e4e45b4eef062916748b1 : SUCCESS in 18m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/063fb850983d4b6486137f6b7bedd531 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b43d7610e39a4f38bc98123b5d0ae9cf : SUCCESS in 5m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5d89728fe4d1482e85a2c1d994386494 : SUCCESS in 14m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/301f879923134a19b9aeda45c8cd9d60 : SUCCESS in 1h 09m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f9679e950294895b09b5efbce6c98a6 : SUCCESS in 1h 21m 06s","accounts_in_message":[],"_revision_number":4},{"id":"115d5394b7e8eb27624d68c868865f0898a27197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-28 06:35:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5d9962cc6256da9bd3205d0def2f25e5955dfa2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 06:49:03.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/40101a72fb704e739754c5acc81c9435 : FAILURE in 8m 57s","accounts_in_message":[],"_revision_number":5},{"id":"af8367e262bb6f2df7f86ccf56dfc97059a4dce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 08:29:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4ab4e48bbb43bd8946ca182454f917 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c59efe097d7473e87552462c61ad13c : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45977ed90e494e8e94ef2d672d41ef99 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e543bad323b4314aceb474f008af794 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b756ea8f92c41b2acb12fa1d5335dc0 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/17b7fa51e9d14f85b50e0b442178c61e : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e22608660e5c4a319a3c12f044d0bfcb : SUCCESS in 3m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e2247bdfc7a643fdafa3119ae1428594 : FAILURE in 4m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7608fcbbbacb43a9a0cd29b44604675a : FAILURE in 4m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/060a3b652ea94eea9131dbfce5d666b7 : SUCCESS in 8m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1e7dcfba97694ca6a742d114e1dcdad0 : SUCCESS in 5m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3b293e1b2b254d1db8ca1270b69cf342 : SUCCESS in 8m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b71a65f8f044338a58de40879b47520 : SUCCESS in 6m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/666df0de87844df88339a34fe2901bfc : SUCCESS in 6m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b70f593f01c742b99300dad1f7b26cf2 : SUCCESS in 11m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ad03a29ec16945dda59aadbe9d0c1be2 : SUCCESS in 22m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fefce01e43bc4a06b331508f6a2ce9ea : FAILURE in 11m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/882eba75e3eb4cb3898050caa56396d7 : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14473e63a9734cf19b4eaa879e54d3a3 : SUCCESS in 21m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d7afda76e46a4cdca00f6fee5e7fb646 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ddc23a95bab34cbbb9ebb765fa4511f1 : SUCCESS in 6m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/994f20ac06f8476f96e1835b8c8a2b4a : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b022db5fb674aaba86ce6c9f153c4ad : SUCCESS in 59m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c627b8e97784995a400516995b3d0c6 : SUCCESS in 1h 17m 26s","accounts_in_message":[],"_revision_number":5},{"id":"749a4c313105bbc26cb0d8ade2ed908c453b2787","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-29 06:19:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"883f56c79633f72156bf5c3b2e752e602ebd99f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-29 06:37:20.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dfd25f7a73474ad589de361f632d8e5a : FAILURE in 13m 24s","accounts_in_message":[],"_revision_number":6},{"id":"a6d4753c1ce730d07f77a52f6a67414b267ac847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-29 08:12:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba423c1d37cb459c86e98db3340554bc : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bce52a3615d0430a84cdfdce968df758 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40c4ac1e3e1a43deaf11fd7816a9e67c : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9fd4103136340ed9efff21635cda0cc : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd9d510e92bc473c8bb3aea215439fa3 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9cb29afa4cc7493e9b36080311f4ad71 : SUCCESS in 8m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/286b0574a3164149830bd5e5c760f863 : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9945f4b8349c48f28545f7e8103d4d79 : FAILURE in 6m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8b009f65668140e495bad1f14eb85f69 : FAILURE in 6m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/acce3fb2af104f258931a8b99d1c7bd2 : SUCCESS in 7m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/297d98c3bd9741ee84b4fe5b483d166f : SUCCESS in 4m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5b84c14b9c6f4ae88ea4e5a66ecee43c : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e993989fa3bd423d929302b9e9e09934 : SUCCESS in 5m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f28cc72e1fc2489eb6a33dfbcdaa0e05 : SUCCESS in 6m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d09db952d7a340c3a5eb84bc369b8565 : SUCCESS in 14m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/506a452a75f34ba0b71bec35d1cdd80f : SUCCESS in 30m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a110dfd61c894e8688da6b6fb8468fc6 : FAILURE in 17m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3d736ec0f57941679134364a198653d6 : SUCCESS in 5m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/801cff2b4f834f7bbccf3e95656655d1 : SUCCESS in 24m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/87ecdfb7d3fc4cc4aab6e7150adb865a : SUCCESS in 3m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6389b921ca7642f497719623c5359672 : SUCCESS in 10m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6ac9d8082ffe43898a52edcd0bf6c461 : SUCCESS in 9m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf90af029d5d4f15a06e03815b1307ac : SUCCESS in 1h 08m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02dfcde5c8c5447bb759c60e87c13ce5 : SUCCESS in 1h 15m 20s","accounts_in_message":[],"_revision_number":6},{"id":"403d71d978bcde1219fced715354c95d7279fbdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-30 06:29:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"05a1969acc8c2a726d8e7df6f66b595b8cdb9c2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 06:46:09.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3efd51f38bcb454c8246585ca908e19a : FAILURE in 11m 48s","accounts_in_message":[],"_revision_number":7},{"id":"36c482dc955557a97445f32094942f5f18489a3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 08:04:31.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eddfebd432c24afb81982d7a20b657e3 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fffb5a440144a009372197d0864b24a : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c372f93c97045b29397fd437a8882a1 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ab7c39140d54baf98f842e9b6c82b5c : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f12c5d1bc2440beb5178cfc5d99c3c4 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9ae928b0834488a8287f33953f92e65 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d86b8c9fe14a46f78b8e55ff4c167863 : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/06a71c7143854dbe98f4c11bde47d59f : FAILURE in 8m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/10b6f45a48ed460683b1eb7f32fb193c : FAILURE in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a26f58cc91584096af5a3af545f01691 : SUCCESS in 8m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fbf250de4c7744a3b19a4fc4aec39e74 : SUCCESS in 4m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ed3646b4b90d4a7b9d474b0937360eb8 : SUCCESS in 9m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83fd45513a8047189b76ee899f0b6778 : SUCCESS in 6m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/38655ae0e7dd4d7680dde6fda79a3444 : SUCCESS in 7m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0082c51804394cd8bad83c9ddce523e1 : SUCCESS in 15m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/99e5dfcbabb64dfea0cc40c9029bbfc9 : SUCCESS in 27m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/57c464b3955246cf9a10c69cce1b5cab : FAILURE in 10m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0221494533a44a56989576e38c6f2a61 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d1a2982ecf024f2dbd3231afeadb84e4 : SUCCESS in 20m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/09592259a44a48178a56c35f0d33bdf1 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8dd57fc0cc80461eb1f115fa996e694a : SUCCESS in 7m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a0703e3c3c6e45fabc14ab1d72c78dd8 : SUCCESS in 9m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2ea1022af0146f58297048666e47b06 : SUCCESS in 49m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f880009117e41d9a700406638330508 : SUCCESS in 1h 21m 52s","accounts_in_message":[],"_revision_number":7},{"id":"61e8c18d1ce815c72a85c087273ada4dd939d3b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-01 06:25:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b359598f0a9beca8893b329344cbe2afcac9f4e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 06:41:31.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e35851fccdc14ac286698bb66c511d0e : FAILURE in 10m 59s","accounts_in_message":[],"_revision_number":8},{"id":"86dc7a50c18fbcb55c3011c334d1c564bbe4d416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 08:23:10.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edd225aa04484362bf4e16fbc4f6b053 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/891135b0eed2456585ba10278a29a0d3 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f9b67f74e354138a97ceac518d72c0d : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7ae65fa299e4de49c7f10488f1f1425 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77f3468bf0c34bf3ac10c1813859d195 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d60ee590b88479d81e190b4eb60de21 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/419a01ec55614ba58e35d6b6f9210b5b : SUCCESS in 5m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8964a385e0e545e2b0fe0f0e3704bf58 : FAILURE in 5m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a8540a287744469a9177321e2807cb0a : FAILURE in 4m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/289c0da0306e4745b058c71af174b574 : SUCCESS in 9m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ff08a400e5cd4f9aa3e05753e823c644 : SUCCESS in 5m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fe0f456bcae5402db5078e7fbadab5cd : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df9ffa6a622d4cdd94f718c2834054bd : FAILURE in 9m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c6c1812c0b71478ab95f6f97eca77b7c : SUCCESS in 9m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/810ac2b38a8b4c8ab36cb2a4d0e332f5 : SUCCESS in 14m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/667d4d749b47468ea18611a5db533a4b : SUCCESS in 27m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eb4d5a4147d44a038b32df850876b3ef : FAILURE in 11m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2b9e3575bb5647c789b3f5b2d2d98d2b : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5f60e9c77c96413eb27217c4e3082606 : SUCCESS in 17m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6b6d296671f94793857f943a6f5dcdf0 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6081e13d0f4f4a2bb829f8423b3aa328 : SUCCESS in 9m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/03def694cbe542da9a7277e581c57469 : SUCCESS in 12m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c55deacea86244618fb1c45953fde020 : SUCCESS in 55m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef55b755df704a898d4d1e23462b5134 : SUCCESS in 1h 18m 08s","accounts_in_message":[],"_revision_number":8},{"id":"a68d5fc104f2855cbea770b85ddeb2db5f96efe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-02 07:19:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1600a72b1e07212dbdfc8f668d4ff5fe632184e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 07:30:43.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b6d1c859330f4cf09ac4953ec4c975ac : FAILURE in 6m 35s","accounts_in_message":[],"_revision_number":9},{"id":"07cd031d418e5ac0193242bc9e446ccbb60fc956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 11:22:34.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb23deb97d894dbe8c704bacbc469eea : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/025343035d774875966e8b0be3de7dac : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cd787212f8042ed8bdcac3cc973d6a4 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7bd7e1eb6a74adb80934c41f0822e2a : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e54aa28c5d248b5a078cf9d4df7fc6d : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc4096a339824d6d859951198909ccfc : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b0e0dbb0ffb94c2db0acfb709f0fbd2d : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6ee20474f2fe402abf221ab6b4a112bf : FAILURE in 5m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f1ffc45a0e5f4796b6a447efafdca94a : FAILURE in 4m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8ab77ae47bd642678ede0406c398bc95 : SUCCESS in 7m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/04016fc77df24f359e7d19f5e2d0440c : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4350bab46fe2441799ed6a9d62896e95 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e9c5c8f738c49dc9bf48c6a9334990d : SUCCESS in 6m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1517ca17da3e49719c3ac0897c177bcc : SUCCESS in 6m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b10d03be3f1346ba8ba17fc531c43d28 : SUCCESS in 13m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/62ff37d4c0454dfb9cf0705b9d23c06c : SUCCESS in 25m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/725f98b5d27f48d4ab43d7a49e84db85 : FAILURE in 11m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/992039c27feb4de889b78fd867a9f743 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9195fe3a000a4f7b999443c353eada7a : SUCCESS in 17m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c6cf5a26bf204d18873d1d2dc1ff5902 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec0afed605f5484bae095ca7be8663eb : SUCCESS in 6m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/00748ae569764a1bb05e8cfcac57c5c2 : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eef0ed0bd9394fb19225165671aeb532 : SUCCESS in 1h 16m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a01006175ec34fe29cbac50c720df050 : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":9},{"id":"2212957154f9cc83dee906e083aa33e2ff696870","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-02 16:54:50.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"cc7408444eadbe9e95214dddc8fc29fbf49483f8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-02 16:55:23.000000000","message":"Patch Set 10:\n\nI expect this to fail on the sphinxcontrib stuff, which was previously held back","accounts_in_message":[],"_revision_number":10},{"id":"1f3a8282c21243790fa33ef639d9eb14f98bca8b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-02 16:55:28.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":10},{"id":"a8f2fb506864958c7ecb8642beb2bab40a5d4e55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 17:16:14.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4381a6056c4548969e418d6849d9fb5d : FAILURE in 15m 06s","accounts_in_message":[],"_revision_number":10},{"id":"9bd8cfdc217ecabb9af078acef9fc331228cd209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 00:49:20.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6035f25a808145d8b97d8c26204a6d9c : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5ce0c8edf35423aad3848d4ca0f716b : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb7bba979e9a4a6d9be57551ad09d709 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89e4a3ee20cd4246a45954bc75b49935 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/917ced0add3e44cb8d03d2d19cc88e10 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eee9a4b02a9f436b8ff6079fb4ceaac5 : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/419d74ee5c364ad7a999bcbc295a474f : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/72ad573b150744c7af2ab98ac8d5c209 : FAILURE in 9m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ad4cd00787664a14beb13ff1b9dfa4a8 : FAILURE in 8m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8cbfdf80895c4d108dd258ac1a87832d : SUCCESS in 8m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5246bec54a8a4b24bc41649ef3c68792 : SUCCESS in 5m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/21617b95e5554f2b9a5a966ecea70b32 : SUCCESS in 12m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b2bf571e6194c75a5afc28a9f45ea63 : SUCCESS in 6m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9f90af6c8415458193f094295daf3b02 : SUCCESS in 6m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a464c80992524a8bbd0eb4c80b0f5d4a : SUCCESS in 20m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/214724d08afc4e478c03e4ef068a4b38 : SUCCESS in 31m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b17180bb1c2f4903a942439b0890f6fb : SUCCESS in 12m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/882355bc64f14f28b28b9896538656d0 : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1176a6bb2c09430597fce9dafdafc8b4 : SUCCESS in 18m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/18568181cb23423c989a3931c9864f83 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f76db8fa361849f0ad6607205c53aa63 : SUCCESS in 6m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9976943f4f704267beba77d5af275487 : SUCCESS in 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7677e85c9614e15b457b3ae14717ddd : SUCCESS in 1h 00m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f06c8336e31f40dd9ea32ac76776afc1 : SUCCESS in 1h 19m 07s","accounts_in_message":[],"_revision_number":10},{"id":"8c316f949f15aeeec10421c7259715e82623eb2b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-03 07:37:29.000000000","message":"Patch Set 10:\n\nSHOULD be working, but isn\u0027t...","accounts_in_message":[],"_revision_number":10},{"id":"76163137049d0548afffa057189c881db7397422","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-03 07:43:23.000000000","message":"Patch Set 10:\n\nyep, fails tox locally, but...\n\n.tox/py38-check-uc/bin/python3.8\nPython 3.8.6 (default, Nov 16 2020, 20:34:15)\n[GCC 9.3.0] on linux\nType \"help\", \"copyright\", \"credits\" or \"license\" for more information.\n\u003e\u003e\u003e import sphinxcontrib.runcmd\n\u003e\u003e\u003e","accounts_in_message":[],"_revision_number":10},{"id":"a903d849665d1e814f3a746d75e97387a9c3bb06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-03 22:54:38.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"712995969e0c932e9bb58e94dd0c637fc1900013","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 23:22:19.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c821a546944342b994a7cf04ef0bedbd : SUCCESS in 18m 36s","accounts_in_message":[],"_revision_number":11},{"id":"8c4de383c56699e56f6619781f31184254790cc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 04:14:11.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/937e7f08fe1140609fb2c15eba48f243 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23eb2f4888d14b5690e4d9056fdbf005 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cbeaf4d06a24e17a79122316ace77e6 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4290b9b8a13e410b8cca47e012c6be28 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68f5c1f9b5da45218848fb31def75807 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a4b3a600eb9414b94d873c9740a764f : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/baa866d10f8b4f2db85fc1f361690300 : SUCCESS in 10m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4a82558187dd4e3e97c5a45481c63507 : SUCCESS in 7m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a14ca3ae88344caab9d526449f24aa8e : SUCCESS in 21m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2025efaec251401ba29b7a35cb90e377 : SUCCESS in 15m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7cd6d1428f1d4a7aacd2b1cc016fe4aa : SUCCESS in 10m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/510d33da521e4f8ca83ff5f6ca4654ff : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d2f63979364c4a4e87b4d2a397213f8b : SUCCESS in 6m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d1dd3d785b05441dacfd3d9d92bacce3 : SUCCESS in 6m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b06c14e9cc6143299e00f297ec1dab41 : SUCCESS in 12m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a279732a6e1b4d71953d90c0fa023ef2 : SUCCESS in 27m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/81886bab19394cdfb95d02d5704d8bba : SUCCESS in 11m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/732bb79df60048c795deccabc1da2ec5 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bf0583d6b137463f9e4d91818c8e75f5 : SUCCESS in 18m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a2997a5225fd4bb7bdf72e1044b3d8c1 : SUCCESS in 9m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/43c6c7d8f38f468192f8e67f02cef6b1 : SUCCESS in 13m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/76a15f28e35e4d9ebd772c3ea9c3e6b6 : SUCCESS in 8m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3920c46665e94a438386c970d6d9e166 : SUCCESS in 1h 00m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a81e3c882914582b5ac5438aa4270a2 : SUCCESS in 1h 38m 01s","accounts_in_message":[],"_revision_number":11},{"id":"69ea878f01082b9338102ff35db3812134cafed6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-04 04:18:20.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"1f1bebc0bea76228edcbf6cb4c9b14f0f36cc5a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 04:18:45.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"310877025ca014c76bb9495556018e9805f0aa34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 06:32:06.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/166db2def33b4687a3f5f9f5d3c41f42 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01ef7a8a669b4ac6835c187cb4f1d4a3 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcc6d49b4d6a4b1e99bdf30a0124c083 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fa8304fdd0c4344ae158326f9843caa : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40bd310214c3417296d54996cb430d32 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81a90f4580c84903a3b3b8e2b8086728 : SUCCESS in 3m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/053e9563b07349589550a0c58bc5c230 : SUCCESS in 8m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/53be0cbb0c77486ba925f4b7ffc2865e : SUCCESS in 16m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cab187fcae924bee861cf8cfe79580fc : SUCCESS in 8m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8bdfc80c23c646dfbe566f6e6abf08e9 : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2557f43f8ec64a15b2d7fcbf5cda6199 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83da9a9930ca47b1b3cb0ea67e81202b : SUCCESS in 6m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ed122fd219a9427eb56eaa0397ce1898 : SUCCESS in 6m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3ecb2610f21346179a430f682be2598f : SUCCESS in 13m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/48c6545ffcc54539a0268e576b37bb16 : SUCCESS in 29m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8e74a32fad2646849022a705df1d914c : SUCCESS in 11m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d349620c52144cddb8b6cd500d05b018 : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/907d2a6641a741c796ebc82df63dc898 : SUCCESS in 27m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b6937a483f5247fc868ebdb07fa74ec3 : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/95f7d377f3ea44d4a881ce4f30d82fcf : SUCCESS in 13m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8422912b788d47ebaef715f7b8d9bc88 : SUCCESS in 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c97f8dbbc024834878b3a7a82908e5c : SUCCESS in 1h 39m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8e44ab5a8e84eb4b161e9521d642b1f : SUCCESS in 1h 40m 43s","accounts_in_message":[],"_revision_number":11},{"id":"acd69d8fb046d796ca59a104cf7f26491bcee378","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 06:32:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"08574adc6b47ae6a471be5a20fff693a23943d21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 06:33:48.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05c3531e862445a89523befd0df785e7 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"3a1d6acab0fd8bca7604952d3c7292f38925f699","revisions":{"3b18bccf5f51988fcfa3baa69503496e6be2ca1d":{"kind":"REWORK","_number":1,"created":"2020-11-24 06:21:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/763931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/763931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/763931/1"}}},"commit":{"parents":[{"commit":"937ccadec9b94ebb111bbde8d4a9b5eb6690365e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/937ccadec9b94ebb111bbde8d4a9b5eb6690365e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-24 06:21:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-24 06:21:09.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I57e2e63899e579e190d32c2e1b9c5ede623b26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b18bccf5f51988fcfa3baa69503496e6be2ca1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b18bccf5f51988fcfa3baa69503496e6be2ca1d"}]},"branch":"refs/heads/master"},"a05f0a57a608d228bc4f8935401b15904b8ccbae":{"kind":"REWORK","_number":2,"created":"2020-11-25 06:32:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/763931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/763931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/763931/2"}}},"commit":{"parents":[{"commit":"937ccadec9b94ebb111bbde8d4a9b5eb6690365e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/937ccadec9b94ebb111bbde8d4a9b5eb6690365e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-25 06:32:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-25 06:32:01.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I57e2e63899e579e190d32c2e1b9c5ede623b26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a05f0a57a608d228bc4f8935401b15904b8ccbae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a05f0a57a608d228bc4f8935401b15904b8ccbae"}]},"branch":"refs/heads/master"},"1d434b58b853ebc90821ea172834d5857c3a23b7":{"kind":"REWORK","_number":3,"created":"2020-11-26 06:27:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/763931/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/763931/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/763931/3"}}},"commit":{"parents":[{"commit":"4c7d26b3e3b3bef05443d66661b6d7b53f755879","subject":"Merge \"update constraint for oslo.upgradecheck to new release 1.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c7d26b3e3b3bef05443d66661b6d7b53f755879"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-26 06:27:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-26 06:27:24.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I57e2e63899e579e190d32c2e1b9c5ede623b26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d434b58b853ebc90821ea172834d5857c3a23b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d434b58b853ebc90821ea172834d5857c3a23b7"}]},"branch":"refs/heads/master"},"27fa076c7f0becf9049503b3731d3193ba0d6f92":{"kind":"REWORK","_number":4,"created":"2020-11-27 06:16:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/763931/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/763931/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/763931/4"}}},"commit":{"parents":[{"commit":"4c7d26b3e3b3bef05443d66661b6d7b53f755879","subject":"Merge \"update constraint for oslo.upgradecheck to new release 1.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c7d26b3e3b3bef05443d66661b6d7b53f755879"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-27 06:16:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-27 06:16:46.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I57e2e63899e579e190d32c2e1b9c5ede623b26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27fa076c7f0becf9049503b3731d3193ba0d6f92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27fa076c7f0becf9049503b3731d3193ba0d6f92"}]},"branch":"refs/heads/master"},"6bb0be2eb852f22a912e4fa34254176bc02b6770":{"kind":"REWORK","_number":5,"created":"2020-11-28 06:35:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/763931/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/763931/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/763931/5"}}},"commit":{"parents":[{"commit":"9b47a48a92683da51228c72d0645f25cca6ab772","subject":"Merge \"update constraint for oslo.policy to new release 3.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b47a48a92683da51228c72d0645f25cca6ab772"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-28 06:35:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-28 06:35:11.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I57e2e63899e579e190d32c2e1b9c5ede623b26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bb0be2eb852f22a912e4fa34254176bc02b6770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bb0be2eb852f22a912e4fa34254176bc02b6770"}]},"branch":"refs/heads/master"},"fa8fdd68f88193d0cebde3a5686f73671c1e9ca2":{"kind":"REWORK","_number":6,"created":"2020-11-29 06:19:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/763931/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/763931/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/763931/6"}}},"commit":{"parents":[{"commit":"f74b67ac0a5a42e7f933c5cc75201e835d3413a7","subject":"manual update because bot is not running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f74b67ac0a5a42e7f933c5cc75201e835d3413a7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-29 06:19:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-29 06:19:41.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I57e2e63899e579e190d32c2e1b9c5ede623b26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa8fdd68f88193d0cebde3a5686f73671c1e9ca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa8fdd68f88193d0cebde3a5686f73671c1e9ca2"}]},"branch":"refs/heads/master"},"a9ffcee6b2babef0fb08aeffdfaf537a784501e5":{"kind":"REWORK","_number":7,"created":"2020-11-30 06:29:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/763931/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/763931/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/763931/7"}}},"commit":{"parents":[{"commit":"f74b67ac0a5a42e7f933c5cc75201e835d3413a7","subject":"manual update because bot is not running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f74b67ac0a5a42e7f933c5cc75201e835d3413a7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-30 06:29:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-30 06:29:26.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I57e2e63899e579e190d32c2e1b9c5ede623b26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9ffcee6b2babef0fb08aeffdfaf537a784501e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9ffcee6b2babef0fb08aeffdfaf537a784501e5"}]},"branch":"refs/heads/master"},"47f896d5abb28493d284ce5a5d71ccf6df075903":{"kind":"REWORK","_number":8,"created":"2020-12-01 06:25:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/763931/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/763931/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/763931/8"}}},"commit":{"parents":[{"commit":"733472f4621641a0a0b85a896f07f3d2f2ea19ce","subject":"update constraint for python-troveclient to new release 6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/733472f4621641a0a0b85a896f07f3d2f2ea19ce"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-01 06:25:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-01 06:25:37.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I57e2e63899e579e190d32c2e1b9c5ede623b26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47f896d5abb28493d284ce5a5d71ccf6df075903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47f896d5abb28493d284ce5a5d71ccf6df075903"}]},"branch":"refs/heads/master"},"f5995c33b648d49974cbb2795c542c3b8c5706e3":{"kind":"REWORK","_number":9,"created":"2020-12-02 07:19:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/763931/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/763931/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/763931/9"}}},"commit":{"parents":[{"commit":"6dec47e08a0ee95ab787cd4779cba8a1f25713cf","subject":"Merge \"update constraint for python-magnumclient to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dec47e08a0ee95ab787cd4779cba8a1f25713cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 07:19:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 07:19:52.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I57e2e63899e579e190d32c2e1b9c5ede623b26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5995c33b648d49974cbb2795c542c3b8c5706e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5995c33b648d49974cbb2795c542c3b8c5706e3"}]},"branch":"refs/heads/master"},"a0a3849252528013be140af8ea3ca85363dbf243":{"kind":"REWORK","_number":10,"created":"2020-12-02 16:54:50.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/31/763931/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/763931/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/763931/10"}}},"commit":{"parents":[{"commit":"6dec47e08a0ee95ab787cd4779cba8a1f25713cf","subject":"Merge \"update constraint for python-magnumclient to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dec47e08a0ee95ab787cd4779cba8a1f25713cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 07:19:52.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-12-02 16:52:33.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nneutron-lib has it\u0027s own review\nmock held back because nova\ndocutils held back because boto (iirc)\npkg-resources removed (not sure what\u0027s bringing it in)\nsetuptools readded (we need to either fix the job or remove it)\n\nChange-Id: I57e2e63899e579e190d32c2e1b9c5ede623b26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0a3849252528013be140af8ea3ca85363dbf243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0a3849252528013be140af8ea3ca85363dbf243"}]},"branch":"refs/heads/master"},"3a1d6acab0fd8bca7604952d3c7292f38925f699":{"kind":"REWORK","_number":11,"created":"2020-12-03 22:54:38.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/31/763931/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/763931/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/763931/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/763931/11"}}},"commit":{"parents":[{"commit":"6dec47e08a0ee95ab787cd4779cba8a1f25713cf","subject":"Merge \"update constraint for python-magnumclient to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dec47e08a0ee95ab787cd4779cba8a1f25713cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 07:19:52.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-12-03 22:53:58.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nneutron-lib has it\u0027s own review\nmock held back because nova\ndocutils held back because boto (iirc)\npkg-resources removed (not sure what\u0027s bringing it in)\nsetuptools readded (we need to either fix the job or remove it)\n\nChange-Id: I57e2e63899e579e190d32c2e1b9c5ede623b26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a1d6acab0fd8bca7604952d3c7292f38925f699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a1d6acab0fd8bca7604952d3c7292f38925f699"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
