)]}'
{"id":"openstack%2Frequirements~764022","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Id5ca21336f302f5eddc3f569ebfcb4bfd8882313","project":"openstack/requirements","branch":"stable/victoria","topic":"victoria-neutron-uc","hashtags":[],"change_id":"Id5ca21336f302f5eddc3f569ebfcb4bfd8882313","subject":"Add neutron to stable upper-constraints","status":"MERGED","created":"2020-11-24 16:38:29.000000000","updated":"2020-11-30 19:02:29.000000000","submitted":"2020-11-30 19:02:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764022-victoria-neutron-uc","meta_rev_id":"769d6a5ea96625a2915db0172202f60c41bd4045","_number":764022,"virtual_id_number":764022,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-30 19:02:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-11-30 17:28:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-11-30 17:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-11-30 17:29:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 17:08:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-24 22:02:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-25 10:38:19.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2020-11-25 14:37:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-30 17:28:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"80a8c0813838ce247c831fdc2abfa4f069b2de7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-24 16:38:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03c10fc4232b627c5db780ba9d048eacbd020933","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 17:08:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6628fafae2cb4245ad26eabecb76ba85 : SUCCESS in 12m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1b640f820e5bb6f3f36cc026b6012096f53d6fd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 22:02:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a47d8d6ca1f48bc84be0b62025f9c2c : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22deab95417d41589aee06197bef9577 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8df4e044b11847dd9e269e68198a7fbd : SUCCESS in 6m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8fd7849ddbe6461b904039d813558460 : FAILURE in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8b2da6864c7b4eed83e959c91480ee50 : SUCCESS in 3m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/360ca58ce3824ff79880be72ff3b1cc4 : SUCCESS in 6m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/56880a8030344df884d8e8fe3d5d16b8 : SUCCESS in 6m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1e496faa735940188c5cacf36617f622 : SUCCESS in 8m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/44388740d4ab4259813d3ca9c00c1086 : SUCCESS in 4m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c166648553c74242b2a0cfd2c0702e59 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d60ba066b80d4b808b138ce8a0d97e65 : SUCCESS in 7m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1cbf215b2db0433fa029f8fb77347545 : SUCCESS in 6m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a45e8a9ff8054539967921f6c7c321c1 : SUCCESS in 14m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f137b9cb265b4455a8ba611d78f27608 : SUCCESS in 29m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c210ba61992f4f2087a494f919014d8e : SUCCESS in 11m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/49e95ad5aaab4999af0ffe71f593e111 : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/244baf56c20744da8757ec86ad825620 : SUCCESS in 16m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e3209b10c30146e091ca05ae9e9ddc2e : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84322c2d9b8e4f50b465896b87bc1542 : SUCCESS in 6m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/53c296d4851549c5a1d288adfc95b003 : SUCCESS in 10m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/980c3099ee1a4640b6040f7f03cd329c : SUCCESS in 1h 05m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5113f9f0e0641679851235ebed890ba : SUCCESS in 1h 23m 51s","accounts_in_message":[],"_revision_number":1},{"id":"af2d450f1e89fb29fd0ea59b3e48730691ffb8c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-24 22:20:36.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"10ee7d92e10a5572cabb0d2b01aa7a6fef6eccc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 22:36:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a5a2fa2ead34470bb89595b04e8a4f61 : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":2},{"id":"c2570066c1afdfff61d883df9976dd0cd000ac86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 00:22:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fffe1c296b864f67811e241446f31a60 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be0ea8ac1f134fba97019c8953ac023f : SUCCESS in 8m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bccb095656b54c579e0899c3b26be406 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66456cde09f041b09127d8340d91e8d0 : FAILURE in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ea26bfd2709b414fb8cb9f06842761b5 : SUCCESS in 4m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6ae1c1d5b90a4a0fb2636e48f7d1f8ee : SUCCESS in 13m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ee5c18f99495439cb922f0c56d6ea16d : SUCCESS in 6m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3f8f03df174f4f18b5383de791c449bd : SUCCESS in 8m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e2f17a99f08d48c7b7cd8f0d5c5cd97b : SUCCESS in 5m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7d425540645d45ceb4220879a509fa30 : SUCCESS in 12m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6149956bda844215864eeeefc7766710 : SUCCESS in 8m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ff8c72a70f59440ab2a27079724071e6 : SUCCESS in 6m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b763c7e1bc59417d9c9fd84a914e1220 : SUCCESS in 14m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9cf500114314467ab82a57bdc507a702 : SUCCESS in 25m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/92a21064d33c4865a4e9989895b89934 : SUCCESS in 17m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5137f7a130264b029c80f841828b9468 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3e9b7872c3a45e2a39a2f8a66696fee : SUCCESS in 17m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2a43ec052bd74ffba7fe3067392b97d5 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/31f62b025ce04adaa124da51f64d67c5 : SUCCESS in 7m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/04e7edc48657459b919736cb3aa636cc : SUCCESS in 11m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25cfa28e18f2468fa520168da9e312aa : SUCCESS in 58m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/322d22c9a5104315a0f64c0d2b232abf : SUCCESS in 1h 22m 50s","accounts_in_message":[],"_revision_number":2},{"id":"7151a32c5b5b8933a328f9bacdc8867e34133ebd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-25 10:38:19.000000000","message":"Patch Set 2:\n\nas I see neutron is in blacklist:\nhttps://opendev.org/openstack/requirements/src/branch/stable/victoria/blacklist.txt#L30","accounts_in_message":[],"_revision_number":2},{"id":"20d18cb3614ebf839deb16a12b83ab4444f275ee","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-25 11:52:52.000000000","message":"Patch Set 2:\n\nYes I added Andreas to ussuri patch https://review.opendev.org/c/openstack/requirements/+/764021 let\u0027s see how it goes","accounts_in_message":[],"_revision_number":2},{"id":"c5b92a68b117ec68446ed7f87b27766835359606","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-25 14:37:03.000000000","message":"Patch Set 2: Code-Review+2\n\nMatches what we\u0027ve done for other stable branches.\n\nThough looking now, it appears stable/ussuri is running uncapped. That one should be addressed too since that means these projects using it as a dependency are not testing what they should be right now.\n\nAnd of course, the neutron team should really get this fixed so there is a lib that everything is using rather than relying on the whole service project.","accounts_in_message":[],"_revision_number":2},{"id":"26e39ed9a0db2002a878a9d2071a78cc2bcd386b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-25 14:38:29.000000000","message":"Patch Set 2: -Code-Review\n\nOh right, needs to be removed from blacklist to add it to upper-constraints.","accounts_in_message":[],"_revision_number":2},{"id":"8f5740618b469b02099a3539b5a7223cdf16d401","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-25 15:02:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e1e38343d9812300e2f727e7391c0dd4e13964b4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-25 15:05:02.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review+2\n\u003e And of course, the neutron team should really get this fixed so there is a lib that everything is using rather than relying on the whole service project.\nSadly neutron-lib effort has mostly stalled in recent cycles...\n\n\u003e Oh right, needs to be removed from blacklist to add it to upper-constraints.\nDone","accounts_in_message":[],"_revision_number":2},{"id":"6cda873b76b14a30be134987d86e2c7cfe7a8ade","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 15:19:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5e84adb08a434b77bd8928a5c6709313 : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":3},{"id":"0fe8ec1960f7543247a9fd9ccf52dc3889c8262b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 22:10:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7ef7d4735340fd8cb6be4a3d83dd88 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/546ff672ddc64fa3888bf3782cb3d012 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dda9132c67c547cbbeee2152b3109a8d : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b836ae3aa7343aeaabfff57b09086be : FAILURE in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f18018607054d0cbf532318b99a860e : SUCCESS in 3m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1fdc79c3fbb944eb93b0442f0cc60936 : SUCCESS in 8m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/57d712c2c31a4649b3f7648751f651f8 : SUCCESS in 6m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c9c5a73249d2406780638e1ef9abde1c : SUCCESS in 9m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d81412819e164d5284f87650b2a3ea41 : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f03b4ebee20f459da6e0d67432c347a0 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c7305a6706441e4b66c9b49b8b6980e : SUCCESS in 7m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6da55dc2ab694ce08889f682b68a6d54 : SUCCESS in 7m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ca3df47868e148a5935d92b9d734dec6 : SUCCESS in 14m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/152b5da94d574b25b7b88f785fc34719 : SUCCESS in 24m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/877ab62865b84668869216b177c535b5 : SUCCESS in 12m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c1281e351234b4a8652acd32cfa8e6f : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c50b52d5ed534f7d89940557864143a1 : SUCCESS in 16m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/788adf5a8fa44a9ebfb78cd914cb011d : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8be37f09c44546999973036dc9ffecca : SUCCESS in 5m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/369729642042462bac22dcf9c22bdd6b : SUCCESS in 10m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1ee8285ca0a43f588e33b4ff003f533 : SUCCESS in 56m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc25b78e78844b9bb50b5aa036e3bb4b : SUCCESS in 1h 34m 51s","accounts_in_message":[],"_revision_number":3},{"id":"c4f7f13d9d3efa78d668b4960164247989daefb2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-26 07:16:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e983c2087dafab1d2cd4d9df4860da7afef47985","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-26 15:19:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"301f405190ffc325bd0805594118f7510002192d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-26 15:20:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"61ea25ff56121a0113241c3bbb7614e1b9a1faa4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 15:36:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fca12c71330e41288a12b75027a68b84 : SUCCESS in 10m 14s","accounts_in_message":[],"_revision_number":4},{"id":"4a054e15c80928a712923dd67335979e014de2b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 18:43:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb832105ca5041d489fd5003c7211906 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11f33825eb3546bfae0b788bf0df3a46 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18858c1b4da74846911c958a04542da3 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74bc767471de42f5aa98be8b51ba2d05 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/01b07306a0b148eb91a1666da55b33c8 : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fa5234b4653046bfaa2144455394d652 : SUCCESS in 8m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aafb28a25a1243969bdde98c9a0b4d75 : SUCCESS in 7m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/62c0b22f421b441f9bede172ff29efa0 : SUCCESS in 10m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/770459bd15c049389eca86db33f23648 : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bed6dfb5f08c4dfdbd0029f2790176b3 : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48454c92673549b99454e7b801133555 : SUCCESS in 11m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2d46746e68b04bb1b5a26c64f996bf59 : FAILURE in 18m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/baf01a2e0ab548cbaf271b71eec2441f : SUCCESS in 16m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d557a29425b24fa198cb666910b92f93 : SUCCESS in 28m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/05a5e21c6998439e911445d8f7fc36cf : SUCCESS in 16m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d7e7b0cba9b4470895cc2d842181ca9b : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/84d4ed577ffd4405b231912353a88a85 : SUCCESS in 16m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/35d6c3944f8446678ed24b9f61964c05 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc71452152e54302ad542609b8bcdee3 : SUCCESS in 6m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/43972ccc7e4d4c0c894f55285c8c7614 : SUCCESS in 10m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ecf2aee3c4446778574a2869f99dcd1 : SUCCESS in 1h 01m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a857966d46524dc98f5c5c0e279bbb40 : SUCCESS in 1h 22m 28s","accounts_in_message":[],"_revision_number":4},{"id":"8b3cbddd1ce2d14796436ac37ecbb5c547b29850","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-26 23:39:38.000000000","message":"Patch Set 4:\n\nrecheck ironic test_upgrade_and_version","accounts_in_message":[],"_revision_number":4},{"id":"f8f627e5aa563c56f929b51c5359118d40e046f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 23:53:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2a9dc5b09a7d462aab1aca97ba4ac646 : SUCCESS in 9m 21s","accounts_in_message":[],"_revision_number":4},{"id":"9d7b6657e1ff381a003ec93f055a6ac5cbfd49de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 01:04:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2f6051407a34cb68f0141a7a5756a95 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1387bfe4b5294ca7b58618efd7cd3d36 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39a8ecb48cc34c63b02ecf21f93ebc3c : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4f1a0c7e1b443cf95eadf48962f2a01 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aa3e496fe5c54483ac88ae87f74b3669 : SUCCESS in 3m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c65ed738f69d4f0eb0330802993ae473 : SUCCESS in 6m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ae5bc044b7a040fe933a11eca35b0856 : SUCCESS in 6m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/17b43f13f0f64c24a8ff769d23f9fe66 : SUCCESS in 8m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4862e1e6010a4943a98da49d55b33006 : SUCCESS in 4m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/36a5fabffb004799bd0fe69e0c8e05eb : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f724e8eed9447e7aa6315cea1c41de9 : SUCCESS in 5m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e03e3e41d8f94ca7914aa1b5fcf148c4 : SUCCESS in 6m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/326ae1d99a174fa3ac2c94e7ba86e1e4 : SUCCESS in 13m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cee2ec2df89242a8a1e1bf3c1e887031 : SUCCESS in 29m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a6d53169a3d24b65955bd1d1c2d31388 : SUCCESS in 11m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/36203e0944854670bc9e1e1783db71c9 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d577deccc5546038b4bf25f2be2b61d : SUCCESS in 16m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/77fa36537c794ecfbc5bd106f1867920 : SUCCESS in 3m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b3cb30341d084c88a1ce301fcc32d2b2 : SUCCESS in 6m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9b1003d959904cd1957b91d1675d121f : SUCCESS in 9m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c6e5a874f2d4030b2d64cec6fe4efc0 : SUCCESS in 47m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65503f6b307c4b2281b4477bafe10a4a : SUCCESS in 1h 19m 08s","accounts_in_message":[],"_revision_number":4},{"id":"281db6040be77260e249bf58a42c6b770872342b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-30 17:28:17.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6d50804ff43e94ad72bac0781af176a3f0cabd29","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-30 17:29:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"aa1e29d1744162a2460db752223dea1f88b080d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 17:29:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ca0ff6dbceb3ac4c6628e51201ace944cad46864","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 19:02:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd5219e69e724c1fa66992dec215a1b4 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b799427abe9b4f7eaa37b321b76a3ecc : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be2ba4dd04204a8dbaf2da4b7b0c58c4 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c113734bfb984715ba7cd5d1eed7423a : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eaed22e0aa9142a18fce7d3f926202ca : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f9fdccce225947f0ad8408b7d24699ce : SUCCESS in 7m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7d85b9d3669543c8a35906df6dd0ddc5 : SUCCESS in 8m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1b074f908189471db9d859fc64106ea6 : SUCCESS in 10m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ca8e63221ec744518097f7e654e09e5f : SUCCESS in 5m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cfdaba71e82e40cfb532fb8220b43b4b : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fb523ef4229647e88bbf5c96a3451fdf : SUCCESS in 7m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/eb5046a705bc4197b6b71e515b131ea1 : SUCCESS in 8m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5eac327b251f46b2be19d5b7b3b0f304 : SUCCESS in 13m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/692e39681df04857850e1cf232efe600 : SUCCESS in 24m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cbef534b56d045ee8705d4be545ae162 : SUCCESS in 12m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2b08c408bd9e4720a60e99eb97e3f9bc : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/96420b8fee7d4b2db0e420334b185cf3 : SUCCESS in 17m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c3d9b96389534eec9b97572b449ee15a : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e051a9fc1c0b4bbfb85cc0c239c9df1e : SUCCESS in 5m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/681709111a3e48f0ae4e108bcc1596f8 : SUCCESS in 10m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/955d0e382e914e5990b4d942103a1426 : SUCCESS in 55m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18c156a535284822aa1181ebd57d6f18 : SUCCESS in 1h 26m 27s","accounts_in_message":[],"_revision_number":4},{"id":"769d6a5ea96625a2915db0172202f60c41bd4045","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 19:02:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"36710831f3154ec0844fb2a38317815435da46bc","revisions":{"a5e15354bb4ee8b953b8368dd65efd449ff02262":{"kind":"REWORK","_number":1,"created":"2020-11-24 16:38:29.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/22/764022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/764022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/764022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/764022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/764022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/764022/1"}}},"commit":{"parents":[{"commit":"0c6ba9b1ee76e5670d201bed97fa44611b93ba8b","subject":"update constraint for oslo.config to new release 8.3.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c6ba9b1ee76e5670d201bed97fa44611b93ba8b"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-11-24 16:37:49.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-11-24 16:37:49.000000000","tz":60},"subject":"Add neutron to stable upper-constraints","message":"Add neutron to stable upper-constraints\n\nSome projects, notably neutron plugins, are using the neutron service as\na library. There is the neutron-lib package, but apparently they have\nnot moved everything needed over to there yet.\n\nThis is being added directly in stable branches only. Master needs to\nget the latest version available, so these will need to be added at each\nrelease so the stable branches use the correct version of neutron for\ntheir respective stable release.\n\nCloses-Bug: #1903689\nChange-Id: Id5ca21336f302f5eddc3f569ebfcb4bfd8882313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5e15354bb4ee8b953b8368dd65efd449ff02262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5e15354bb4ee8b953b8368dd65efd449ff02262"}]},"branch":"refs/heads/stable/victoria"},"6d11273cb0cce1a25d47598672512047b51458e0":{"kind":"REWORK","_number":2,"created":"2020-11-24 22:20:36.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/22/764022/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/764022/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/764022/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/764022/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/764022/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/764022/2"}}},"commit":{"parents":[{"commit":"0c6ba9b1ee76e5670d201bed97fa44611b93ba8b","subject":"update constraint for oslo.config to new release 8.3.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c6ba9b1ee76e5670d201bed97fa44611b93ba8b"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-11-24 16:37:49.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-11-24 22:20:21.000000000","tz":0},"subject":"Add neutron to stable upper-constraints","message":"Add neutron to stable upper-constraints\n\nSome projects, notably neutron plugins, are using the neutron service as\na library. There is the neutron-lib package, but apparently they have\nnot moved everything needed over to there yet.\n\nThis is being added directly in stable branches only. Master needs to\nget the latest version available, so these will need to be added at each\nrelease so the stable branches use the correct version of neutron for\ntheir respective stable release.\n\nCloses-Bug: #1903689\nChange-Id: Id5ca21336f302f5eddc3f569ebfcb4bfd8882313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d11273cb0cce1a25d47598672512047b51458e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d11273cb0cce1a25d47598672512047b51458e0"}]},"branch":"refs/heads/stable/victoria"},"01161128e5754a18b25ee76e77d0de41fd91f967":{"kind":"REWORK","_number":3,"created":"2020-11-25 15:02:58.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/22/764022/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/764022/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/764022/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/764022/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/764022/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/764022/3"}}},"commit":{"parents":[{"commit":"0c6ba9b1ee76e5670d201bed97fa44611b93ba8b","subject":"update constraint for oslo.config to new release 8.3.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c6ba9b1ee76e5670d201bed97fa44611b93ba8b"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-11-24 16:37:49.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-11-25 15:02:20.000000000","tz":60},"subject":"Add neutron to stable upper-constraints","message":"Add neutron to stable upper-constraints\n\nSome projects, notably neutron plugins, are using the neutron service as\na library. There is the neutron-lib package, but apparently they have\nnot moved everything needed over to there yet.\n\nThis is being added directly in stable branches only. Master needs to\nget the latest version available, so these will need to be added at each\nrelease so the stable branches use the correct version of neutron for\ntheir respective stable release.\n\nTo allower upper capping, this commit also drops neutron from blacklist\n\nCloses-Bug: #1903689\nChange-Id: Id5ca21336f302f5eddc3f569ebfcb4bfd8882313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01161128e5754a18b25ee76e77d0de41fd91f967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01161128e5754a18b25ee76e77d0de41fd91f967"}]},"branch":"refs/heads/stable/victoria"},"36710831f3154ec0844fb2a38317815435da46bc":{"kind":"REWORK","_number":4,"created":"2020-11-26 15:20:30.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/22/764022/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/764022/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/764022/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/764022/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/764022/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/764022/4"}}},"commit":{"parents":[{"commit":"0c6ba9b1ee76e5670d201bed97fa44611b93ba8b","subject":"update constraint for oslo.config to new release 8.3.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c6ba9b1ee76e5670d201bed97fa44611b93ba8b"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-11-24 16:37:49.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-11-26 15:20:13.000000000","tz":60},"subject":"Add neutron to stable upper-constraints","message":"Add neutron to stable upper-constraints\n\nSome projects, notably neutron plugins, are using the neutron service as\na library. There is the neutron-lib package, but apparently they have\nnot moved everything needed over to there yet.\n\nThis is being added directly in stable branches only. Master needs to\nget the latest version available, so these will need to be added at each\nrelease so the stable branches use the correct version of neutron for\ntheir respective stable release.\n\nTo allower upper capping, this commit also drops neutron from blacklist\n\nCloses-Bug: #1903689\nChange-Id: Id5ca21336f302f5eddc3f569ebfcb4bfd8882313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36710831f3154ec0844fb2a38317815435da46bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36710831f3154ec0844fb2a38317815435da46bc"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
