)]}'
{"id":"openstack%2Frequirements~764149","triplet_id":"openstack%2Frequirements~master~I097554c1e0a10a6cb27f07d6df124577350747d4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I097554c1e0a10a6cb27f07d6df124577350747d4","subject":"update constraint for glance_store to new release 2.4.0","status":"MERGED","created":"2020-11-25 10:17:08.000000000","updated":"2020-11-25 20:13:57.000000000","submitted":"2020-11-25 20:12:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764149-new-release","meta_rev_id":"87c55839cc31140a4232f6b240ad8b7e3298a415","_number":764149,"virtual_id_number":764149,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-25 20:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-25 18:38:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-11-25 10:21:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-11-25 18:38:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-25 10:21:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-25 10:38:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-25 14:46:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-25 18:38:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"0d3cc1b27ee14e9189f2efb814e8adb4f7d34970","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-25 10:17:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05e99a032b1af03ddecde38cde4ca83231db67e3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-25 10:21:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0278f7998004d9be6eea91e46c342da859b0acca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 10:38:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0ddc23af72324d92aaae5dc8139efe0c : SUCCESS in 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f01d6833f78c43926e8974214a34e817dcb8b35a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 14:46:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d492f7ce6fd64986be2e1973ec12aa87 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3670062b907d49f2b7dc4bd5cce16e20 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3e01210de6b4f8da346335262a8947b : SUCCESS in 12m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6e09bcbe88a4887a37c2e8fd18862a5 : SUCCESS in 12m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acb9c23fefbf49a3a28a15025d957880 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe7f8cf17a854c688442216304b370f7 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/76807ab89e0f4ef6b2f446f9608368a6 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f2a42e9cca81474cbcdb7ab5a3820e54 : SUCCESS in 9m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9c68d9de28a14ec78a781129ad692d11 : SUCCESS in 8m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e8603be94610439f8914a73506e0869b : SUCCESS in 8m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/db1a40b6b99a4305aecf48ac74c310e6 : SUCCESS in 5m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/62b162d59bf44af5a8a190a4cae9f3b0 : SUCCESS in 7m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5036d3523c22488fb96f60076f8dd3ec : SUCCESS in 5m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6b82c2350385420b824a72ac90b61bfd : SUCCESS in 5m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/80c8c26184014eff9a515933ba62fa90 : SUCCESS in 13m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/74e83dd4283745aabc73accb7d8d3268 : SUCCESS in 24m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2976044f85574a22956174f41bb6af67 : SUCCESS in 13m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8d54671b273f4265976ab2817d304f00 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8aa78362a2d5486b88f687fb8b17c4ac : SUCCESS in 18m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/32951f370e084b0daab3acd562563daf : SUCCESS in 10m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b36960495a048b19742432b638a2844 : SUCCESS in 6m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a41416b7ee3c47ff8be06bffbe83e2c4 : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5849189605c049bc84884c58df2bd438 : SUCCESS in 51m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a97dfe44c904230b82fb4bdbc848419 : SUCCESS in 1h 26m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f2eca0b030fd90f61c92cbbaca1c24ddb895ea11","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-25 18:38:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1504ee8690ff10c5d56a2bdffecfd87f9b6d9f1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 18:38:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18f6c2342395588af680d696a236ac0f4122d0c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 20:12:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/682a190089ab4295b0ab958950545d28 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c04ec1915b3f4bec89ba750875fbfd97 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c1f66e902a54ecdbb3a9e0387d59ad0 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a5bf22a57364e25beb0b4940631f1d7 : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc5945f8497c4353a4a2f42ec06826cc : SUCCESS in 6m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e83a1d733574eaeb59aa77f934dc6b6 : SUCCESS in 7m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f82811e332074f74810c974f12cb8243 : SUCCESS in 11m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a6a72ce553f6434f8518ff402e7c5f2d : SUCCESS in 7m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b001fdefbf234c65a7a9c9f23230e04c : SUCCESS in 9m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0e34e982998a468b845a83a659ea5d08 : SUCCESS in 5m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/db6342c8c9b249d2968a61458510a9e9 : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/abd379f2e51747d1aaae0df0883e3cdf : SUCCESS in 6m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/66f39bd1d78d4839b7ba2a49d9a5fa62 : SUCCESS in 7m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/11a9adcbb512477bb698d226f679421e : SUCCESS in 12m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f3adab61e32b4b86839f90780eefc0be : SUCCESS in 22m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e7a3633fc5d746e9a56fd29f6d2ec19e : SUCCESS in 12m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5ad4152a73cd4d50a67dc3ea2a303bca : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ecc552fb67c4e9c834964af78fb3c6b : SUCCESS in 24m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cca009aa1b6941cab47013b939bd54a8 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f91a48491b14723984f5bd428e127fd : SUCCESS in 6m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/361d8e45724848df90310dd879743072 : SUCCESS in 8m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a7fa1a7613c48688f18886506d2940e : SUCCESS in 57m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8964c0d5cd6d4c5dbc12999af8b0c125 : SUCCESS in 1h 24m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0e3e8e9fa7f584e89d5828ac290edf6603ade421","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 20:12:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"87c55839cc31140a4232f6b240ad8b7e3298a415","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 20:13:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/303bc603ab854482865083786eae8ce5 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2fe6ef1aec5fb3d40c46657e41d10523d53786da","revisions":{"2fe6ef1aec5fb3d40c46657e41d10523d53786da":{"kind":"REWORK","_number":1,"created":"2020-11-25 10:17:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/764149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/764149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/764149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/764149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/764149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/764149/1"}}},"commit":{"parents":[{"commit":"937ccadec9b94ebb111bbde8d4a9b5eb6690365e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/937ccadec9b94ebb111bbde8d4a9b5eb6690365e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-25 10:17:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-25 10:17:05.000000000","tz":0},"subject":"update constraint for glance_store to new release 2.4.0","message":"update constraint for glance_store to new release 2.4.0\n\nmeta: version: 2.4.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Commit: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Change-Id: I21dd27f6f917f89f38a38cacd502ede5c1adba40\nmeta: release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I097554c1e0a10a6cb27f07d6df124577350747d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fe6ef1aec5fb3d40c46657e41d10523d53786da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fe6ef1aec5fb3d40c46657e41d10523d53786da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
