)]}'
{"id":"openstack%2Frequirements~764214","triplet_id":"openstack%2Frequirements~master~I7f5911edf31725328b8581276b029bae5a38d9b4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7f5911edf31725328b8581276b029bae5a38d9b4","subject":"update constraint for oslo.upgradecheck to new release 1.2.0","status":"MERGED","created":"2020-11-25 16:09:39.000000000","updated":"2020-11-26 00:09:03.000000000","submitted":"2020-11-26 00:07:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764214-new-release","meta_rev_id":"80aa9dc7e2e42fc3c844e11662131438840b1f00","_number":764214,"virtual_id_number":764214,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-26 00:07:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-25 18:38:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-11-25 17:34:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-11-25 18:38:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-25 16:28:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-25 17:34:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-25 18:38:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-25 22:24:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b233d847b01af394db43b38c9e6bd2a8719b77ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-25 16:09:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee9eb92e72ec65d8f3d007494229af9c8cd8bf1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 16:28:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fbb7e41ad9b94844ba98cc7120c68f48 : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b4c387ab30b521c33b49ac08d86306af4ca69995","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-25 17:34:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5b7c912010346b865a4ff5074115fbfddb7539f3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-25 18:38:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f2cf122530348713485289275d3366f896ea0b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 22:24:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0412e85131654d3181e2d3ad7fc00c35 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e20cf0c6335497aa97e60f00181f5e5 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcb6a0f6b1eb4ca88c1373760e34a268 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fb1d228082245729932d17f925c9c4a : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7257f1ee366a41d2983da06626d25f49 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/afd25030d1f74d21b12649cf8ca8e1e4 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f9e03b243b1b450ca4c772948f88aec5 : SUCCESS in 3m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/87673745f9ad42118188110372963ee3 : SUCCESS in 7m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b206af29cad3429a85fbede7191ac16d : SUCCESS in 7m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d2d7af7387f143fb867748e55dc1e93f : SUCCESS in 9m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f53d456fa1ec429e895ee6dbeb5e2367 : SUCCESS in 5m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/87b5b0ad1e1f404789e721eecff2fe4e : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df90f9323038406fb73e90e340deb937 : SUCCESS in 6m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b280fa9d57864b26b3ba810622e08599 : SUCCESS in 6m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/25d07401bd6b42b78d21e013203215a5 : SUCCESS in 13m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bc984020b5f74e3d911e62622573795e : SUCCESS in 30m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e4bdd1989241471cb5763c4823acbd8d : SUCCESS in 11m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0104e52abe6f4ff0865a82fcd2fb2eff : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/406d46487e4e4fbd9b168f375d0f988f : SUCCESS in 16m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/955be396bc684b4095a78bacd528661c : SUCCESS in 4m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c06a87cf42d4486197942f4675970615 : SUCCESS in 5m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3046e1cc37d74444a1ed45814992abad : SUCCESS in 14m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b16ee30dbded460fbb0e3433a58eba7b : SUCCESS in 44m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6360fc419a4147d28bbd50078c1c343c : SUCCESS in 1h 20m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2f5de0305eda679742942ddade5bc4059f0be4e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 22:24:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6362fa651a74fbcaf6fe269c30599685d44d629a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 00:07:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8051fccbc134a9888f9ec59531c6e8c : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb39b95035a04ba39118c43baa87976f : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9932ececd9d4f5fa32911b958db4a0f : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26ebbbc9310348c091f0e6278dd6ecf7 : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f1de2d729dd48e2aa36248d7ad5bc92 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/70c0b21e88e144ad8ade5951d41fae5e : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3aea2add4a5f443e8aebdb27ff9588e6 : SUCCESS in 7m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4389d734d1074bb2bb319ef2c077a86c : SUCCESS in 7m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a575d7cb857e4dfebdb5de1f7091936a : SUCCESS in 9m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b91b532d8d1f4fbc876eaf9958a5b93a : SUCCESS in 4m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7de4b5a583bb48bb95ba5e4c0a8944fa : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/503a9cabb76c46538635c7541b7f67b4 : SUCCESS in 6m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/21b8e28bdb4a4f138e7f571e22270d4d : SUCCESS in 6m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/26b23ef58fe6482083b3fc5dc0569d81 : SUCCESS in 13m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/89ff529dca5d40f984dd648e321f6d4a : SUCCESS in 24m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5e819e4b86c744eb903ee5ffe36155ad : SUCCESS in 14m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/09e5c71b426141938e94dc19dafc6e17 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b03539c4863b48a9a3ad1ba576b44566 : SUCCESS in 18m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bf6c9090f39f4334a3252932601bc1ed : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24ed823d581d45e895e5df559fca31a2 : SUCCESS in 6m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/046302bfc2cf4ea88e619e270391a517 : SUCCESS in 10m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0c7835af21e467d93df885112f7e348 : SUCCESS in 57m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5e8ede6614544ee81a5eb75789eaa47 : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9bd75c162c534899f4379a65d19a04634738c728","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 00:07:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"80aa9dc7e2e42fc3c844e11662131438840b1f00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 00:09:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/451810568dc9488ea954146ff2ed90df : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5ddedf3a5f5c99b717b5a050a62c2a7a60bb670","revisions":{"d5ddedf3a5f5c99b717b5a050a62c2a7a60bb670":{"kind":"REWORK","_number":1,"created":"2020-11-25 16:09:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/764214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/764214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/764214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/764214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/764214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/764214/1"}}},"commit":{"parents":[{"commit":"937ccadec9b94ebb111bbde8d4a9b5eb6690365e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/937ccadec9b94ebb111bbde8d4a9b5eb6690365e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-25 16:09:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-25 16:09:36.000000000","tz":0},"subject":"update constraint for oslo.upgradecheck to new release 1.2.0","message":"update constraint for oslo.upgradecheck to new release 1.2.0\n\nmeta: version: 1.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Commit: Ghanshyam \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Change-Id: I6415ba757112ec84d9ddb27cd0742674a95517c5\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I7f5911edf31725328b8581276b029bae5a38d9b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5ddedf3a5f5c99b717b5a050a62c2a7a60bb670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5ddedf3a5f5c99b717b5a050a62c2a7a60bb670"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
