)]}'
{"id":"openstack%2Frequirements~764216","triplet_id":"openstack%2Frequirements~master~Ie7390021db1f6171327c3deea1dd8c80b71e043a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie7390021db1f6171327c3deea1dd8c80b71e043a","subject":"update constraint for oslo.policy to new release 3.6.0","status":"MERGED","created":"2020-11-25 16:15:42.000000000","updated":"2020-11-27 11:52:09.000000000","submitted":"2020-11-27 11:50:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764216-new-release","meta_rev_id":"13baf5739bca6d9ca04a8b8d79490a3c92230f13","_number":764216,"virtual_id_number":764216,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-27 11:50:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-25 18:38:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-11-25 17:34:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-11-25 18:38:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-25 16:33:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-25 17:34:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-25 18:38:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-25 22:51:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-27 06:17:33.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"bd68cf692549a43fdb04ca890d63b12a7b45bd7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-25 16:15:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ab0461e2be92af02cccdfce0efa9bb97d1dc84b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 16:33:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8ad06b70112242fa8648bfa2467c9b57 : SUCCESS in 14m 08s","accounts_in_message":[],"_revision_number":1},{"id":"88f6b9055804b52cb8e64598626c750dd4477311","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-25 17:34:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f54dc48d7f18cf1babf940febba8d4b23a8ce781","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-25 18:38:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c742a77b80600a284e8035959291a507cb5bace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 22:51:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26320a1b2afe4a9aa561b87bc02cf393 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1d877aac0654700af6e40e8f719014d : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecb74dd404004b3198ccaad269a3aad6 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9b6c1e861f44825953803a156fe1f13 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04b535c2eef142f6bd4ecdc7edd0edd8 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bffcecde722a44df9ba280703937ceb1 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e61bd27bd78b451cb1c203fb54702f0b : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c196f8549993423aaee62cce297be610 : SUCCESS in 6m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c729f961f43a44f5ab35c886fd734ca3 : SUCCESS in 7m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/081ac20f5453444d98117d30329de823 : SUCCESS in 7m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c2a45e0fe273487b9caa98f041e88e10 : SUCCESS in 5m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/782e5c92f1bd4680b99b1d7beeccb953 : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/70332a841b1e4cecb282436e2960848b : SUCCESS in 6m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fefd6ce2d9874bc2b886cc636b0d364d : SUCCESS in 5m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/598efe938b3b496695c9b37c0ff98494 : SUCCESS in 12m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7891c9c3dc3e4b55ace9d8405926a955 : SUCCESS in 29m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9c7a6b94a1d74f72b43d13561155305d : SUCCESS in 17m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/801e72d78b7149e387d3e2359f5fa507 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ab0f64fb86d41928cea5f03aeeb0645 : FAILURE in 32m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d5c0a7ff3d7e4b48b5e30d892b6d141c : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5dfe925af2fc4a71935e1772663483bf : SUCCESS in 6m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/74ad5bc2eb3143cbaf52624893e5d4d6 : SUCCESS in 9m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18b7bd692a85475abe8119516eb0168b : SUCCESS in 59m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80a21c0d00264a638396f2cd56f8be54 : SUCCESS in 1h 24m 52s","accounts_in_message":[],"_revision_number":1},{"id":"9153601333fefc8d61517f1355968fa836f376d8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-27 06:17:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6ee219821a88a11e96cb271e9cd792e07226da4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 06:51:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fd25591d1ed347ec82dcf07c28233cfb : SUCCESS in 28m 42s","accounts_in_message":[],"_revision_number":1},{"id":"528c48baa343cd934bde3e1f9b5224dbff569238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 09:12:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1177adbff8e14bf0b715135d81d9171f : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c05f1ffc20487aaf28efe5d0a4e476 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ebeb778a44944c4b95cf3eaf0128b3a : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd304194c6dd4ee5bf29039f6e0bfbea : SUCCESS in 8m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62fb4be5388f46b6bc68eaad0d92b51d : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12d2ea208b33484fb03ffc5642341cd9 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8695ab002d86478ca7b227c5890d0379 : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9e59aff4136b401fbc102fd348250ec1 : SUCCESS in 8m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/af0573f2a9014651acfa87a44a7b4a8b : SUCCESS in 7m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c7a23e77567b40bb8bb60ae85f28baae : SUCCESS in 13m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fd45414f05ad4bed9f4ec4ad784c44e7 : SUCCESS in 8m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c66cf6555684322b470c66a36de8454 : SUCCESS in 13m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1922c7e3f65246d782af297079708c43 : SUCCESS in 12m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/382739fb64d54596af909f339d3f9fad : SUCCESS in 6m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3bdaca5199dc4c2ebdf82d6a07fa4143 : SUCCESS in 13m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fa7cbc1e08c54331a77410b3e8d0fc73 : SUCCESS in 29m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d4a62ad7bb1444ba9355be5130596606 : SUCCESS in 18m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/21aa4965e31e43b297b7f7dd700ae2c4 : SUCCESS in 5m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/28547b2e02f942218d4033f0bbd8614e : SUCCESS in 22m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/790b9a6105f14b8594aadf395e7cf56d : SUCCESS in 7m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/765c0943c5b946758f535c597c98f24e : SUCCESS in 7m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/44f265f29fd04fd3bbbbcee28b292050 : SUCCESS in 10m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a92cc95099e340ceb32362871f7b04ef : SUCCESS in 55m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9592b486f3e64ba7b423b80ca4882d9f : SUCCESS in 1h 31m 11s","accounts_in_message":[],"_revision_number":1},{"id":"dd570132d7a66ca133107261f4572946acd7d32e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 09:13:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d6d9ba58239d19b24cd9002f55688785f16621e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 11:50:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c5a8892311402f8b796f13bb80898c : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e6ef5664c4443568c55e0da7ec8e9e5 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ea9a2ebef874ef4af32283aeec23960 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/469e43fa409b43049043c1e6c79f1fdb : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ba249e7805841eea78aefbe87e83a27 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4985a447be104560b2fe02047e051a84 : SUCCESS in 3m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/71549a71cc5f41aab52fe4e365beffec : SUCCESS in 7m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2a48e8df298c47038dba58eee9737447 : SUCCESS in 12m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e8aa7a4d458b463998847818287ec1e7 : SUCCESS in 13m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0aca1d0083c746218195918ba479c2ac : SUCCESS in 5m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/92ceb0d5f3dd4ce7b5ca56afcff875ce : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0248a47dba234653946ce774e1f38299 : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/12f175aa2971490a86ea806c12bf558c : SUCCESS in 6m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4ca5072a989e443c8a4e7a29d73ce3b3 : SUCCESS in 16m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3c5be31c93924edfa351461a4e9e4729 : SUCCESS in 29m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bee8d7242c304a6a9222ff769a62823d : SUCCESS in 13m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/685f96945d3940fa82d6304aec9a890a : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/865b2291679d4e4383559e2fe2131e9a : SUCCESS in 18m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/364d9a6cb0764870a2367299fa9af5cb : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c9e4e688e8140d2a48285b97245c83b : SUCCESS in 6m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fd74cf9a2e684414b761a41d8e456fab : SUCCESS in 9m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e20a4860de174544b50df5ea05fadca4 : SUCCESS in 52m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f320f65adce42488eaf1a4d177639a7 : SUCCESS in 1h 21m 51s","accounts_in_message":[],"_revision_number":1},{"id":"747ce9659d8a26f9ac9eb9c7d4ecc3caf475d94b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 11:50:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"13baf5739bca6d9ca04a8b8d79490a3c92230f13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 11:52:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c57e1c7a29aa4fc0bddf4172f0e1c5cf : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"961c5686e090dc7f5b538f41a17ff065adbe8ec9","revisions":{"961c5686e090dc7f5b538f41a17ff065adbe8ec9":{"kind":"REWORK","_number":1,"created":"2020-11-25 16:15:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/764216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/764216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/764216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/764216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/764216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/764216/1"}}},"commit":{"parents":[{"commit":"937ccadec9b94ebb111bbde8d4a9b5eb6690365e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/937ccadec9b94ebb111bbde8d4a9b5eb6690365e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-25 16:15:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-25 16:15:39.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 3.6.0","message":"update constraint for oslo.policy to new release 3.6.0\n\nmeta: version: 3.6.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Commit: Ghanshyam \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Change-Id: I6415ba757112ec84d9ddb27cd0742674a95517c5\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie7390021db1f6171327c3deea1dd8c80b71e043a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/961c5686e090dc7f5b538f41a17ff065adbe8ec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/961c5686e090dc7f5b538f41a17ff065adbe8ec9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
