)]}'
{"id":"openstack%2Frequirements~764374","triplet_id":"openstack%2Frequirements~stable%2Fstein~I1768b386ebc9cfb89e418bf9a8f7ef97adcf3227","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I1768b386ebc9cfb89e418bf9a8f7ef97adcf3227","subject":"update constraint for neutron to new release 14.4.2","status":"MERGED","created":"2020-11-26 17:37:40.000000000","updated":"2020-12-06 11:01:33.000000000","submitted":"2020-12-06 11:01:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764374-new-release","meta_rev_id":"36f161c88ec6ca12f5703c59712acbf08af35725","_number":764374,"virtual_id_number":764374,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-06 11:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-11-30 17:30:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-11-26 18:30:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-11-30 17:30:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-26 18:30:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-26 20:18:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 17:30:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a5720abea12b66cbc1d24bea6e6766fc03365e12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-26 17:37:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7028b5e97ea7178eb74f444c5089f59d3a679355","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-26 18:30:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"746924e5fdf2b33fa000b2357ed72d45efd33367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 20:18:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ecbd608a3cd48a1bba55374c4eca98d : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4a2bd34d0b04753a3fc26cd2f09adcc : SUCCESS in 4m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/82ab1dedbb404d5ebad20baad013c75c : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5b2e217f7524b4fb33988457d3bca6c : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/20f0b17b1ebb405db870fd8f694e35d0 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ef3a046ac0d4ab89edbc88739e9af06 : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/50fe9ee6b74a4aefadc0562c0c017242 : SUCCESS in 7m 08s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/efb3b3483a6741589b719850fc9e757a : SUCCESS in 7m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/35486e77639e4688bdf729820b46dca1 : SUCCESS in 7m 15s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a0ba4b54adb849b1b9a36cd8df289e62 : SUCCESS in 8m 17s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/e7ffce4cbd474b1785515cc5fc5f897e : SUCCESS in 10m 34s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/21a03ff52e704303ac15f3b26737570b : SUCCESS in 5m 47s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/a1ee4a1dd4674042a83bf2c5b74f222a : SUCCESS in 4m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/80951649274e41a38c1af7596872bf18 : SUCCESS in 10m 37s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/14e6dbdb19b74be093f7229d96d25ebe : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9196c47e52384c2a84aa75e180d6c22f : SUCCESS in 8m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2adb4ed8d98049ca897385b0a5b657e3 : SUCCESS in 24m 50s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/49b4f078c41740cb8a1dd772412c3c98 : SUCCESS in 29m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8f464afc3c1a4da8aa69862c8d4eeb8e : SUCCESS in 20m 42s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/0adf9589c7c94356ae3a26f23bf020db : SUCCESS in 20m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f69bdefe3c234d07b0f17176d89e6b1e : SUCCESS in 12m 13s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/0459e256703a45c9b0adcb9488564117 : SUCCESS in 13m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9438f06b3d9649ff9dbda3e9129739de : SUCCESS in 19m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/be85ae81943247f69b5b6ef3428c460a : SUCCESS in 8m 40s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/32ce211af9ac4c07ba107564767d0fb0 : SUCCESS in 7m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c74b9c59e2d44e1ba2110501ebd95313 : SUCCESS in 1h 08m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9c1df7d53e7944318e71439726542b91 : SUCCESS in 1h 35m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4ace67a87873492da71bc85c08044251 : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":1},{"id":"937bab443d19a5b2e030d42dffe6477fe3ba4110","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-30 17:30:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d862f95bacac5aa0f3dbf3d56963d61cac814b87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 17:30:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc1a7ce47bd6eb34ab830ced25a9852a2e1e14d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 19:02:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef943139c36440839ef7700619717f60 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab84f6642fc440bcb0e81b81cf985df9 : SUCCESS in 4m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2f37883fe22d4bdf8d815efd91b09250 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db8b1715f051436e82ecacbbc938e9e5 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf51e96c7d31449fa8ad8b47f0f89dd9 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d7881a41dc3b493d8e93adf622e392f4 : SUCCESS in 3m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e0bfc34e11504b108ed96e1f2f74d72d : SUCCESS in 18m 11s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/627aac64551d444997eaac67476ad9ae : SUCCESS in 21m 09s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/166753b9af494aa791acdf56474d807e : SUCCESS in 10m 12s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/3dda25105f0b46d28661215039e1816c : SUCCESS in 9m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/889bb4eb1bc4439a93e2620a5b0548ca : SUCCESS in 5m 34s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/25932e22d06c4591a2d6a5fdac11514d : SUCCESS in 5m 22s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/038c873e74834a5eb4d2a28132187590 : SUCCESS in 11m 06s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/cb98b0b116fe4d64b74508155a7c56bf : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a7ec86c5d9c14a8abf4a1d9547c3e35f : SUCCESS in 8m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3bbd3ac749494d308e119bd7ccdd4d3f : SUCCESS in 14m 03s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/3ecb3a2922e74a3a8eb9f4c1a4cc5cc2 : SUCCESS in 40m 48s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/20a9e98d4d294aaa80a56d9b328d42f2 : FAILURE in 21m 28s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4e5dbdb827a44b888995e0bdeb7d6f05 : SUCCESS in 18m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d61f1775eaa14dacacffcd59311b07aa : SUCCESS in 11m 42s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/6ad30ecfb06244189c9f0e02a0b62715 : SUCCESS in 14m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5cdb9d5416bb4dbea3172f891874310e : SUCCESS in 13m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bf753d29e4944ab7983cf57e02ddb1b8 : SUCCESS in 8m 38s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/0045b6aeac644205ab0f00ee429277b8 : SUCCESS in 8m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0d97b09979e14fbf84326dd73bcfe3d6 : SUCCESS in 1h 02m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a065e26a64054d2dbb56099a40646cfd : SUCCESS in 1h 26m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7664ae06d95349d2b0db8371aa7a4493 : SUCCESS in 1h 00m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ead8d2c8495ac3605fc185fdcee16df82d1a3226","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-06 08:20:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1076d26022a82cbebfdb7d5c26cb71b121b64edf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 09:38:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2efc73cb70945cc8504cf1ad635c447 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04dac952b5d4453d9f73ed6fd06251b1 : SUCCESS in 3m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ec6c8cf0c5d54504868e7089f4ddc505 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/762ff658f2534154b7beb922f2cfbcf9 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f87e2a11c9747d2a127673f12cbdec5 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df1161b225934b07b8e02afd0168ada6 : SUCCESS in 3m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/02deb78388f942b78edbb979f058ccfa : SUCCESS in 7m 16s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/496d375610364f759d8e0d47f9fbb772 : SUCCESS in 7m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9f11d3a3f9a1410fb7bcf6d92a96b479 : SUCCESS in 7m 33s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/84fd456b242e4d7e90a24f81568e3af2 : SUCCESS in 8m 52s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/a246a1d7141b441a9ef484588a97a61b : SUCCESS in 9m 10s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cfa22d1ded474e16ae75001849f84a82 : SUCCESS in 5m 40s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/4b3c74baecbe4a6ebd2df2ffb6dc7bff : SUCCESS in 4m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/faeaa53fe0d440858d61cf02e773b2bd : SUCCESS in 10m 47s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/57b55c6f3ce14366914eb87f9a090f9d : SUCCESS in 9m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f89afa33b57e465f94bba9607951566d : SUCCESS in 7m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b8f92a4dc91640e1b8c574120a29449a : SUCCESS in 19m 38s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b4f8befaab7e431585ca955e50e8f562 : SUCCESS in 26m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/87a7560bbe934c9f90bf249efc0cac7c : SUCCESS in 21m 19s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/a613d04daafb4ee2999b442d792d18d3 : SUCCESS in 21m 30s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0b46e8e6212b4805b315f04e3e3c7139 : SUCCESS in 13m 42s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/93cbbdb095aa400abea144bc517fefb9 : SUCCESS in 13m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e246972f69e4679a6ec909031524b26 : SUCCESS in 16m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ed3a89deb29f4441a6083e7b472153b4 : SUCCESS in 8m 28s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/8153f76d6d614397aa2505638a1a865c : SUCCESS in 8m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40f2e996e2dc48e5b5d7787f5462eafa : SUCCESS in 57m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3e5654a792fe4e2599bd5cc9095b482a : SUCCESS in 1h 15m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dc9d5ce196a84f8eb73701dafa0ced05 : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":1},{"id":"01aa1034bd308fdac6139581089d9ccb0bbc59ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 09:38:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d912a16552eb8b6337a57eebdd709eee5dc6b50a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 11:01:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33dfce6840ad430590f6bae28de779d9 : SUCCESS in 6m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64e3eabc3a6b467180442abbbfaf54b8 : SUCCESS in 3m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6fd581279c034be3a13804c803d6b33b : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a13e86524c7454fa7463075e9361ae2 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9ad8d143e944ab4bb757bb9d069c78f : SUCCESS in 6m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae88de39bf1a4fb5b042d5381e8dccd5 : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7ce2279f8a3a46b58bfcc21fab20531f : SUCCESS in 7m 50s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/543a29d80c1c4c3988d25035c95a8ca7 : SUCCESS in 13m 33s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/30aa75e6f39a45f09d46d8410a6a1123 : SUCCESS in 8m 22s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/798ec2fcacc143b3858b026693f6349a : SUCCESS in 9m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/15934f9dbff94b408176718607545087 : SUCCESS in 5m 29s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/8107c38fda734117b339b61df645b5ea : SUCCESS in 5m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9aab34eb2b4148378ef25747ee74611f : SUCCESS in 8m 27s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a8f2b7f8a3834d9f95aa74a004d992b5 : SUCCESS in 11m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e617b370fbd4640acb08fd6536d6b7c : SUCCESS in 6m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/75c62957b67b4dba82c1a465b7dd8b48 : SUCCESS in 26m 02s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/8251a0610a6d4f1dab4f7c51f70bbcf2 : SUCCESS in 17m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/95918b9e83304ea593748fddedccad77 : SUCCESS in 19m 44s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/428fce8959a74138837fa24c2dbb0984 : SUCCESS in 27m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/309bc789ba3343bba5113dfdee95b2fa : SUCCESS in 11m 04s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/1e35293b48a142e0937011433d307a65 : SUCCESS in 11m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ed0eba6a0ac49aa9201963facf3e291 : SUCCESS in 21m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b0ad5240b09142ebbe1bda2a1cf83faa : SUCCESS in 7m 34s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/5f347830c663471c83a1c46750e0d04f : SUCCESS in 12m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/348765a6f30c4f1aa3178516e2d6b5ff : SUCCESS in 1h 00m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7e84144c2fb742b1ae988a9d0fda361c : SUCCESS in 1h 20m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8766e5c5ec47445abedb4526cd019e37 : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":1},{"id":"36f161c88ec6ca12f5703c59712acbf08af35725","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 11:01:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45797e0e4336a65af6d8a7d31dde507b08dd6570","revisions":{"45797e0e4336a65af6d8a7d31dde507b08dd6570":{"kind":"REWORK","_number":1,"created":"2020-11-26 17:37:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/764374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/764374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/764374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/764374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/764374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/764374/1"}}},"commit":{"parents":[{"commit":"e80e3ac77db4d6ecbe958299da302fc02f0b563e","subject":"update constraint for horizon to new release 15.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e80e3ac77db4d6ecbe958299da302fc02f0b563e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-26 17:37:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-26 17:37:35.000000000","tz":0},"subject":"update constraint for neutron to new release 14.4.2","message":"update constraint for neutron to new release 14.4.2\n\nmeta: version: 14.4.2\nmeta: diff-start: -\nmeta: series: stein\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Change-Id: I06d0ecadd63592c5c764bcb21512cfc7a857bd7d\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I1768b386ebc9cfb89e418bf9a8f7ef97adcf3227\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45797e0e4336a65af6d8a7d31dde507b08dd6570"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45797e0e4336a65af6d8a7d31dde507b08dd6570"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
