)]}'
{"id":"openstack%2Frequirements~764495","triplet_id":"openstack%2Frequirements~master~I0e15eeefa3fb8af0e307b8724dc3d749fa3870d0","project":"openstack/requirements","branch":"master","topic":"manual-update-20201127","hashtags":[],"change_id":"I0e15eeefa3fb8af0e307b8724dc3d749fa3870d0","subject":"manual update because bot is not running","status":"MERGED","created":"2020-11-27 23:29:08.000000000","updated":"2020-11-28 10:06:28.000000000","submitted":"2020-11-28 10:05:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764495-manual-update-20201127","meta_rev_id":"14ed794e2bcfeb8089a4b57d2c8cefbe82f63cf8","_number":764495,"virtual_id_number":764495,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-28 10:05:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-28 02:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-11-28 02:01:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-27 23:49:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-28 01:19:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-28 02:01:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"af613f7031bc104bd5075feb58cbb995c98f7b13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-27 23:29:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24ac7921e1e8dc594cfc5eb5914147f8b7f673f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 23:49:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f7d73e89223b4d33ba8801eaafa1ccf2 : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5782718b1a4dcc30c7c83f32df8756026c36069b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 01:19:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dec86daf62784f038b5b874e364eac03 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e17f3fd5665437280dc6a54fde3accf : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaf9925aa5fb459c8b14e7fc93ed4a56 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94b8dec2c0394acc9f88d962bbb209b4 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32e24ec09ce14b0a901d9b25151c9b0c : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be822ffb725a4daf9ef8e9d28f09c25b : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/742b3ccec70d4b0a8d048a960cd5e396 : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f3286b3398554b4e9dea6655ecf4721a : SUCCESS in 7m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1486fe4ed74d42cd81f53dd8265cc476 : SUCCESS in 7m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f8edd2f38bbb4d519971279d1aeeebbd : SUCCESS in 8m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8fe47eafab534dee902d4b711dd751d7 : SUCCESS in 4m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/42eb05042b03451f83401bc69fad9bd3 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/402c15320d2544219005bd19dad591b0 : SUCCESS in 6m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0e448f07f8744eeea51aa7abd99fec44 : SUCCESS in 6m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cfaecf07a3ac49c5befe445a2217754f : SUCCESS in 12m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ce3805b1b4b64c2b8e1a2a30b9f0e04a : SUCCESS in 25m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ca6134115a7941c8b51a53cf92930d60 : SUCCESS in 14m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/79ed12a143fe4bd08d04468f3f38ba6b : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f61a99a04784ba79eb36277c683a837 : SUCCESS in 17m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/121494165d944bd6aef3f00db997c028 : SUCCESS in 4m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc292d44c135424f96508ffa663aecda : SUCCESS in 6m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9a6035b6787846f78bf0c1fceaa28daa : SUCCESS in 9m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ae566ffad4941e3ae91f1574416e63a : SUCCESS in 1h 04m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e35a863abb0b4f29b594c3e0de13d457 : SUCCESS in 1h 43m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9353c61da306287b0bb0134a95b95f99e500d5f5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-28 02:01:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4fac16e4c9ae2ab895e37ec685eb064553ca8115","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 02:02:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70f271376dd3219223d26c5411d3e07b1ce1470f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 04:27:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fabd3b5689fe41daabecb4eacb246038 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f8283331c494d7295883286c00b073b : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a33b0b54da844ed923decf800d72bd1 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2374f0aa22834ddfbfaac631aba85fbc : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d32516fd4534b689bc241ef9662d1da : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9bbe9daf9172446fbf9edc0e891a6443 : SUCCESS in 3m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c4dea0179dbe44a9ab2ce89042564e24 : SUCCESS in 8m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00b2575f66304a3f83937ac5abc49c22 : SUCCESS in 9m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/70d6d8a4825e404a842b114ae0f20e31 : SUCCESS in 11m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aed9497b6e0a4b94a35ff7b3bc467487 : SUCCESS in 5m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/877efdaa08d54a829640fa782fc1140f : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef8e23186c064761b57cc54314c1a822 : SUCCESS in 7m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dffd6c7f17f643c183330aa18b4c1b72 : SUCCESS in 7m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f1a50bd4384147439f51f4144bf552ac : SUCCESS in 15m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5759df5d0ebf4fdd9339bf302741be5a : SUCCESS in 21m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f3bf2362f1e54645952b0a2b7c969088 : SUCCESS in 11m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8f56e0cb642b44a48008c7a0dceba7ae : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f91a14562c7a436fa50682a1c4771c5f : SUCCESS in 17m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/014264dfb53a4432bd23814b925b47e4 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38069d0df7574a1a88d111df692a20d8 : SUCCESS in 6m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/95d3d48ad9dc498c8bb72c4553479ce0 : SUCCESS in 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d582c691cb244ccb94d9a024568964a1 : SUCCESS in 48m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c77666aac1348849cd496e58f37e55a : TIMED_OUT in 2h 04m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0c89f0bb215e35dba03590bc71e96bf572928e42","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-28 06:27:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b82d5bd414499377dd4910baba51279ac9d53a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 06:50:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fcdfd8394d9b4602ad558c47f10b9e9e : SUCCESS in 21m 17s","accounts_in_message":[],"_revision_number":1},{"id":"954a0199709cb364a1b9a8624c4cc7323fb7b6e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 08:33:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9965576006847579f194efc1e0f79af : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f309c04f834066bc1f3b2b648f4b7c : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0aa3b343dfd4caca1128c2479649689 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba081f25edbc4b4a91765b87fe1c0855 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05dcf0cc7dad42a58da6634b720adab0 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f62bd21516d34d579ba3df7a48827f5f : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/30976b4acc9848818b15bb7df23abc51 : SUCCESS in 3m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/87e5742a70f143cf94305b83c609d7fa : SUCCESS in 9m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9b34567f04d946a39ca7db6093d113b1 : SUCCESS in 8m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d981834e9e5346ce93db91570e79b9df : SUCCESS in 8m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2cdc1cf7ffc84ff18f979733eda91e04 : SUCCESS in 5m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/810a4d60865242c199e9bca133175fc6 : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0b4d1d288cd7471bac7e98fd39236795 : SUCCESS in 6m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/17779b4521f84dd4a609c4f20a898121 : SUCCESS in 7m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c0aa853000064508be01fddc4d30c6a2 : SUCCESS in 16m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8bc694327b364015a8cf72f898655b87 : SUCCESS in 25m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9fdc8bb3b80a40f69afeb21b0f59ebdc : SUCCESS in 12m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/724228fc097f43939aa6eaa093486bcb : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/837933f1d5db40e791d89394b618ffd6 : SUCCESS in 17m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c55ef032b78d4bb0bcaf0475d045e1be : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cea2c1e057cf4ae98f400ace25ffccd7 : SUCCESS in 6m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cd34ed132d56429ba992de1c0f674783 : SUCCESS in 9m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c68ba9715b414cd48ded78a87f264b01 : SUCCESS in 1h 01m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d72e3cd5938f434ba468a858a4b65b30 : SUCCESS in 1h 23m 17s","accounts_in_message":[],"_revision_number":1},{"id":"b6b07c79ab14ca302a6b454f322c338f0c935d7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 08:34:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2cc9ce57e66e9e83de5cc4f437f55042cbe2dea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 10:05:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16177c9ed640493097f4d383d07ca0b0 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abd8908effcd4a05b4cd80cc4e7d83e4 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e207c30245e4504806af039f43cb347 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95f5435f714c4951aeda717dd083a1ba : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a550074f799e4c1b81e2699c18f6633b : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab545b79bbe64c77893bed090ff829c5 : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2bbc9b0a9224c89925011158d84af48 : SUCCESS in 7m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/581dab08a61642f7be660a8e65a23709 : SUCCESS in 10m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9ebe736fbe284f66a80c508c6affd688 : SUCCESS in 8m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/73634c750a1e42cf9d529746a0ea16d5 : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e92a406501394816916d52ff22cdeab1 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/757a4a009c1f4f3d84aee6e84f587f9e : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/432aa778dd5644898b8bede19567131b : SUCCESS in 7m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c31a4d8541a54afa96937ee97a38c560 : SUCCESS in 12m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9ef59193cf7f416b80a6aba6c1d948d0 : SUCCESS in 24m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/89b7664562074dcb966b348db59891b9 : SUCCESS in 15m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b63fbbdd62b74b7e82c02ecc5062174f : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/550dd9d2c7384c8e91ce256ad26a28de : SUCCESS in 17m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1e5c72395db942a09f8c1f78a864edf0 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/269dfe10e65744d7bb1b7be26d923c47 : SUCCESS in 6m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f049b211889c4b4fb3ece7f3b91ab33a : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bd1f0147ad74c6896bb9a6f5f399eab : SUCCESS in 46m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ee764d29f4c45f28ffa0692f62bbff6 : SUCCESS in 1h 25m 35s","accounts_in_message":[],"_revision_number":1},{"id":"12757af00894819c941d9786b161c4b822f4921b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 10:05:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"14ed794e2bcfeb8089a4b57d2c8cefbe82f63cf8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 10:06:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/344dcdd5e7ae4159b081a5d4fc21c8b0 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f74b67ac0a5a42e7f933c5cc75201e835d3413a7","revisions":{"f74b67ac0a5a42e7f933c5cc75201e835d3413a7":{"kind":"REWORK","_number":1,"created":"2020-11-27 23:29:08.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/95/764495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/764495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/764495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/764495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/764495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/764495/1"}}},"commit":{"parents":[{"commit":"9b47a48a92683da51228c72d0645f25cca6ab772","subject":"Merge \"update constraint for oslo.policy to new release 3.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b47a48a92683da51228c72d0645f25cca6ab772"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-11-27 23:27:50.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-11-27 23:27:50.000000000","tz":-360},"subject":"manual update because bot is not running","message":"manual update because bot is not running\n\nChange-Id: I0e15eeefa3fb8af0e307b8724dc3d749fa3870d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f74b67ac0a5a42e7f933c5cc75201e835d3413a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f74b67ac0a5a42e7f933c5cc75201e835d3413a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
