)]}'
{"id":"openstack%2Frequirements~764517","triplet_id":"openstack%2Frequirements~master~I0fca8d724c431e006dd05d1251a65c88146e58df","project":"openstack/requirements","branch":"master","topic":"retire-qinling","hashtags":[],"change_id":"I0fca8d724c431e006dd05d1251a65c88146e58df","subject":"Remove the qinling deliverables from req","status":"MERGED","created":"2020-11-28 02:10:22.000000000","updated":"2020-12-01 23:55:43.000000000","submitted":"2020-12-01 23:53:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764517-retire-qinling","meta_rev_id":"faf2110b99d709f587d4bcb0b5cc2e1ff1f9e447","_number":764517,"virtual_id_number":764517,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-01 23:53:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-11-30 17:30:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-12-01 20:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-12-01 20:11:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-28 02:31:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-28 03:26:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 17:30:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-12-01 20:11:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"6483cdefdd519a712535fceffed13a2e3391cf1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-28 02:10:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b29ed52808afdc8cca398c4dcab33b72ed1dd3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 02:31:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d17385f718af4ea3a085584db09f6ad3 : SUCCESS in 16m 25s","accounts_in_message":[],"_revision_number":1},{"id":"cc41f178b3515b90fc9b42569d772826ae1f4b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 03:26:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f092685646843bf8e91f1a14c70fb2a : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c07b57f20e7a4d3095b2581804688f4a : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/142489ec5d87496cb031af2a6f51f10d : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3494247bbb3345f0a382ea01c9f15b63 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d6c929a15fd4ee780aa4e07a91fe2c1 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/742e0665400d43f6816523c83910520a : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2d55af84f38c4438b27a508415b73af8 : SUCCESS in 3m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/68a73e4c1af14e1ca1f4c8e448558b4f : SUCCESS in 8m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a725e7bd064e47be8dc8a7603e97d32c : SUCCESS in 6m 45s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/787c5160e7244c828eb43bb069e63e05 : FAILURE in 3m 38s (non-voting)\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f6b2b87facfb4cee952133a41d5dd290 : SUCCESS in 8m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5fdbfe8a21a548b49b20244558913803 : SUCCESS in 4m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f4e74aa15c1a446c94dd94d7567a1fa4 : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b52a6228583e42d8a70ee7ca361997e3 : SUCCESS in 7m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/35bbefffe4054599b3176922c63ad0f1 : SUCCESS in 5m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bc922e5b35a549c3869002c457a7e897 : SUCCESS in 12m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/72db8cc25bf747b5a25d9f37b5370963 : SUCCESS in 26m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7a4b1828e6064a349ef285dbaa6002e3 : SUCCESS in 10m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fd9abed6060b4f56aa967c927c3a7f7a : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f56451ea81f044e8a1ba376f9aed1b87 : SUCCESS in 17m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/830c564890604eb58b3bd94ae328dc0a : SUCCESS in 4m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d9a1f3ec98e94460b8fa839dc208bdf0 : SUCCESS in 5m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2e9e932d62284e858e6eb8fe4a1da0f5 : SUCCESS in 10m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f597a520fd5f44d2a62d88a700532e06 : SUCCESS in 55m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d894d6c942254b8baf4922e0c67e3fe7 : FAILURE in 1h 14m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d25615dc12fdbbbb1664e146224fd5c1ccf30985","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-30 17:30:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6416a9bf91fc54e9df2432bc2eceb0234463f769","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-30 17:30:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"36094759f6c0c929ae1e988b90fadd0163525ced","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 17:45:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/06d215a76a0047688b57f465959e2c86 : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5643755ea57f64975e5e193a9939b53774b3c751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 19:02:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be26db302674fa589c50eb118b1e3c9 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e57d5f67abe44659e7604ccf5443cb0 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be705c3431ce46b4b75e0bedaf64a720 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f16dfc579bb4de4b6add1ad03556b52 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/576e7a6370ba4a41809e839a4d577324 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/01a0061318d3459fad6f3bea069859a5 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ef75244c009d4a17a56edbee8b1af1ff : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b858f7333bf44a1bdcd43dc80240d49 : SUCCESS in 9m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3709cb78bf0342eabedcbe3f78ad6198 : SUCCESS in 7m 57s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/7847eb1acb3140e896a4309df43d9bef : FAILURE in 4m 29s (non-voting)\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/11a996e6426e414e9d4a7386ddde45e6 : SUCCESS in 9m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fce70e4fe8f94a8192b5285e46500427 : SUCCESS in 5m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f1d8310007ea4daeaa0c4fa0b7bba39f : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/36697a4ae8174ef8a96c68178413ac71 : SUCCESS in 6m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/73c9bbd58c554d6888887c9ff8f5a5fd : SUCCESS in 7m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4f19d1aa888d4b77b41ddfc05d7792b4 : SUCCESS in 13m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ca15bf4d804f47f7b1f489bd91a8f51e : SUCCESS in 24m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d3b336800ee14487909f136950326a7e : SUCCESS in 12m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f30d5cbbd4124b4680cb56a7998e4f19 : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24d01b10094e4c51ae1412c7a733e045 : SUCCESS in 18m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dbb67231d10c4b6eaecff38bd53cda71 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ddc4f083f0e9408e9059e9ea4f3ea78e : SUCCESS in 6m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9d3c321abe874046b2bb2f423da3d834 : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52e31e5efae648a0b2d78c0075a052b0 : SUCCESS in 53m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17b6b0f2a3064774bf18505337dfc136 : SUCCESS in 1h 25m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0ad96e0da8fd9952d0f94a430fcb8a71926fcca9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-01 20:11:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nDoesn\u0027t appear anyone is using this.","accounts_in_message":[],"_revision_number":1},{"id":"bb024a98c9b2a6e1a08003ac890e77bfcc88f8f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 20:12:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f58f1fe14417e0d7414d32bbbc40e18e9ee7d1fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 23:53:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ea0c2580d0e4cf58c01602b6534428d : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d54f6d31e544273ad509ea716e42c8a : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f26281fe94a44a893840fb54ca47aec : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97e3f87f7fa94cb59324c0c025db42cd : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f794022234ee41f88e80e7a4774ca511 : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/193d3767e73043a7a8d7b50a76152034 : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5a06051ee3314beda15f2bb4c350a6ae : SUCCESS in 9m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8a6243b6d07c46f69665d2cc4684eea6 : SUCCESS in 7m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b66cacb6819847708831e28bf3adf08e : SUCCESS in 9m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/be40c5254a93450d88d2bc5900d4d822 : SUCCESS in 4m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a3286805b3ba407b9c5dd52c09fd485a : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7cfb0f6f6a3843cd993bbb113e62a0cf : SUCCESS in 6m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a44806888b7d4dedbf7e0f6521f1038d : SUCCESS in 10m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/611ffb0a06a247da9ba635f9db8ca587 : SUCCESS in 29m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a0c162f330054b0aada00affb7e4490f : SUCCESS in 24m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dead4328c2974198b6efcdaf2355f37f : SUCCESS in 12m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c7694f3095124d2c8e826bbb910e3bb8 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3da76071e27e45628c33af5cc5b1061f : SUCCESS in 18m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/58114f6ce2b84247a54dccb92d9582aa : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a76e49e127f848b8929fd14aaaf439db : SUCCESS in 5m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/486c2eef91f84a8bab366d14965141fe : SUCCESS in 10m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38742941cb884b7da89d5916c2a4a5fc : SUCCESS in 1h 00m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c818f52d85341e2853492d8c838c63e : SUCCESS in 1h 18m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c86fe8f4eac188fed4d81457579a0dd84f92eeb5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 23:53:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"faf2110b99d709f587d4bcb0b5cc2e1ff1f9e447","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 23:55:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b69a2549e57846bbb75839f89825005e : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fec5f64f3945346b8395b4d21f475350b2299c31","revisions":{"fec5f64f3945346b8395b4d21f475350b2299c31":{"kind":"REWORK","_number":1,"created":"2020-11-28 02:10:22.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/764517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/764517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/764517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/764517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/764517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/764517/1"}}},"commit":{"parents":[{"commit":"9b47a48a92683da51228c72d0645f25cca6ab772","subject":"Merge \"update constraint for oslo.policy to new release 3.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b47a48a92683da51228c72d0645f25cca6ab772"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-28 02:07:12.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-28 02:10:21.000000000","tz":-360},"subject":"Remove the qinling deliverables from req","message":"Remove the qinling deliverables from req\n\nQinling deliverables are going to be retired in Wallaby[1],\nso let\u0027s stop syncing the qinling deveilerables req.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018638.html\n\nChange-Id: I0fca8d724c431e006dd05d1251a65c88146e58df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fec5f64f3945346b8395b4d21f475350b2299c31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fec5f64f3945346b8395b4d21f475350b2299c31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
