)]}'
{"id":"openstack%2Frequirements~764518","triplet_id":"openstack%2Frequirements~master~Id8c8f38644740d80a32c3bde3fc58b088222d1d3","project":"openstack/requirements","branch":"master","topic":"retire-searchlight","hashtags":[],"change_id":"Id8c8f38644740d80a32c3bde3fc58b088222d1d3","subject":"Remove the searchlight deliverables from req","status":"MERGED","created":"2020-11-28 02:16:01.000000000","updated":"2020-12-25 18:46:15.000000000","submitted":"2020-12-25 18:45:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764518-retire-searchlight","meta_rev_id":"bc6ab28137646a72d6eaeb21321ca8e4ac934710","_number":764518,"virtual_id_number":764518,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-25 18:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-25 17:04:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-12-25 17:22:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-12-25 17:22:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-28 02:34:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-28 03:27:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 16:43:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-12-01 20:10:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2020-12-25 17:22:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"b6b572dc47772de5fc1925adaf9c1d8bb45e7244","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-28 02:16:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b64890601cc7df30074327568ee7e1900c8a668","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 02:34:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9689f0a6c0544aa8a5b1288014461cb6 : SUCCESS in 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d3b0318a5c3dafc89583ea9cb24256875d4bde99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-28 03:27:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47f01529a4f448f18a94dcf478b7149d : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39e3096474c241a3a78816e065b5de6f : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1359d00205dd428bac0441b8091a2f89 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69744309e0a548ccbebb890ed27f77f7 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a224930b13354c6f9e07d292548aa21a : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a376fc51a244e3e8af8406cb2298451 : SUCCESS in 8m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0ddcaae8b254eb5ae2381763039f0d3 : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/45dafbaf9c274ebb81f7c0f04331a16d : SUCCESS in 6m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eeba33de5ff5492b88678c8c77ecd2a5 : SUCCESS in 7m 27s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/19b26955d48f474d8d36025ffd307871 : FAILURE in 3m 28s (non-voting)\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2bc2c6fd983c4bd29ee973709d7e62a9 : SUCCESS in 7m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/457ef441ffd8472b809af8f8d9ecc33f : SUCCESS in 4m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3a44fa7754e644ff84de597ae79c9589 : SUCCESS in 8m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f54287e493694b35a9c877a7e88e26ed : SUCCESS in 6m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b90ce6cdc8e949429891bccc891f9520 : FAILURE in 20m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/03229f2f425840819e762f74a34c1da4 : SUCCESS in 24m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1c87b3e9c05e44e9b23c44185d8fea6b : SUCCESS in 16m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/adafd63fe9e14c10b82711671a5ade01 : SUCCESS in 13m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a48bc549fdd94125beed3c65bbc9edc9 : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/33af5cb9121149dcb34a191ef2ea6465 : SUCCESS in 12m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/65dcbf0cdebb4599a2faa4c9d134938e : SUCCESS in 3m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aed9180e334e4a9abf843bf9ebf6faaa : SUCCESS in 5m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2b23113458824312bad5d57a9564099a : SUCCESS in 14m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0c72e937fc04b57acbb00fc9d11cc6f : SUCCESS in 56m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/227c0b57ccb54d0492b3a3d0e1c7c9c8 : SUCCESS in 1h 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"927d2fc395d0162c735415af1915b3df517f0641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-29 00:19:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f83c83479ea81c1bc698d22969be63ea715bf2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-29 00:45:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/258d30dda11c410bb6c835c6fe496b89 : SUCCESS in 21m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b9352b73af1f683b6c35ddbdc19382b8d76db8fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-29 01:40:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ffba79058e14bb7ba456bc0636a9f39 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3896d1a736cc463fa0e0342304db6478 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23fcd00465c44828b7829ab0aed59063 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb34b32fbcce41a4899c45c5d7a56174 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c2c19c586954da68cc88099aadde0e1 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b3e95b2be5a4bfca8bf482efd47e76d : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/64d02bbc12b944f9b6a1396ce7baa770 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d937685aabe24273a3c411c9cf6d5f03 : SUCCESS in 8m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/018d393e7e99430b80c3d96502678565 : SUCCESS in 9m 29s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/c28e7bb7084345a0b338cddd287989be : FAILURE in 3m 25s (non-voting)\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8b607680f2f84708ab5cadaf229f6692 : SUCCESS in 9m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c96e690b6b7249109907c504e451a58d : SUCCESS in 5m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0d4c053827a942a3bb0fb9eaf963a510 : SUCCESS in 8m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d70d23a813ae410894f914f235aa7114 : SUCCESS in 5m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b6309b56fecd4f7684802fa0ca6f7f59 : SUCCESS in 8m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/695fe776e7ce4891b6ce7f9f4d9679eb : SUCCESS in 14m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e3b9e15057ec481d9e39e9c5cdc51070 : SUCCESS in 21m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ea39258fb32c4f14b1580a248cc20f26 : SUCCESS in 10m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4279356c805f482e9d7b0f683e02c04c : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3aed0f8d058745239028736bd06fb7e1 : SUCCESS in 17m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/49e6bb9f93ad4c399018ffa4e1cbb3dc : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32de568e9bd74f9697966adb8c3f46d1 : SUCCESS in 6m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5a323546126a40ca8fb6f6c9328b43fb : SUCCESS in 9m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2a3dfb4930e4eb39ca602514a42cfdc : SUCCESS in 45m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/380c0290ce244d839fa9237616094796 : SUCCESS in 1h 18m 21s","accounts_in_message":[],"_revision_number":2},{"id":"284590d3823c53dd28d9e32758913eec474f31c1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-30 16:43:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7b739a98e8ebc611a8124206db8a9042ce82b267","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-01 20:10:51.000000000","message":"Patch Set 2:\n\nDo we actually need to do this? Other projects can continue to use older versions of the client if they want to continue to work with past releases. python-openstackclient still has it in their docs requirements at least.","accounts_in_message":[],"_revision_number":2},{"id":"0f7411576379b4b571aa1e40084a58a9b3f4a23a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-01 20:29:04.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Do we actually need to do this? Other projects can continue to use older versions of the client if they want to continue to work with past releases. python-openstackclient still has it in their docs requirements at least.\n\nyeah plan is to remove all the usage - https://review.opendev.org/q/topic:%22retire-searchlight%22+(status:open%20OR%20status:merged)\n\npython-openstackclient - https://review.opendev.org/c/openstack/python-openstackclient/+/764569","accounts_in_message":[],"_revision_number":2},{"id":"39a3d5baafd304dda3cacf6c07e5d55a53dea510","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-07 19:14:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"74275179faa755bc767d99a176a0f7c0cb82936c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 19:39:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/19284191778c46978733e690e0140189 : SUCCESS in 20m 10s","accounts_in_message":[],"_revision_number":3},{"id":"0c884301de41dfc8a3c05714050b146682fcd6a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 22:03:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd852e4e5d804e29a0972d48f1edc235 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/128045285ba0440e8d85340807b1a42a : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4b518a1cc6645d6a07595c45ae635ef : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8004b9ced87f44d7b0b183d68bac1ba1 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da5a47601fdd4a9e9b581f0ca038a032 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/776528e710f14ca19c53b2d2290f97d1 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d1ebb54f8624708a6430b7544308145 : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5dc356de10a2482a84393a7adaaad88b : SUCCESS in 10m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4619bcf4f01f42279634117d6b3fb988 : SUCCESS in 9m 44s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/e5323d059da64a859a971cec62b00638 : FAILURE in 4m 11s (non-voting)\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/456031e55e8145c095ef2f674a979c8b : SUCCESS in 9m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b24eb31a50454e4f9a47a3947e7050dc : SUCCESS in 5m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/af1afb3d112c42c399370aba6bf2e44b : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aa23471011024d04be991efb86780919 : SUCCESS in 7m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1d2098de49234e638a71426b09bfc176 : SUCCESS in 7m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c8a85a59de9e4ff8870dce0de5ae8845 : SUCCESS in 13m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bf1108f2505843fbb6e79df0f6889237 : FAILURE in 3m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ac60079b60654dc6b5b7e141530eef0d : SUCCESS in 11m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a638eb3410b54fcd978fa6627440b4d0 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/58e70b2a7b474832af3a8d63f1bc35a3 : SUCCESS in 21m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d46c9d6f9acf44cc98f39f80c9861ebc : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6bcdd2ff222c4dc7b342c1a2ed9c4dce : SUCCESS in 8m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7a3e2649636145a6bb8a0d9aa94bdae5 : SUCCESS in 14m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90a5d5e4ade641b7b5df3ce352904d5f : SUCCESS in 1h 04m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/634c52e586b841008a16639f5a3ec5f7 : SUCCESS in 1h 21m 41s","accounts_in_message":[],"_revision_number":3},{"id":"be5b028505a4e064f84d20fe24a24a40ff504615","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-19 05:28:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c3efd8ab6e55c16e569da0f9dbeb45fd23a21a04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 06:13:24.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/860fa7d96675479c96f88f6c9b930c5e : RETRY_LIMIT in 2m 14s","accounts_in_message":[],"_revision_number":3},{"id":"4244dd9b323d4d888bf7787c3a13704b7f91ebee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 06:55:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c67da11987194fa581cef95febe635b1 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/054642436eb245f9a0de3b4070fb34c9 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25f10d919af04c628ab2b5527c225083 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87902c44dfb04aeca6facb888b45811a : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a60d4df9184f42c5a69df4ffcafac09e : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/359ae95ac02045f28d5133bc2c906e42 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/946cceb04467401585f305a02fc7d68e : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0989484a6dce4162a68b829a58967fc8 : SUCCESS in 9m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9b8edf18a82b4fbdae8a1d7a4f81f253 : SUCCESS in 9m 28s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/8558db106324418e9ec7d743caef2c34 : FAILURE in 3m 16s (non-voting)\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/350011d74a9e4f69a75fa20f51c55c99 : SUCCESS in 9m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c2383c28a6214f2097b3d734c2652bff : SUCCESS in 4m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c75139d2ff314ae5843e47d659672de0 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9fe56cd921054b3d92331d3f9717a7cf : SUCCESS in 7m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/16e02c1b110b4706864d46809d3c790c : SUCCESS in 6m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1f17619d1af948338d7a6b515abd89b3 : SUCCESS in 11m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ec0ba88d5a534c9292b63ffe48a43601 : SUCCESS in 22m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/645a6ce7707b4a4a83f5b864be3cf394 : SUCCESS in 12m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a9c7cc299ccc4e1b826f7e7df27d3b90 : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/847554f4b1e340ff83d52a9728642780 : SUCCESS in 18m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1d80408befb4467480d9430fa73b44c1 : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ca10a9eaa0b495095c4105ff0fcd0d7 : SUCCESS in 7m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9bec77dc6abb4a2ca88853b2eba90a6d : SUCCESS in 9m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a24fe4aae8884861b76c24ae4bf3a9e3 : SUCCESS in 55m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92db7d66abd34452b70523f6469e2a19 : SUCCESS in 1h 24m 27s","accounts_in_message":[],"_revision_number":3},{"id":"f5b2120eec451b260324e712471c1ba6cefd5684","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-21 20:59:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7feb3597695fe4edc4e98d5cbd152704e152103c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 21:16:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0a42db3135834edda468515b9aaa3881 : SUCCESS in 10m 45s","accounts_in_message":[],"_revision_number":3},{"id":"6a7dfe5609127f841c00f97f24cefee317de4f65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 22:21:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad83601ef71b415a9e57c38b8f9d4bed : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef09997a1c8e41e0a169a802bca2eb50 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0912c2ca300d4fa1868a03aa82340c07 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a384bf2de8c4d00b14f5f7ba188a2aa : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efb47242e52f4c83ae7b40b2c263e9a7 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a931fbb9d1045af8d1489c435002c20 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3a2ccb1e89de49bcb3f3ae08a0ca0e6c : SUCCESS in 3m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ecd30cb763bf46ed85d0959a17448bd1 : SUCCESS in 9m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fadbb98fff7f416f8c7a263358dd2e7d : SUCCESS in 8m 42s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/9d23727839da40668903ad3edf36844f : FAILURE in 4m 02s (non-voting)\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5b0bd73b5a074bd188510b04a1b0f416 : SUCCESS in 9m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7d87e8b55c9f4126ac1a630affa95620 : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e828425951e847079d53350d4e76fffe : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b2e0d39600284682a46a591cfbad82a8 : SUCCESS in 8m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/33dd8020f22642dfbf02c02b0a0d7e77 : SUCCESS in 7m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fc924282222843e18125f62704d8ab8c : SUCCESS in 27m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5c64f2bec8c942059954ab034becfeea : SUCCESS in 27m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5348687633ae44aa82c521a26a74d594 : SUCCESS in 14m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fa7c3398fefd42db9e08372223ec25f1 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7aac06c5b689444f9db82fb84483d6f3 : SUCCESS in 16m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/72a00008deaf465493deb1e94241c7e9 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2468df97328c43d7bcdef4c86ac24f62 : SUCCESS in 7m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d3109ddba93d4a30b3a9f47b11fe23e4 : SUCCESS in 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98534a27404745909124e220df96b023 : SUCCESS in 58m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98f0fd2e96f54ec6bd020e427614362e : SUCCESS in 1h 18m 24s","accounts_in_message":[],"_revision_number":3},{"id":"107355ad9dc3c797e86ce4faaafee22ddd6c24e3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-25 17:04:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d824ca1d9cdda2ba64aa656a732d93d907bbf372","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-25 17:22:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2e88e6551081eac5c0377e2fc4841f3c129fdcb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 17:22:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b8bc38841cffd42302cd8cf96f45f67d7db1c424","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 18:45:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2b36a7197a468d82c03645af2e169a : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/823f9d4b8d6149a8b11d14170b81b6e2 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fe6bcfce2c24e2f8ee49cc13b288086 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dc3292a7f684653a1efb821e637483b : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08fade588f7442cab240bfbc84a60acf : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/19782b336b9f4b77abac19a14961c21e : SUCCESS in 3m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8bfb924b2ee64d6d97e4a769779cb480 : SUCCESS in 8m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/80be19fdf168441d817f4c77c2cb1cb0 : SUCCESS in 7m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fdab3120ed8d4ae9894b5f705f6aea71 : SUCCESS in 8m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a3322fb0339d42fca3661b2392f8002d : SUCCESS in 4m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7757bd2f00e949c297debd5894a35d70 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d000d23cb7474024935d87ae1d3a2188 : SUCCESS in 7m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/aabbda83c9fd4f73b3d1aa9729214473 : SUCCESS in 6m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d83c3377b2784f4cb7d1a640cc4e3fb6 : SUCCESS in 13m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9a261bb847934416a8510238eb4b742b : SUCCESS in 29m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4731ebd08a0f413d8ba58aad423a1c2a : SUCCESS in 11m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fc7d47261a044c8e863c74779a16e048 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/765655bd26d74d198729982fc43b40e5 : SUCCESS in 17m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0d2650cc63c54004b9e1d0149a7599c5 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6ee1e26b752643648765352f84cf623a : SUCCESS in 6m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/01b2078b9612435fa5f28dfd3bc88b43 : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a348d0cab7d43359a1743851458e38c : SUCCESS in 57m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5b575d3bc4c475ba95a431a54be85f4 : SUCCESS in 1h 19m 09s","accounts_in_message":[],"_revision_number":3},{"id":"02380c389fbff5a01b81233d2f75f1e298f1e303","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 18:45:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bc6ab28137646a72d6eaeb21321ca8e4ac934710","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 18:46:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b268ad74718443c9346f460aa985b54 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f695891a7b4ff6ce086f44be35e59a2edc27b7ca","revisions":{"66a9b029ec32cbd18b9e351bfec3035c1b4efc72":{"kind":"REWORK","_number":1,"created":"2020-11-28 02:16:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/18/764518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/764518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/764518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/764518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/764518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/764518/1"}}},"commit":{"parents":[{"commit":"9b47a48a92683da51228c72d0645f25cca6ab772","subject":"Merge \"update constraint for oslo.policy to new release 3.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b47a48a92683da51228c72d0645f25cca6ab772"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-28 02:14:27.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-28 02:14:27.000000000","tz":-360},"subject":"Remove the searchlight deliverables from req","message":"Remove the searchlight deliverables from req\n\nSearchlight deliverables are going to be retired in Wallaby[1],\nso let\u0027s stop syncing the searchlight deveilerables requirements.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018637.html\n\nChange-Id: Id8c8f38644740d80a32c3bde3fc58b088222d1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66a9b029ec32cbd18b9e351bfec3035c1b4efc72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66a9b029ec32cbd18b9e351bfec3035c1b4efc72"}]},"branch":"refs/heads/master"},"cec5b52dac691f1ff0616b4a37021bc9aa28fe51":{"kind":"REWORK","_number":2,"created":"2020-11-29 00:19:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/18/764518/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/764518/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/764518/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/764518/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/764518/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/764518/2"}}},"commit":{"parents":[{"commit":"f74b67ac0a5a42e7f933c5cc75201e835d3413a7","subject":"manual update because bot is not running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f74b67ac0a5a42e7f933c5cc75201e835d3413a7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-28 02:14:27.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-29 00:19:12.000000000","tz":-360},"subject":"Remove the searchlight deliverables from req","message":"Remove the searchlight deliverables from req\n\nSearchlight deliverables are going to be retired in Wallaby[1],\nso let\u0027s stop syncing the searchlight deveilerables requirements.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018637.html\n\nChange-Id: Id8c8f38644740d80a32c3bde3fc58b088222d1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cec5b52dac691f1ff0616b4a37021bc9aa28fe51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cec5b52dac691f1ff0616b4a37021bc9aa28fe51"}]},"branch":"refs/heads/master"},"f695891a7b4ff6ce086f44be35e59a2edc27b7ca":{"kind":"REWORK","_number":3,"created":"2020-12-07 19:14:40.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/18/764518/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/764518/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/764518/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/764518/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/764518/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/764518/3"}}},"commit":{"parents":[{"commit":"b4e2a19d7d1e7a717fe27d5b167d93ed3371e345","subject":"Merge \"update constraint for python-saharaclient to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4e2a19d7d1e7a717fe27d5b167d93ed3371e345"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-07 19:14:05.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-07 19:14:05.000000000","tz":-360},"subject":"Remove the searchlight deliverables from req","message":"Remove the searchlight deliverables from req\n\nSearchlight deliverables are going to be retired in Wallaby[1],\nso let\u0027s stop syncing the searchlight deveilerables requirements.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018637.html\n\nChange-Id: Id8c8f38644740d80a32c3bde3fc58b088222d1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f695891a7b4ff6ce086f44be35e59a2edc27b7ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f695891a7b4ff6ce086f44be35e59a2edc27b7ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
