)]}'
{"id":"openstack%2Frequirements~764635","triplet_id":"openstack%2Frequirements~master~I1c587182219f63234770004cbcca04dacf683f40","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1c587182219f63234770004cbcca04dacf683f40","subject":"update constraint for python-troveclient to new release 6.0.0","status":"MERGED","created":"2020-11-30 11:12:41.000000000","updated":"2020-11-30 18:18:16.000000000","submitted":"2020-11-30 18:16:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764635-new-release","meta_rev_id":"58de56830e58ebcced2ca07274f52ab7cae6c693","_number":764635,"virtual_id_number":764635,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-30 18:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-11-30 16:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-11-30 12:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-11-30 16:43:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 11:32:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-30 12:19:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-30 15:26:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 16:43:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"61a1b56edb8f4c14c3d1f28aca93ac8fe4d32a0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-30 11:12:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d5fdff7fd61eccf7944f62345bcd8f366022254","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 11:32:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a4005c9fbe1b45e2a43df04227c63b2f : SUCCESS in 15m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a7dc247f140e32e71235f70a9aa54e0b8cfe3080","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-30 12:19:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5af065462d9c3eea6bfe7a5e214b757c3b5dfa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 15:26:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/929d2866bee74be5abd548d5b73f5b28 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f8f5ceb666a4bfa993b7b2bd45e5fda : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adab411ca2cb4cb79339e260c4d3b915 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cdc987f43ab421fb8eb0dce5fdb150f : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11bf3baea500437ab0e24dcc613aa6a9 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c2313f653f44a659caecda1d91f0630 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55ade55bfac642f480e2e5a73373675a : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/abf6459c247b4661999955d1c9fbd5b9 : SUCCESS in 8m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a85cf41f3c294111b04c86bd508f3a2a : SUCCESS in 7m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b426409f6bcd45449ff140366b8e5204 : SUCCESS in 8m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fd9fe2a6340c4bc38a16c7b87df505dd : SUCCESS in 4m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f1b0aa8538444b50940864aa88c1a94a : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a5e39d3719f9416ba0719176d80421f1 : SUCCESS in 8m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/682d0cdc7061499ab23a35ba6282df0c : SUCCESS in 7m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0f865ff270c442e98fb9f62c7ecdc0c6 : SUCCESS in 23m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/968f4a94938a47a5947f9088a77f3344 : SUCCESS in 29m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/23f2541adb8644e9ae33d28e2a171a0a : SUCCESS in 11m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cfea7e212c904adbb17ceed0a34efa0e : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e90434e52d54633b45672dce36538b8 : SUCCESS in 24m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2a2a95a629d54f77973bfffa7bc827aa : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3a9b7480f0194349beb154bef8121052 : SUCCESS in 5m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/38670f9585ab4cdd81f1e701f40118a4 : SUCCESS in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca20a66b118f4bf6b06f2ca97c84755f : SUCCESS in 1h 01m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24222210160a4e869eaeb9f5c264c2f0 : SUCCESS in 1h 19m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c620ea55e377377fe09395a3a1450c4b82006431","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-30 16:43:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"449608e4be606bcd708f49caa1b017c48f49744a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 16:43:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e09d271bf7b1b5ae172be207b6ea90ca3ef604e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 18:16:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b14811cc97ca4a78be66f6d14642ccb5 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b1f0ff17a8479b8655c19523146e2d : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38fb2414956645d9bf597d96d164fc3f : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ff8b7eb9eb243788032007ac43c4783 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/39b578fe3b5a4a73bee06b3aa1814501 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17610e72c6ea470db6c5d3b6f005554c : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3ee544a50de04a60b162ae485d57705e : SUCCESS in 7m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/96879b6788184eb2a5790f878bbeb836 : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c343c8a89cae4baca293b3514fc1dfd8 : SUCCESS in 9m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cada61915f23458784992a6348b82f33 : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c9e07ef80e9441a7a4ab5c40fa76997f : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0009d3a45b814dc592e4d9ea737ee45f : SUCCESS in 6m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3cc06636381e4f9daca438ab3339b2bb : SUCCESS in 7m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1906398aebc04c5eb9927e9fbdc78b11 : SUCCESS in 13m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3fc1c5e0a21b47849115470724111616 : SUCCESS in 23m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e4bfe3f9e6fd4775910800307a585a39 : SUCCESS in 13m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/69ee77f6525e475ab84331057ffa9485 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8d3c16a83af241a4acb83cb796bacf78 : SUCCESS in 18m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/16494a15cc6d4a81b1c819353c8eebd0 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4a696410a9ca479fb0256343d1568c52 : SUCCESS in 7m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f351671ff7c44975bb8a25705e64d082 : SUCCESS in 10m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd4d30e97fa24ebca40aef47ce2acd32 : SUCCESS in 59m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/039dc313a1e94e049ab2fea53ff6f561 : SUCCESS in 1h 26m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b25a6edc87d64d14d1ed57c9895c873a8d00e7e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 18:16:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"58de56830e58ebcced2ca07274f52ab7cae6c693","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 18:18:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11fd7a4820d44b00865fee5ddab7f2f2 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"733472f4621641a0a0b85a896f07f3d2f2ea19ce","revisions":{"733472f4621641a0a0b85a896f07f3d2f2ea19ce":{"kind":"REWORK","_number":1,"created":"2020-11-30 11:12:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/764635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/764635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/764635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/764635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/764635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/764635/1"}}},"commit":{"parents":[{"commit":"f74b67ac0a5a42e7f933c5cc75201e835d3413a7","subject":"manual update because bot is not running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f74b67ac0a5a42e7f933c5cc75201e835d3413a7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-30 11:12:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-30 11:12:39.000000000","tz":0},"subject":"update constraint for python-troveclient to new release 6.0.0","message":"update constraint for python-troveclient to new release 6.0.0\n\nmeta: version: 6.0.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta: release:Commit: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta: release:Change-Id: I841969913698591984d6920a84913968f41fa618\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I1c587182219f63234770004cbcca04dacf683f40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/733472f4621641a0a0b85a896f07f3d2f2ea19ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/733472f4621641a0a0b85a896f07f3d2f2ea19ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
