)]}'
{"id":"openstack%2Frequirements~764636","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I57219027d23e35ddc3147becffcfda099efadf9e","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I57219027d23e35ddc3147becffcfda099efadf9e","subject":"update constraint for castellan to new release 1.3.4","status":"MERGED","created":"2020-11-30 11:13:17.000000000","updated":"2020-12-01 13:20:56.000000000","submitted":"2020-12-01 13:20:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764636-new-release","meta_rev_id":"1a227be0152baa4b341b0651e35c4964a515df67","_number":764636,"virtual_id_number":764636,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-01 13:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-01 05:03:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-12-01 11:41:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-12-01 11:41:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 16:32:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 16:43:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2020-12-01 05:03:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-12-01 11:41:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"5e0b33df0cdc56188c1b7e44dd81051cfaf02bfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-30 11:13:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c042fb7386607f0872e87e296e38b5fce5eedfaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 16:32:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d6084c94bcf43e1a9bd6eca554eba7e : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/951095c06ddd4ca6aa9b34fe3ba00d8c : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fec928c8fd7447bab6535f94dceb3452 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2a7233bcd694b1788481a2f3bc3cb2f : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa9532a141d74b0688b117fe77c94561 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/46cd8ca44a864b43a79f0cf54e777ade : SUCCESS in 3m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/08b8a3c61b9548ff9d2049540e7a425e : SUCCESS in 7m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9386cad7f1fd4bd0820f3a6dbb7f1d1f : SUCCESS in 7m 07s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9b057c02a612490b8929d84e16b6f4d7 : SUCCESS in 8m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/067e0885e9cb423598e57d60a4ea47cc : SUCCESS in 9m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c82016c2e5d34f2ea6081295573b6f39 : SUCCESS in 5m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ede42ee0155f46f68e63f8ec9f36047e : SUCCESS in 5m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3b3230f0f47b4c6b8d8d05cc40490c04 : SUCCESS in 11m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c9867a5a83924f1e9d595f3a848aad3e : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8a6abdd34a5441a6adf17e2dc251d6dd : SUCCESS in 7m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3e092d0ac8a148b08f52d8b81af53fa9 : SUCCESS in 16m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/55cd3aa49ff740bab357149b10aeb2eb : SUCCESS in 13m 38s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0a7dcedb99624b7f82400f9f890e849a : SUCCESS in 16m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/11527f9ddcb6452b9c48bc1aea499c16 : SUCCESS in 20m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4a6f781a521f4675ac1d0122b1dfb01d : SUCCESS in 12m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/be182b1792db4596a6983b2348a94489 : SUCCESS in 10m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/715e3bf978e84d408e7bb2698254ec94 : SUCCESS in 12m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cef5d52ae5c949e7a7eb29b4b4113004 : SUCCESS in 20m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/14ea502abef348c8accc8ee238ffb7f9 : SUCCESS in 6m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/80aec9abebe548aaa819829f957a363b : SUCCESS in 8m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0086fb55edce4be79cde17ee640d6801 : SUCCESS in 9m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/313021235eb041a2bdf1caf94b6e6d77 : SUCCESS in 2h 24m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1d7766efffc64a9388c71292d4889435 : TIMED_OUT in 2h 04m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f25d9396aba74f2cb1d912d1ed2db667 : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":1},{"id":"207f2fa72e7dcbd7046046b7be884c6cbd7943fe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-30 16:43:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5fd71d71b098c4da075468a16f6d03dff82aa1f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 18:19:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe4571a59fe54c0099fbb83f0b9d9adb : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0117671efb2b44f996448446a59daec8 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31694d4e2044411a823e3746f6dbf332 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27d07985404d4f34925a6985e6f8d6e7 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8f3074f0a8d4bea8faabbb16ba7db84 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/db9f27b9ca49407881ce96bed92afc66 : SUCCESS in 4m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e18df0c1001747f6ba2c166c86100090 : SUCCESS in 8m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c4ee98fc59c8478eb435d45dd944959e : SUCCESS in 7m 01s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9e12d5fb93914415a22e5bc227e528c5 : SUCCESS in 8m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2980a5fffe4f4bfd8510617d4e5210f3 : SUCCESS in 9m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/63aa3d005cd349d29aba19ea6461ecf0 : SUCCESS in 4m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/709fe22b304d4e9e984c0ddb513ad30d : SUCCESS in 6m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7dfd28def16b4df99be51ae239a28258 : SUCCESS in 9m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c265398a0fc64ff3b9666add861870ac : SUCCESS in 9m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc8ab2276699455aa5b9e525da57ea0c : SUCCESS in 6m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a698d614ea5e496fa827f153c2709bd8 : SUCCESS in 31m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2e8ac177d070475fa2dd660a5730b611 : SUCCESS in 12m 49s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/155a48f08f4d4655b131e1df0ed5b60e : SUCCESS in 27m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/90c262c78f3a439095d475b455fdbe1e : SUCCESS in 21m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/58252a7e18ee4eb0b3f89048c1da25e4 : SUCCESS in 17m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b25da3f2e6af47aaa6cc86c15f259c9b : SUCCESS in 13m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/745fdd9876d54fc0b085c3d3529a1731 : SUCCESS in 11m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e642d03f99c041a88a8c57261bb6db23 : SUCCESS in 19m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f589dad80f1e4882951ddb9911b70776 : SUCCESS in 5m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/26e23e0b112245eebe8b26031ed15375 : SUCCESS in 8m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4eda36de1dbf4eb6a2ca139087a211e5 : SUCCESS in 9m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9aa98ce92f3f4f68a1207bf29b422981 : SUCCESS in 59m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/caa58860675546c5bb4baf4e0433c22c : SUCCESS in 1h 33m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/68db7f9714c9457eb9e3c4a19844afac : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0806ffdbfc024a1d07e13f0d71e90a746b96799b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-01 05:03:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b17d8e98265f10886ceb208a9582fb47ccb568c6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-01 11:41:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8b07dfc43826be7f8a6e6736a405ba9d11037fe6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-01 11:41:49.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0052fb46d87c990e5063806d7841f33e6457066","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 11:42:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5812e6586e8564c59173f0dafee296c797621f53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 13:20:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/753b26d70d204751954d2fd92bec78c0 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71579f324e824c1baaf8c60e1b1ca9c0 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5be6b0a9880e4f58b6436ddff53c9a4a : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad4c3abc6879402886d56966911c720a : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/081753fd3ddb4d3987aa59c5115d16b8 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1112a05f9cbf4644840608bd294276ab : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fde271ffbdd74856bb84023c5af1d3a4 : SUCCESS in 10m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/13d58caa46f54ba5ad16161781965afc : SUCCESS in 6m 21s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/55b789d1bd2c4304aed67b410db73e43 : SUCCESS in 8m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a15fc3737bf940918165525a4fe1a7e8 : SUCCESS in 10m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/002fdd1fb85f495ba4c63b748ea6a3e2 : SUCCESS in 5m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3f24aa4587ae41fb9b081203d5ea94e1 : SUCCESS in 4m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/89e46c7a898140e8830d00fd3e4cb13e : SUCCESS in 10m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/918c6b345c3c4c618d4e55b80b817414 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1cdafc1ee1fb463eb37aae6ee2e7a2c4 : SUCCESS in 6m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/81cec71673834eafa305df620e238eef : SUCCESS in 18m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3278f618c8674bdb83c28312329beea5 : SUCCESS in 14m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6463f25836524242a08e1129e78db8cf : SUCCESS in 23m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b7f4cc90482442b0b937a454eeee22aa : SUCCESS in 19m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f0e7629b125f4e9a87972cb4b65990a9 : SUCCESS in 12m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6df0952dbae34652a4a276729a84d7c5 : SUCCESS in 12m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/dd50d3e468614814a71d8c0a0624df9c : SUCCESS in 13m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6bd1dbd360f24aa7a0a389036613f860 : SUCCESS in 18m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bf18d992cf8a49188f468f2abd6fa01d : SUCCESS in 5m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7708565a00924a27a12871165bfbf82a : SUCCESS in 9m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1f880891a26a4da9ae57234bb1981b10 : SUCCESS in 8m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2699408c977c4077a0888e9b9e0647d2 : SUCCESS in 1h 08m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5c35eaa2964c434fbb8feac66ae64c97 : SUCCESS in 1h 33m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/db2d3e19be1949b499ed1ad65c542ac2 : SUCCESS in 55m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1a227be0152baa4b341b0651e35c4964a515df67","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 13:20:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"453513c0f8619357a8fb1b624d898d4cff819eb0","revisions":{"453513c0f8619357a8fb1b624d898d4cff819eb0":{"kind":"REWORK","_number":1,"created":"2020-11-30 11:13:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/764636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/764636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/764636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/764636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/764636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/764636/1"}}},"commit":{"parents":[{"commit":"66f1d12ac510c816e0361ed86a43b9ed14236987","subject":"Merge \"update constraint for oslo.concurrency to new release 3.30.1\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66f1d12ac510c816e0361ed86a43b9ed14236987"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-30 11:13:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-30 11:13:15.000000000","tz":0},"subject":"update constraint for castellan to new release 1.3.4","message":"update constraint for castellan to new release 1.3.4\n\nmeta: version: 1.3.4\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: If5898b9139fab8378a193fcd26ea1cc024c57dba\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I57219027d23e35ddc3147becffcfda099efadf9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/453513c0f8619357a8fb1b624d898d4cff819eb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/453513c0f8619357a8fb1b624d898d4cff819eb0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
